)]}'
{"id":"openstack%2Frequirements~820372","triplet_id":"openstack%2Frequirements~master~I6f81188a4f835a636a113b405993582ce5e05141","project":"openstack/requirements","branch":"master","topic":"new-release","attention_set":{},"removed_from_attention_set":{"14288":{"account":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"last_update":"2021-12-03 15:26:55.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I6f81188a4f835a636a113b405993582ce5e05141","subject":"update constraint for openstacksdk to new release 0.61.0","status":"MERGED","created":"2021-12-03 10:57:15.000000000","updated":"2021-12-03 18:22:49.000000000","submitted":"2021-12-03 18:21:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"820372-new-release","meta_rev_id":"72b062b5a9933efdab211d006206462383a9414e","_number":820372,"virtual_id_number":820372,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-03 18:21:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-12-03 15:26:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-12-03 15:26:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"}],"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-03 11:36:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-12-03 12:41:13.000000000","updated_by":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"reviewer":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"state":"CC"},{"updated":"2021-12-03 12:41:13.000000000","updated_by":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2021-12-03 12:45:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8dff7888f5a38b392102fcc091bfdc3ca97c0d17","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-12-03 10:57:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"481fbed7d808d74f0530cfb189552dd172ea5673","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-03 11:36:36.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/db888033dc034e7c9915e20d76b692e1 : FAILURE in 16m 34s","accounts_in_message":[],"_revision_number":1},{"id":"62b71c9913d65fb842c992c64c82f7d538eb969d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-03 12:45:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbedd14eace64d7b8fc1fd7bb567075a : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3d64bc0ded443e99ec88484d15cf188 : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6b28d760f0964b71b88752b61052356e : SUCCESS in 5m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fbe2008a09cc40958bcb29e5fe9338e6 : SUCCESS in 5m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/05e12bf89e7f4afc81d5af32dba590ed : SUCCESS in 5m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7dfd2400b3dd4f8eb7720aca9b879301 : SUCCESS in 4m 34s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/bab3f1cb18e9476782545f428516fa48 : SUCCESS in 9m 57s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2243e00a152744c385ae3bc36e2cdf8e : SUCCESS in 10m 26s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/ab7d55d5f58043bca1db129369f669a1 : SUCCESS in 6m 26s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/84d757257fdc4e5c8df794e624cc5e5f : SUCCESS in 11m 24s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/b34474c2071d4185b86bef7a1723abcc : SUCCESS in 6m 50s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/ca328754a9d04e3aaafcd9103ed9c2b2 : SUCCESS in 11m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/96228cfbf23c42f594886d418119b8bd : SUCCESS in 8m 37s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/09d5fad7a5f9458f99acdcad368e205a : SUCCESS in 8m 17s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/7956528e5f8347759d48f161cfc8ad68 : SUCCESS in 29m 41s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/48e810ceed144b2fb2f961d4a52ca6b7 : SUCCESS in 5m 13s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/bdb916dd84b74de485f96466184f42fb : SUCCESS in 18m 36s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/19e82586c02b48d78743389b70aa91d4 : SUCCESS in 5m 55s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/e6069e0dbcad483982df025dcd3e80d9 : SUCCESS in 23m 08s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/33e72b7520104bf2b4062ebc3161734e : SUCCESS in 12m 41s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/8eaca0e3cb834a3da93c3b61fa76f9c7 : SUCCESS in 6m 18s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d4f75888dd9e4cb69acd8e17c53330d7 : SUCCESS in 4m 43s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/19a77b9a2701434884e68ec69887ec20 : SUCCESS in 23m 15s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/5f2bd582e6734efda4748c3eda65324c : SUCCESS in 6m 41s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/41d02635b7ee47ffae9ba43a1b6399a0 : SUCCESS in 6m 40s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9a371c4fe34f48548ba623bcfac1c039 : SUCCESS in 8m 36s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/837ff9c4978d4e1c996ff430ba78842f : SUCCESS in 12m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/284d4221d28c4ddea7a4d5a430260204 : SUCCESS in 5m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7f17084df6944cafa793a7c69a909b8d : SUCCESS in 57m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ddb2a0f710bf463b9adbed68c4fd1fc1 : SUCCESS in 1h 25m 37s","accounts_in_message":[],"_revision_number":1},{"id":"c359e1f0658947b4028642a8661d0f5eba1ca079","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-12-03 15:26:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ac1c1cd6d71e2a663c3887a390762078faf33e15","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-03 15:28:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0b030c25c08db628b2df09452df0ac0e8f07880e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-03 18:21:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa04fe409ad246e29b26981b2110c3ea : SUCCESS in 5m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1cada64912f7407fb641c9fccc3fbad2 : SUCCESS in 6m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b94a90a354de4253947aa1beefa1ea85 : SUCCESS in 5m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5222e4e33238499c9bc1c129aaa1f95a : SUCCESS in 6m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/15f6abca9ba24f08aa4b55e8aee8d121 : SUCCESS in 4m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/21304ca2e1f84b18bf57b2519b275e9c : SUCCESS in 4m 50s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ae9b157ae95d47a8b44897a2c1963ed6 : SUCCESS in 10m 13s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a3bbfcf99b59457e8f4c71b3e83aa567 : SUCCESS in 9m 50s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/4a3bd780c724481eb14af87a675859e1 : SUCCESS in 11m 10s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/3b126124f9e24a4f984a5702f7d01862 : SUCCESS in 6m 05s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/5db5885725854ac29f31732c0359b9c5 : SUCCESS in 13m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/350d9c6ea7454978b59c3bdd4c44c30e : SUCCESS in 8m 47s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/95df868229504cc1b1f2448f5c29576c : SUCCESS in 9m 18s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/5e9b1c09e36e4560a72fbcb94cca732d : SUCCESS in 16m 39s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/a881da269e8145068ac9a6b81f212bdf : SUCCESS in 5m 50s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/132ab4d5da3e48a4b9780f850a5dd6fc : SUCCESS in 19m 10s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/554780ac607b45d59d5ec18a2d932e56 : SUCCESS in 6m 28s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/128ab8de8f6b498594b818474ba1b23d : SUCCESS in 25m 36s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/58276fc367f8455b88660e710cb0431d : SUCCESS in 12m 33s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/f1c1098424414f9f95348df9e1db9724 : SUCCESS in 5m 54s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c5466c4be29e4f7181b138f5a9830b4c : SUCCESS in 4m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ed51d9cde43f42cca96ec67c206e5056 : SUCCESS in 21m 44s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/66fb28d55dbc408ea0c530befda6604f : SUCCESS in 7m 47s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/ebd910efd02c4731a9145051b0236c1c : SUCCESS in 6m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/80e7ad52f0dc41adb9336cc6e1e9b959 : SUCCESS in 7m 51s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/64ef45a15ad74251a63cc347ce1df5ff : SUCCESS in 12m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c8dfb46db580401e89b450ea5ecb1be8 : SUCCESS in 4m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1a8fba6e6973400fa1c449eee2a8b3ab : SUCCESS in 51m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2a34826bfe3b4180a90809cbdc61e302 : SUCCESS in 1h 29m 02s","accounts_in_message":[],"_revision_number":1},{"id":"dae091c91adcfaf40cf7d58acd05eb3389e297dc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-03 18:21:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"72b062b5a9933efdab211d006206462383a9414e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-03 18:22:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e77027ff0ebb45f3a8f04b1d241d2c98 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"552d7a310dba597b328d810a2175a90b5c2a79da","revisions":{"552d7a310dba597b328d810a2175a90b5c2a79da":{"kind":"REWORK","_number":1,"created":"2021-12-03 10:57:15.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/72/820372/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/72/820372/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/820372/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/820372/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/820372/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/72/820372/1"}}},"commit":{"parents":[{"commit":"0583cb9421b5b8a9aca9aadff5c0011866fa51da","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0583cb9421b5b8a9aca9aadff5c0011866fa51da"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-12-03 10:57:15.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-12-03 10:57:15.000000000","tz":0},"subject":"update constraint for openstacksdk to new release 0.61.0","message":"update constraint for openstacksdk to new release 0.61.0\n\nmeta: version: 0.61.0\nmeta: diff-start: -\nmeta: series: yoga\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Nurmatov Mamatisa \u003cnurmatov.mamatisa@huawei.com\u003e\nmeta: release:Commit: Artem Goncharov \u003cartem.goncharov@gmail.com\u003e\nmeta: release:Change-Id: I289b582d653b49e81bcf5d8f91997cc53e812455\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I6f81188a4f835a636a113b405993582ce5e05141\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/552d7a310dba597b328d810a2175a90b5c2a79da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/552d7a310dba597b328d810a2175a90b5c2a79da"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
