)]}'
{"id":"openstack%2Frequirements~820377","triplet_id":"openstack%2Frequirements~master~Ifdfe6a69e31ce4deddb4df8553b8656bbe3be282","project":"openstack/requirements","branch":"master","topic":"change-820377","attention_set":{},"removed_from_attention_set":{"34120":{"account":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"last_update":"2022-03-17 11:57:24.000000000","reason":"Change was abandoned"},"32926":{"account":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"last_update":"2022-01-13 09:39:09.000000000","reason":"jpodivin replied on the change"}},"hashtags":[],"change_id":"Ifdfe6a69e31ce4deddb4df8553b8656bbe3be282","subject":"WIP/Test - Replace paramiko with libssh","status":"ABANDONED","created":"2021-12-03 11:55:41.000000000","updated":"2022-03-17 11:57:24.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1176fb078f2a6f68342a6053046de767cd9ef79b","_number":820377,"virtual_id_number":820377,"owner":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-01-12 18:18:15.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"}],"CC":[{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-03 13:24:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-04 14:34:57.000000000","updated_by":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"reviewer":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"state":"REVIEWER"},{"updated":"2022-01-13 09:39:09.000000000","updated_by":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"reviewer":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"state":"CC"}],"messages":[{"id":"ef82f783ff21ddbc0a6e6caa85b5245b176670a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"date":"2021-12-03 11:55:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ac72868fcba3313ffada2eb6df4e84510c1d2854","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-03 13:24:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b7b9068aced4aaabff6b9fdbe9a5ae7 : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cadb12c98bc6442d819f0ac618ea88a8 : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3ec4663add52438a9beb0c38ad99132b : SUCCESS in 4m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/46d184bc9b9a40cdb982dddd177fedb0 : SUCCESS in 4m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/64a8ec6ac46544b68588da40a14eff0f : FAILURE in 3m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e501c789ed6e4d53805546206d00543e : SUCCESS in 4m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6306009695774da3b6830be8baef3cf1 : SUCCESS in 53m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b7e1a7842ba2454f9f5ce91729e2482b : SUCCESS in 1h 25m 41s","accounts_in_message":[],"_revision_number":1},{"id":"ecf7d027335d9d9dbe5ef4d3bc8a1bc6d8762f05","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"date":"2021-12-03 14:29:18.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d90259c957d86870cae2343b1ba1c9a709818c17","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-03 14:54:21.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/2ced51b39db74eb39bf85d20f29a0f26 : FAILURE in 22m 39s","accounts_in_message":[],"_revision_number":2},{"id":"333a76144af59cd4321f24ad34158ea1f73b0398","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-03 15:53:14.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a78d131d67644790968c004ad6152bf0 : SUCCESS in 5m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2dd6231d56a24fe084efb2bba0976acf : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0a19482bfb964fdc86e14de34da4e798 : SUCCESS in 5m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ab01b65b94b545b3baa0dee6526a227f : SUCCESS in 5m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a9de149755d6423a94f80afd631da854 : FAILURE in 4m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7bb9f99743d041f58aa88d53f9dbec41 : SUCCESS in 4m 08s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/59143b7cc04c4125944dd3c6dcd8833c : SUCCESS in 10m 35s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8c738ab655844a6894c60bfb9ea1171a : SUCCESS in 9m 52s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/170b5a12a7a24648b59dfd6a9a9fb36d : SUCCESS in 5m 40s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/052e4b9bc15c480985c87740595eb336 : SUCCESS in 11m 12s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/3d888eba9e5e40ba8c057485c5774dd6 : SUCCESS in 6m 05s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/b3959a44a3fd4359a9542a133cf194be : SUCCESS in 10m 26s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8e8335c7dd2f4f73aa67baffc7301659 : SUCCESS in 8m 32s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/043e5c6d8b7f456aa4487bc2c81add93 : SUCCESS in 8m 41s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/53bd321ff07a42da87d854204bf72f8f : SUCCESS in 19m 07s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/99f4e1bca9994a72a165da0ec243aa69 : SUCCESS in 4m 31s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/45a66d6e3a5645e59f811a209636c4c1 : SUCCESS in 20m 19s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/7defa101563e4b778356ef220097d0d0 : SUCCESS in 5m 39s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/40689ebc7ad44dffb7b0c853671b2afb : SUCCESS in 24m 31s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/24bc8fdb65c54e128ef1cfcf77f05007 : SUCCESS in 15m 39s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/77b25f3b6e7246a7aa66ee0b750119c4 : SUCCESS in 6m 10s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/49ad78aa1615456ab64a685b2ff00e3e : SUCCESS in 3m 45s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/aa9c914e44e84bf98e87e85ba2b4793f : SUCCESS in 21m 05s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/1e636e4708884f49a216a6d3c17ea5b6 : SUCCESS in 6m 47s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/2d5e5e051fd44a16b17cf810b7fd911d : SUCCESS in 5m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/391eec7faacb4830aecbb9b383ad60c8 : SUCCESS in 7m 07s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/1c5b5c4dfbee404aa0d0553b78abef96 : SUCCESS in 14m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bcd88d4a70e748b2af9d6641616ad3f1 : SUCCESS in 4m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d889bce2acd04d0c94048d6b559252d9 : SUCCESS in 50m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d94fa25080b74c7ab2863701a3baa103 : SUCCESS in 1h 21m 29s","accounts_in_message":[],"_revision_number":2},{"id":"2fc424365f9fbf04296a26bdcbbddaa2d51de31c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"date":"2021-12-06 10:26:54.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"fc303114e7c7cceb23fb72fc820c36658283c964","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-06 10:45:09.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/50563f2718d64cab88789d4ec3712491 : FAILURE in 16m 12s","accounts_in_message":[],"_revision_number":3},{"id":"b48160040b8a104f3522a2ed17f6a8c4d431255b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-06 11:57:28.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df040dd7ffbb4aada6ff3d00f7409262 : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6800e92f15f40dba9bcba907c7226e1 : SUCCESS in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b964cc05513341c3985782935156500b : SUCCESS in 5m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d6be489fe8104cbd8ca989843a1ad93e : SUCCESS in 5m 52s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/27c9c4ee48e34b2cb5501d29dc726529 : FAILURE in 5m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/845b0f6447724ccdb4b3279ffca4e9d4 : SUCCESS in 3m 45s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3da89e4574f948b091abe9c7eb5126e8 : SUCCESS in 11m 24s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/127061b7f0e248f5a9d327313b36f195 : SUCCESS in 11m 51s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/ac2cdcae29c24708ba8692420d74ed09 : SUCCESS in 6m 03s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/fa01f656846e45e58811ea3b3c19f71a : SUCCESS in 12m 03s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/eaeaace547dc4dfcb7cc10b7b6949af4 : SUCCESS in 5m 59s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/c630a8acedaf4db08b2019e97448df8c : SUCCESS in 11m 57s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/88125812ab504a7c98c138e52866a949 : SUCCESS in 8m 07s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/adadfcb21e294db18f4235ea655f0af8 : SUCCESS in 9m 58s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/adb6a8dc5ead4fa99ee4f80a74d22f52 : SUCCESS in 15m 47s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/782b37b8924341cfa49930572f9bee6e : SUCCESS in 4m 18s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/9572ad6c6b1c43d5ae477623f90b317e : SUCCESS in 18m 54s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/1991f75d435542789cd63c01d832ee33 : SUCCESS in 5m 37s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/e2e7fc3ca3904b17b3417a417a2ac52a : SUCCESS in 26m 44s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/baeff6c58dd54506b0c0ca1b40972bfa : SUCCESS in 12m 34s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/eac119c80b3042ff8376705a8559c83f : SUCCESS in 5m 22s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b593adb0bf40439ca6e5c4eedec8595e : SUCCESS in 3m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/25694df88e164babaf7cc158bbc7f247 : SUCCESS in 23m 24s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/f804954260c441bdab963fcb0c5da25f : SUCCESS in 5m 36s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/b1ec34c1cf7c4b25b779f627351d9b3f : SUCCESS in 4m 55s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/590e843a3f8d435bbabf9b1cdd8ba417 : SUCCESS in 6m 12s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/8d6dacbf6b4e4a8a8dd7136420f51f2a : SUCCESS in 11m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9da0f4a4652344eaafeed10ab5b85ba4 : SUCCESS in 4m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/181f88dc98644fb48df1c7c038c835fb : SUCCESS in 44m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/72dfaa7ed27149fab71caa5f6f0d7597 : SUCCESS in 1h 22m 57s","accounts_in_message":[],"_revision_number":3},{"id":"b29817b5e0ccc29c7d53e5ec18c7d110929b2e9d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2022-01-03 14:40:39.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"32f81a6fefb6cfff0074abb0369dd797627183db","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-03 14:56:37.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/e3c5c0eb73bc45c9833ad73de590d898 : FAILURE in 11m 43s","accounts_in_message":[],"_revision_number":4},{"id":"9ed2ced2f7c5b2dad143d9e89c4243ce8c521ee8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-03 16:03:44.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fabe66ab3ba94c84bab03d4e15bf9112 : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa35fd9f8eeb4ad08abab058367cb84c : SUCCESS in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4a906f8964fe498d9ebbcc74be1f84be : SUCCESS in 5m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5f6f79ea8e4249a68b589df8b661e42f : SUCCESS in 4m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3ada92a8ed6144dbb07913e95721480b : SUCCESS in 5m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7e4f514156154f848fbae450da2b38b7 : SUCCESS in 4m 45s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8da74d052689464a831cb3b565bf94a2 : SUCCESS in 11m 41s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2acd3ffd40ff4f26a7413ec5ab45d223 : SUCCESS in 11m 18s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/e0ad827eee9c4a10ba8da1c57d7b4521 : SUCCESS in 6m 29s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/143ccb0dc052417089f9c6fcf1c48c73 : SUCCESS in 11m 47s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/2566c34caa144e8a966da5821eabdf64 : SUCCESS in 7m 06s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/f737eec7f94f4db3bd2cf8b88fd0c867 : SUCCESS in 9m 42s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c44939f7e10b4b379bf855eefee67021 : SUCCESS in 8m 04s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f26c266587d7443fb03020aa9f7001d6 : SUCCESS in 9m 33s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/0c0bf156749140ee8ca73130150799ad : SUCCESS in 21m 10s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/88d977b2c87e47e089d2f70f01d1563b : SUCCESS in 4m 33s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/4df2d3dc4f99481d89a565a360924c80 : SUCCESS in 15m 02s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/044df7c316a6498c97a40a0ff0ba171f : SUCCESS in 6m 38s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/d32f28e12a62483d80124aa8c620f0a3 : SUCCESS in 23m 33s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/e737e80d86e5433db3ba1597b9204ca0 : SUCCESS in 10m 43s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/e8bea3ac502c4c9caed4b3b45804b254 : SUCCESS in 6m 20s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f7a43532c3bc427596d5233ac5c8b429 : SUCCESS in 3m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0d0ad170373e4fef8a00fe39b9963aab : SUCCESS in 20m 10s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/ad59da793ca8403284a3a1d39c22a3de : SUCCESS in 5m 55s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/b3789ec4c8ba4798b6d265b72aac2185 : SUCCESS in 5m 46s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c886863115de4762b2cc2dd86619d4ff : SUCCESS in 6m 28s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/0d6fcffbe72d46bfb4360cca81e4c904 : SUCCESS in 13m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f891c90fca4244a9b7d12ecca7ab071f : SUCCESS in 35m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3419e3d644de4ff3a2117ecd889d0ebb : SUCCESS in 1h 20m 42s","accounts_in_message":[],"_revision_number":4},{"id":"e25022f887cf928939bd4a7fb7b02db96cf1d916","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2022-01-04 14:34:57.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"9c5fceece1d91e4be25b6f529d59666f7e2ba1b2","author":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"date":"2022-01-04 14:37:17.000000000","message":"Patch Set 4: Code-Review-1 Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"9c68d6b15d9667ffef0e0330d6d9365179c4001d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2022-01-04 14:47:49.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"b950d9f1d2eabbc877e868700b79e6e4550fa359","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-04 15:06:18.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/4ea01095b61c402fa492e1dbdfa524c0 : FAILURE in 16m 34s","accounts_in_message":[],"_revision_number":5},{"id":"e662db419b04dfc6a85fe14d3330ccbbc8aec2da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-04 16:25:59.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2335b1e3fb74f898c4bc3370cfec072 : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06c226bd4020458d8476f6a486f1df34 : SUCCESS in 3m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bc27b6f666704f8a8d77e5a2bf3aa249 : SUCCESS in 4m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/361fef5a6bd74c4db23b22fc78e823e6 : SUCCESS in 4m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b81b8781f1c94bf9b816db8fb6a03445 : SUCCESS in 4m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/be7c458cd45d44b89de3eb22130c551c : SUCCESS in 4m 13s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/fed4676dbeb74a72b037b00daf0850f0 : SUCCESS in 9m 58s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/da3c9968e3fa449bae014aaf1989288e : SUCCESS in 11m 28s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/921a23b4e00441c998ccadd5ea719800 : SUCCESS in 5m 43s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/2a924b29fb104c2ea00d21d04b8784ee : SUCCESS in 12m 22s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/a92fa6e0833f49e988b5f815c765589c : SUCCESS in 5m 50s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/3178a95a7cc2447b9acdba8c33ce094c : SUCCESS in 11m 04s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4aac99e71a1746fe98b67ae0cdc6a0cf : SUCCESS in 8m 18s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/aafa02c8e3d14d1e86bb3c3feca2e5b8 : SUCCESS in 9m 52s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/bdc3ca8135f042b487bd7b6590d1c67c : SUCCESS in 13m 43s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/099cb6745ba74b2298e6fea33946f4bd : SUCCESS in 3m 49s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/822a1dbda2bf42f6a2c60893b052308d : SUCCESS in 23m 56s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/6a3467d264194d5dbb8517b3e9e22b47 : SUCCESS in 6m 26s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/1b05388bb5de4f1ba0bf225f355d1654 : SUCCESS in 24m 47s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/94aad998a743442192685018063d7073 : SUCCESS in 11m 58s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/c767d94a295e4a3a90e07e062b5f28bd : SUCCESS in 5m 34s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a4b0495a01434650beb9ca01688b3370 : SUCCESS in 4m 31s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f86fff8e3d6a45eab61a6d872a0a63ab : SUCCESS in 19m 46s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/b77255f0aadb44769791bdb6a199e4cb : SUCCESS in 5m 53s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/487f409232dc40f6b347df2ff61794e8 : SUCCESS in 5m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8fdfdd6b991f48ff83d3825d49b20dea : SUCCESS in 7m 15s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/ba15a5f61939482c8df8872d954ecb26 : TIMED_OUT in 41m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/493cfd984bd04abb97918a1db3f53ba3 : SUCCESS in 49m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/db4984115368400b9993971a5922caf4 : SUCCESS in 1h 35m 19s","accounts_in_message":[],"_revision_number":5},{"id":"d78ee2a3738155bc790a85ed4461ca6b02251133","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2022-01-12 09:49:40.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"38e5f434c43a849bba8fa9cb7182b11fc57db75f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 10:13:08.000000000","message":"Patch Set 6:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/c77f296d3f21433aade9d7005a6b9838 : FAILURE in 17m 39s","accounts_in_message":[],"_revision_number":6},{"id":"b2fe6c46c8717a605ce0dd481603eee35f28eff0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 18:18:15.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8dd132f5d854409a81126613ea1068be : SUCCESS in 6m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb77cd1ebd644d589811cec04f4dc8ad : SUCCESS in 5m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bdeabee642f64835835b6ba4eaa118ea : SUCCESS in 5m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ffd39987bb8b468eb549149c522d4853 : SUCCESS in 6m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/78d32bdc8f91419c96ee9a59b6604d32 : SUCCESS in 5m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/549f9058dad345e5b9b49fe21b15d2ba : SUCCESS in 5m 18s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6587b6e50f7940ee808a6d3449c00ecd : SUCCESS in 14m 16s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f44c4fdb874e4c84bd4c4b58cd8c9348 : SUCCESS in 13m 43s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/24c804bb7cd64793a872b70c8b5a17a1 : SUCCESS in 9m 17s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/a98a5677dfa44907a58e9b28c2674c39 : SUCCESS in 11m 42s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/f5052aa7c2fa4ba48bc7eb7ad7618f94 : SUCCESS in 7m 54s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/c0810400403d4ba39f99a59466df2fce : SUCCESS in 11m 57s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0da11950d9814050a33fdeb7378bbb2d : SUCCESS in 10m 16s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/2e28366d480140feb8035453bac482ba : SUCCESS in 10m 52s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/73cc06c8c5b649e8a4c8f3828eb0a9de : SUCCESS in 18m 27s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/6dcc2f0d693e49a1a0b2aac763e443f2 : SUCCESS in 6m 09s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/0be0e55803774d918cc3b98831a3e16f : SUCCESS in 18m 24s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/a982e2aee17649a79904ebde529a9571 : SUCCESS in 7m 50s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/98aeb149eedf447a9c40ce6dc2fc6d1f : SUCCESS in 26m 28s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/c3c3783c867048f38ad508011225907a : SUCCESS in 14m 07s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/e19b387040bd444097b90890ddb7771c : SUCCESS in 8m 13s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/131aa86e071643e7ac622e9532d9d23d : SUCCESS in 5m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ecc2e347687b46e182cde74026956e80 : SUCCESS in 24m 37s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/d71ebce22d934d7f8b5af592ecb813fe : SUCCESS in 8m 24s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/6b1b4b0d69a44e608381caaeccff66f7 : SUCCESS in 7m 26s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3c6d29c1907d45bb9102d50bd858273c : SUCCESS in 8m 49s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/144e40a230684f2994686ff26bdfd5b9 : SUCCESS in 13m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f6beb5cde82a4821b4e040f998b07a17 : SUCCESS in 44m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a4f87b8e5323414da6acd96f75913256 : SUCCESS in 1h 41m 13s","accounts_in_message":[],"_revision_number":6},{"id":"5586f2b9f9d2ca6d39d8fd45a9849e30287b6f4d","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2022-01-13 09:39:09.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"1176fb078f2a6f68342a6053046de767cd9ef79b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"date":"2022-03-17 11:57:24.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"ba2e3b28285f435a00d728946d317cc14118bf40","revisions":{"679074ebaaf27eaa0cd3a29c5b2b54f8b584da30":{"kind":"REWORK","_number":1,"created":"2021-12-03 11:55:41.000000000","uploader":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"ref":"refs/changes/77/820377/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/77/820377/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/820377/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/820377/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/820377/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/77/820377/1"}}},"commit":{"parents":[{"commit":"0583cb9421b5b8a9aca9aadff5c0011866fa51da","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0583cb9421b5b8a9aca9aadff5c0011866fa51da"}]}],"author":{"name":"Andre Aranha","email":"afariasa@redhat.com","date":"2021-12-03 11:48:50.000000000","tz":0},"committer":{"name":"Andre Aranha","email":"afariasa@redhat.com","date":"2021-12-03 11:53:04.000000000","tz":0},"subject":"WIP/Test - Replace paramiko with libssh","message":"WIP/Test - Replace paramiko with libssh\n\nparamiko is not fips certified, and so, cannot execute on a system\nwhere fips is enabled. This is an initial patch to replace paramiko\nwith ssh-python which uses libssh.\n\nChange-Id: Ifdfe6a69e31ce4deddb4df8553b8656bbe3be282\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/679074ebaaf27eaa0cd3a29c5b2b54f8b584da30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/679074ebaaf27eaa0cd3a29c5b2b54f8b584da30"}]},"branch":"refs/heads/master"},"420208fbb1114a46fc9e9faeeae94663823f72ad":{"kind":"REWORK","_number":2,"created":"2021-12-03 14:29:18.000000000","uploader":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"ref":"refs/changes/77/820377/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/77/820377/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/820377/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/820377/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/820377/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/77/820377/2"}}},"commit":{"parents":[{"commit":"0583cb9421b5b8a9aca9aadff5c0011866fa51da","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0583cb9421b5b8a9aca9aadff5c0011866fa51da"}]}],"author":{"name":"Andre Aranha","email":"afariasa@redhat.com","date":"2021-12-03 11:48:50.000000000","tz":0},"committer":{"name":"Andre Aranha","email":"afariasa@redhat.com","date":"2021-12-03 14:26:35.000000000","tz":0},"subject":"WIP/Test - Replace paramiko with libssh","message":"WIP/Test - Replace paramiko with libssh\n\nparamiko is not fips certified, and so, cannot execute on a system\nwhere fips is enabled. This is an initial patch to replace paramiko\nwith ssh-python which uses libssh.\n\nChange-Id: Ifdfe6a69e31ce4deddb4df8553b8656bbe3be282\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/420208fbb1114a46fc9e9faeeae94663823f72ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/420208fbb1114a46fc9e9faeeae94663823f72ad"}]},"branch":"refs/heads/master"},"2da0cc5f382030daecd0c2836f700e3c4286bdd7":{"kind":"REWORK","_number":3,"created":"2021-12-06 10:26:54.000000000","uploader":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"ref":"refs/changes/77/820377/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/77/820377/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/820377/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/820377/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/820377/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/77/820377/3"}}},"commit":{"parents":[{"commit":"14e7426f7e6e9d55d3f5d52999306fdc01bfda12","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/14e7426f7e6e9d55d3f5d52999306fdc01bfda12"}]}],"author":{"name":"Andre Aranha","email":"afariasa@redhat.com","date":"2021-12-03 11:48:50.000000000","tz":0},"committer":{"name":"Andre Aranha","email":"afariasa@redhat.com","date":"2021-12-06 10:24:01.000000000","tz":0},"subject":"WIP/Test - Replace paramiko with libssh","message":"WIP/Test - Replace paramiko with libssh\n\nparamiko is not fips certified, and so, cannot execute on a system\nwhere fips is enabled. This is an initial patch to replace paramiko\nwith ssh-python which uses libssh.\n\nChange-Id: Ifdfe6a69e31ce4deddb4df8553b8656bbe3be282\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2da0cc5f382030daecd0c2836f700e3c4286bdd7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2da0cc5f382030daecd0c2836f700e3c4286bdd7"}]},"branch":"refs/heads/master"},"0b58e46dbc3001d6b8a81003f1613c5eac5de1fb":{"kind":"REWORK","_number":4,"created":"2022-01-03 14:40:39.000000000","uploader":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"ref":"refs/changes/77/820377/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/77/820377/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/820377/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/820377/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/820377/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/77/820377/4"}}},"commit":{"parents":[{"commit":"9fdc8d1e641e4e5c84a6805baa1deca308c98508","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9fdc8d1e641e4e5c84a6805baa1deca308c98508"}]}],"author":{"name":"Andre Aranha","email":"afariasa@redhat.com","date":"2021-12-03 11:48:50.000000000","tz":0},"committer":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2022-01-03 14:28:06.000000000","tz":60},"subject":"WIP/Test - Replace paramiko with libssh","message":"WIP/Test - Replace paramiko with libssh\n\nparamiko is not fips certified, and so, cannot execute on a system\nwhere fips is enabled. This is an initial patch to replace paramiko\nwith ssh-python which uses libssh.\n\nChange-Id: Ifdfe6a69e31ce4deddb4df8553b8656bbe3be282\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0b58e46dbc3001d6b8a81003f1613c5eac5de1fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0b58e46dbc3001d6b8a81003f1613c5eac5de1fb"}]},"branch":"refs/heads/master"},"fe5d9c7d0c76b9e47117271411e849b00e7dcf79":{"kind":"REWORK","_number":5,"created":"2022-01-04 14:47:49.000000000","uploader":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"ref":"refs/changes/77/820377/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/77/820377/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/820377/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/820377/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/820377/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/77/820377/5"}}},"commit":{"parents":[{"commit":"9fdc8d1e641e4e5c84a6805baa1deca308c98508","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9fdc8d1e641e4e5c84a6805baa1deca308c98508"}]}],"author":{"name":"Andre Aranha","email":"afariasa@redhat.com","date":"2021-12-03 11:48:50.000000000","tz":0},"committer":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2022-01-04 14:44:07.000000000","tz":60},"subject":"WIP/Test - Replace paramiko with libssh","message":"WIP/Test - Replace paramiko with libssh\n\nparamiko is not fips certified, and so, cannot execute on a system\nwhere fips is enabled. This is an initial patch to replace paramiko\nwith ssh-python which uses libssh.\n\nChange-Id: Ifdfe6a69e31ce4deddb4df8553b8656bbe3be282\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fe5d9c7d0c76b9e47117271411e849b00e7dcf79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fe5d9c7d0c76b9e47117271411e849b00e7dcf79"}]},"branch":"refs/heads/master"},"ba2e3b28285f435a00d728946d317cc14118bf40":{"kind":"REWORK","_number":6,"created":"2022-01-12 09:49:40.000000000","uploader":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"ref":"refs/changes/77/820377/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/77/820377/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/820377/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/820377/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/820377/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/77/820377/6"}}},"commit":{"parents":[{"commit":"803231efa6cf05368eaecf61a902a9e2c44f03f1","subject":"Remove centos-8 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/803231efa6cf05368eaecf61a902a9e2c44f03f1"}]}],"author":{"name":"Andre Aranha","email":"afariasa@redhat.com","date":"2021-12-03 11:48:50.000000000","tz":0},"committer":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2022-01-12 09:49:20.000000000","tz":60},"subject":"WIP/Test - Replace paramiko with libssh","message":"WIP/Test - Replace paramiko with libssh\n\nparamiko is not fips certified, and so, cannot execute on a system\nwhere fips is enabled. This is an initial patch to replace paramiko\nwith ssh-python which uses libssh.\n\nChange-Id: Ifdfe6a69e31ce4deddb4df8553b8656bbe3be282\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ba2e3b28285f435a00d728946d317cc14118bf40"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ba2e3b28285f435a00d728946d317cc14118bf40"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
