)]}'
{"id":"openstack%2Frequirements~820378","triplet_id":"openstack%2Frequirements~master~I647a046d26c7d75e327b6ea677483858a2e01d8e","project":"openstack/requirements","branch":"master","attention_set":{},"removed_from_attention_set":{"30742":{"account":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"last_update":"2022-05-23 10:57:23.000000000","reason":"Change was abandoned"},"10135":{"account":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"last_update":"2022-05-23 10:57:23.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I647a046d26c7d75e327b6ea677483858a2e01d8e","subject":"Add icmplib to global requirements","status":"ABANDONED","created":"2021-12-03 12:17:08.000000000","updated":"2022-05-23 10:57:23.000000000","total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8562e6f96887b7610e44e90ba1c95e5bb3062201","_number":820378,"virtual_id_number":820378,"owner":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-12-03 13:50:39.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},"all":[{"value":0,"date":"2021-12-03 16:36:21.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":-1,"date":"2022-03-01 11:37:17.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"date":"2021-12-03 18:41:22.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"all":[{"value":0,"date":"2021-12-03 16:36:21.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},{"value":-1,"date":"2021-12-03 18:31:27.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"}],"CC":[{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-03 12:54:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-12-03 13:50:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-03 16:36:14.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2021-12-03 18:19:32.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2022-02-21 14:34:19.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"CC"},{"updated":"2022-03-01 07:34:37.000000000","updated_by":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"reviewer":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"state":"REVIEWER"},{"updated":"2022-03-01 11:31:06.000000000","updated_by":{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},"reviewer":{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},"state":"CC"},{"updated":"2022-03-01 11:37:17.000000000","updated_by":{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},"reviewer":{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},"state":"REVIEWER"}],"messages":[{"id":"e55420c4ba510b2b3f124e680f30e478d701100d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2021-12-03 12:17:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"352847b4232d424ee015972ece2bb88477a2041d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-03 12:54:23.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/a929256c72b54e18a83406ec8e98fd79 : FAILURE in 13m 49s","accounts_in_message":[],"_revision_number":1},{"id":"380cb3c76db6e4b7cf78b5a1af2f0085f21a1e4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-03 13:50:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b1ba844baa04620933d6e7f5a499d65 : SUCCESS in 6m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b145d10667e4fc58a26a071d1da642f : SUCCESS in 6m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/90fa0cc3d7ee450e9f6be0ee1a21e238 : SUCCESS in 6m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/99abd0e25cbb4bd7aac389bf0b98a21b : SUCCESS in 7m 26s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/190defb3fbea452bb8d82e6f239cb998 : SUCCESS in 5m 30s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/28aa7e5fdcf744b6b8594462a28b3395 : SUCCESS in 5m 09s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ff2390b0b3aa4615a3b16c7fed7c7f43 : SUCCESS in 12m 49s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/cc86abb833bc4caf84b1a4d09a57830d : SUCCESS in 13m 35s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/f25fdb4b47c14570aa4bbb7c498c9b22 : SUCCESS in 7m 31s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/fd04e4bb0cdd4a7b82369de571bcdee1 : SUCCESS in 15m 25s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/4a7fb10457ef47df9be7370838915a85 : SUCCESS in 8m 16s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/46703d81baa9449e827b2f3650d057ab : SUCCESS in 14m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5730c86c05f445a983cbb4d3facbae43 : SUCCESS in 9m 19s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/15be3884351140079091b4d997acac40 : SUCCESS in 10m 04s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/5bbedfde16b34847b80b6f27c05bf3dc : SUCCESS in 14m 59s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/cc4e5f1eed2148c9bf21e61d4e8c093a : SUCCESS in 5m 58s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/2cdb83859b8a4542a14f3bda3d2effda : SUCCESS in 26m 41s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/62101eb4d4d1470f9a8c83e1a8e3cb5a : SUCCESS in 6m 52s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/edbf26ea39c44e3990888793a6ae5fa7 : SUCCESS in 28m 03s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/e5176006aec94758a0a6d365977ac925 : SUCCESS in 17m 02s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/a0677f247d1a4deda91631622665c1b5 : SUCCESS in 10m 54s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/792915d3880347f9b4728d9fdf8406fa : SUCCESS in 5m 32s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5bb498f92d66407fbaccdc1ae8b56591 : SUCCESS in 24m 51s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/76a480df2a164fcfbd297b4a38e8d516 : SUCCESS in 7m 16s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/9e1e7326784e4cb989ade366f4397d4a : SUCCESS in 6m 47s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c6fc9969ecc64891b61ce3d39801e971 : SUCCESS in 9m 11s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/4978e9b579c74f27afccb16dacd5a14f : SUCCESS in 13m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ebab5b2ceb2047539110d4c7d73d992b : SUCCESS in 10m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/45eb020b41db437794891ad6bf52ee77 : SUCCESS in 51m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4fc611be610846ca9cceda40909de211 : SUCCESS in 1h 22m 24s","accounts_in_message":[],"_revision_number":1},{"id":"e9fdd2fa59c5504c0aa34fd36aad1ae5e9bb6a83","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-12-03 16:36:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\ndo we include tempest only libs in GR?  also for any new lib we need https://docs.openstack.org/project-team-guide/dependency-management.html#for-new-requirements answered in the commit message","accounts_in_message":[],"_revision_number":1},{"id":"631fa5950a89e5f993b48fbf58adc1258b7091a5","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-12-03 16:36:21.000000000","message":"Patch Set 1: -Code-Review -Workflow","accounts_in_message":[],"_revision_number":1},{"id":"27d324c63e673e662579c90fd5d39b9add82a791","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-12-03 18:19:32.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f2efa6e82679c197b0525b07741aea820a5c4acf","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2021-12-03 18:31:27.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7fe5bc95d11c0b70cb0bc8c4491ac972cf2498f4","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-12-03 18:41:22.000000000","message":"Patch Set 1: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9588ea53f42a343465617e45856ed41dc60bd15e","author":{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},"date":"2022-03-01 11:37:17.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7efc21f76145465221a7b195909c3b014143901c","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2022-03-01 14:53:59.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8562e6f96887b7610e44e90ba1c95e5bb3062201","tag":"autogenerated:gerrit:abandon","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2022-05-23 10:57:23.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"04862fb4e130081b84c3b6290c1b9c6d7e0fcd82","revisions":{"04862fb4e130081b84c3b6290c1b9c6d7e0fcd82":{"kind":"REWORK","_number":1,"created":"2021-12-03 12:17:08.000000000","uploader":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"ref":"refs/changes/78/820378/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/78/820378/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/78/820378/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/78/820378/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/78/820378/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/78/820378/1"}}},"commit":{"parents":[{"commit":"0583cb9421b5b8a9aca9aadff5c0011866fa51da","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0583cb9421b5b8a9aca9aadff5c0011866fa51da"}]}],"author":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2021-12-03 12:16:41.000000000","tz":0},"committer":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2021-12-03 12:16:41.000000000","tz":0},"subject":"Add icmplib to global requirements","message":"Add icmplib to global requirements\n\nUsed by Tempest and Ib14fa7dc5c8093eed498049cd0a56c8ac6853660 to allow\nus to wait until an isntance is pingable before proceeding with a given\ntest.\n\nChange-Id: I647a046d26c7d75e327b6ea677483858a2e01d8e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/04862fb4e130081b84c3b6290c1b9c6d7e0fcd82"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/04862fb4e130081b84c3b6290c1b9c6d7e0fcd82"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
