)]}'
{"id":"openstack%2Frequirements~820974","triplet_id":"openstack%2Frequirements~master~I0310989941d0b0fafe478369ebd8075134a89574","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","attention_set":{},"removed_from_attention_set":{"14288":{"account":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"last_update":"2021-12-08 05:10:26.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I0310989941d0b0fafe478369ebd8075134a89574","subject":"Updated from generate-constraints","status":"MERGED","created":"2021-12-08 02:52:01.000000000","updated":"2021-12-08 08:09:17.000000000","submitted":"2021-12-08 08:07:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"820974-openstack/requirements/constraints/noclob","meta_rev_id":"4ef899db24e827ca812b5b81baff9dbf43804fad","_number":820974,"virtual_id_number":820974,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-08 08:07:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-12-08 03:32:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-12-08 03:32:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-08 03:32:11.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2021-12-08 04:43:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-12-08 05:00:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2d30987205f1e0105cdf91b2f0dd2385aadc80ff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-12-08 02:52:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1128ab52763f4934b446715b55044cce78591cc5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-12-08 03:32:11.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"12fcef07c84165702f0441eda66b623c7c2cadf7","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-12-08 03:32:14.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":2},{"id":"d59f35efca2eef9d66ac8586e81a909ffeb722c8","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-12-08 03:32:48.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"654daca3a1716db8ddcfdc0e4dcbf52ee851a4bc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 04:43:52.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/e9ca04f675a5446d8681f483cea52e1f : FAILURE in 20m 07s","accounts_in_message":[],"_revision_number":2},{"id":"ca085bde84f53ac632cf52cede6257410cb1eef5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 05:00:11.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c26f6b53104d48e5ab0b836a538f3307 : SUCCESS in 5m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e338585d53124644a8e99da55a6307bb : SUCCESS in 4m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5fb649d6031f41f3838ad36dd0e062d2 : SUCCESS in 4m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9c7f9863c6ff4845b67d9ce3f8ecca98 : SUCCESS in 4m 53s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8dd3296f025f435c8385ee88f88eb050 : SUCCESS in 3m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/47e9025beb33479a814f297e59b7caf1 : SUCCESS in 3m 39s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/eb0729f42564490794c8401e135227e8 : SUCCESS in 10m 24s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/42f889e187db48aa9dc619164b14f830 : SUCCESS in 10m 35s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/d25c89364c824996bab4eb63b8c17735 : SUCCESS in 5m 22s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/1080439a031c49049e190e5aaca91cdc : SUCCESS in 13m 12s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/a5a42fd3fec94c3a82f3d8aa5940564a : SUCCESS in 5m 55s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/70c853543fe94e95ab5f8ef6b8c0fd07 : SUCCESS in 10m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a4ab195926ec436a8d06b8a73ccfb240 : SUCCESS in 9m 44s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/41e98d381a1f4625a24ff00c98f8cc59 : SUCCESS in 8m 59s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/7ff85a0e3d524ef2a10a62cb1ee18932 : SUCCESS in 42m 59s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/2106076cc76a4c39b8f63990aafa467b : SUCCESS in 5m 49s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/052e7848b61c4fc7ada1527f97023a6b : FAILURE in 20m 04s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/67375a6894b241ab899f0c15980e087d : SUCCESS in 6m 26s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/cbd4e0058f2a4743ab17502ccee95931 : SUCCESS in 34m 12s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/d126fcd8559741e6a51ada194b55b844 : SUCCESS in 18m 49s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/025a727f7d7d4669b02b6769d829c11a : SUCCESS in 6m 53s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/92753e1bfa3940fd9259a6c7c052eeb3 : SUCCESS in 5m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/75fda8865917405aa420d98835db8bfb : SUCCESS in 22m 04s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/6b40aba56fb244328538978526a38545 : SUCCESS in 7m 09s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/f07d3c115f284ff7974024023b0120f5 : SUCCESS in 5m 22s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6db0f622cc004cf69ca5c6fbec87fd43 : SUCCESS in 8m 05s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/b963104081c44f4eaf85bf23138ad063 : SUCCESS in 11m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2876985c388d412d8786f5c63b5717c5 : SUCCESS in 4m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/419b03fcdbf14a0d85c0eeec5fb5d5a3 : SUCCESS in 47m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/be8d26e5c4e641ef88b90e7d6623453a : SUCCESS in 1h 23m 09s","accounts_in_message":[],"_revision_number":2},{"id":"e2faf613305f5c17a2881f5c9fbc070a98476c89","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-12-08 05:10:26.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"e96ddc61aadd13457b6fb755a29371490d31a3e3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 05:28:33.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/ee27b2c2acdd4c19b9565449a2857109 : FAILURE in 15m 38s","accounts_in_message":[],"_revision_number":2},{"id":"93dfaa4e107986b6bf5725a4c4dff5e4dceab0fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 06:40:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fafcc83ac35d47cf84893c12fdaa8311 : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb5d08cbab994342bc06c991051d4ff8 : SUCCESS in 3m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a6b89aa1f0654099bf7a8f5423a3df42 : SUCCESS in 4m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2678e48b27de48ae9f7c0c17466b9d04 : SUCCESS in 4m 44s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5c942fa9c0f8432292cc0cb59caff189 : SUCCESS in 3m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f3f3942437204290b74ead23cb95ce73 : SUCCESS in 4m 37s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ff075910834c4351ac30320ea040a26c : SUCCESS in 9m 06s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/00e8e8b449324d7e8b26df85480b5ca0 : SUCCESS in 11m 18s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/b125b330601c498987d2760cef1fe433 : SUCCESS in 6m 16s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/c973dad921e54e8584f02044307484d5 : SUCCESS in 10m 46s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/65ea978537a6407aa5cdabbf32ebfa61 : SUCCESS in 6m 22s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/e1e3a537b7a849f29d8335f0c89b8948 : SUCCESS in 10m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/65d1ef6e50e9481a90ebdd9a76c75663 : SUCCESS in 8m 32s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/a4d176d4589e496bbc779fd2ecb0be67 : SUCCESS in 8m 23s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/979b045046cc4d1c8abdf53f30199423 : SUCCESS in 14m 42s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/2300662544f44aa2ac20e934a6694b25 : SUCCESS in 4m 37s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/84624989b25d473ebbe58af62a0fbe87 : SUCCESS in 16m 15s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/654d923865c54d65a00f36c7f038a300 : SUCCESS in 6m 19s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/08283bd5c7de479fa8409a72ce21138d : SUCCESS in 25m 10s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/ed088b49c1bc4b64841534b5315c5828 : SUCCESS in 11m 44s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/776ecee7f83f4d129b45e07a4fd7985b : SUCCESS in 6m 37s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a646067d343f4030ad09a1a5ddee8245 : SUCCESS in 4m 43s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/da5fa8c60f6e4a299358932b3dc80c72 : SUCCESS in 24m 26s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/d879d63cee7342f9ac27088102954b26 : SUCCESS in 5m 29s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/eb9ca507462643d5ad19177792919b71 : SUCCESS in 5m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4eda0dc6761e43cfa1ddaf3dc92d00de : SUCCESS in 6m 45s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/01b87c62431a4d7ab2baafeca575361c : SUCCESS in 12m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6551560296144a98a65fd877f3321a12 : SUCCESS in 4m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c4f64645ece84b4da791a2d47667af74 : SUCCESS in 43m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f4affa47653d4240bad9822837eaa048 : SUCCESS in 1h 25m 03s","accounts_in_message":[],"_revision_number":2},{"id":"3e738ddb1845bd05aad5eb5acf32c2b574005ed6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 06:41:42.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a5e9b5bf8792499871a9bf5dda8197c02cc1a8e3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 08:07:34.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ff72e35ccc44116be38d01b7755c1a9 : SUCCESS in 5m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a114f614f6124488946c3e2ff9941125 : SUCCESS in 5m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7d1f8632e46447fcbf8b1934d46d5197 : SUCCESS in 5m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ed9d3d38a8f6490daa1fb281115185ed : SUCCESS in 5m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/79d12eb5b4e449e18d4d89cb9f0bdc99 : SUCCESS in 4m 26s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/18960644b3bc4c06a0f4db774b64c8d1 : SUCCESS in 3m 35s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b8bd4583006a442b96875276d3ed737a : SUCCESS in 10m 28s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2c50e96a4af94c8cb4984be1b711ca3d : SUCCESS in 14m 10s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/ba00c1f60221406e9b8bed71871fde6d : SUCCESS in 9m 57s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/5e874f79f4de4f00ab4c613b3c7b1dbd : SUCCESS in 6m 35s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/47403371a5644db1a56e828ab7b22d2b : SUCCESS in 8m 33s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/32a1131b02924f2694a1399e85ed623d : SUCCESS in 9m 11s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/db0d5ccd02eb480f817c77643c72231d : SUCCESS in 8m 54s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/e25e79f1b38541639d421232d3a92f2b : SUCCESS in 26m 36s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/ec18bc08bf2240f092ba1a50310f2160 : SUCCESS in 5m 18s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/41dd6440762d416b920ccabf55c07801 : SUCCESS in 18m 13s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/698086b0a44d4a74a73df6283c336b79 : SUCCESS in 5m 41s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/5040907ba5ef4789b423df528bf31a7f : SUCCESS in 24m 58s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/49487472630f4fc983d5facc89f0c643 : SUCCESS in 11m 41s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/03db4dfefbec4c33a6704823b3555c1b : SUCCESS in 7m 00s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c72a93c98dc1471490cfbdef97d0740c : SUCCESS in 4m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6cb9b236f0df40118204ca7c4ee8ad6a : SUCCESS in 19m 25s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/db511d3f9b8d4839b8a37d485daaedf1 : SUCCESS in 5m 48s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/37911dead798456b9c3673d20d6a22ee : SUCCESS in 5m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/57c5f17a7a014d9f88b05729356971b0 : SUCCESS in 6m 53s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/a1d271ea00054bdf81a8c8cc228c37af : SUCCESS in 12m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3ed9ac8988ec4fb298b5c487ae644d03 : SUCCESS in 5m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0b4215ccb27a4146a28755220a436c17 : SUCCESS in 39m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3877dc9d47434942941b7831889e0b4d : SUCCESS in 1h 22m 00s","accounts_in_message":[],"_revision_number":2},{"id":"8714da73bfed62681771c78228089d3a4bfc65e5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 08:07:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"4ef899db24e827ca812b5b81baff9dbf43804fad","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 08:09:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6abaf86bcdb94a09b5579816f9c96b01 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9cb6243b26fdd507a2425d4d82b0251420a76e42","revisions":{"f3b08fa788430c73ce0556146658a5f9bde36390":{"kind":"REWORK","_number":1,"created":"2021-12-08 02:52:01.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/74/820974/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/74/820974/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/820974/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/820974/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/820974/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/74/820974/1"}}},"commit":{"parents":[{"commit":"d329ca5f440340b6b90ad67373970b6a6d5b08e2","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d329ca5f440340b6b90ad67373970b6a6d5b08e2"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-12-08 02:52:00.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-12-08 02:52:00.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I0310989941d0b0fafe478369ebd8075134a89574\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f3b08fa788430c73ce0556146658a5f9bde36390"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f3b08fa788430c73ce0556146658a5f9bde36390"}]},"branch":"refs/heads/master"},"9cb6243b26fdd507a2425d4d82b0251420a76e42":{"kind":"REWORK","_number":2,"created":"2021-12-08 03:32:11.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/74/820974/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/74/820974/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/820974/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/820974/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/820974/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/74/820974/2"}}},"commit":{"parents":[{"commit":"d329ca5f440340b6b90ad67373970b6a6d5b08e2","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d329ca5f440340b6b90ad67373970b6a6d5b08e2"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-12-08 02:52:00.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2021-12-08 03:31:55.000000000","tz":-360},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I0310989941d0b0fafe478369ebd8075134a89574\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9cb6243b26fdd507a2425d4d82b0251420a76e42"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9cb6243b26fdd507a2425d4d82b0251420a76e42"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
