)]}'
{"id":"openstack%2Frequirements~821795","triplet_id":"openstack%2Frequirements~master~Ie14c48c76b9edc594facfa4c156b6e979bb9250f","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints","hashtags":[],"change_id":"Ie14c48c76b9edc594facfa4c156b6e979bb9250f","subject":"Updated from generate-constraints","status":"ABANDONED","created":"2021-12-15 02:48:44.000000000","updated":"2021-12-15 18:48:14.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"07f4119510cf052daf723a5216827647a13f7743","_number":821795,"virtual_id_number":821795,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-12-15 03:21:32.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-15 03:03:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-12-15 03:21:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ef4e76f4900ac6ec2650e8ca740010e0282693b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-12-15 02:48:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"079ecfa1c9437bb500d12fb747a94ba9d9dc8044","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-15 03:03:14.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/28a48b4f60d144f8a0436f8640ec782d : FAILURE in 8m 53s","accounts_in_message":[],"_revision_number":1},{"id":"3afc91696d5bd3d0ff13ead89aa9c6a26d3ed82e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-15 03:21:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24bd23b40d8c4e3ab682a8f3cc11169b : SUCCESS in 6m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0de1279c498f4634a126e59ba1d9bf7a : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fbafdfcb8a0041d4bccbcfa64aec819e : SUCCESS in 8m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b77bb39dc9324c34b0245a9f13296578 : SUCCESS in 6m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/09cf7e79c45c48c2a9b9a3ba2a4e379d : FAILURE in 5m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/22773310d5414384b1bef563265cb7f2 : SUCCESS in 6m 21s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ec6ecb8d363845108bb856f81f4b0ebd : FAILURE in 7m 12s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/10074369525f4ea8adf70887e142375f : FAILURE in 6m 03s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/e824231b263f46dc95865a02bc5408db : SUCCESS in 9m 06s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/648f9593ad5e4919bfe2fe0842374ba8 : SUCCESS in 12m 40s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/1e24725aa7974578891496c4bea45688 : SUCCESS in 8m 24s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/16cdf01376d34236be0875aa27d43310 : SUCCESS in 12m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1c71ab6fd0b14c4aba7146591528e6ca : FAILURE in 8m 01s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f3d01533cd6648b28b72d99236415e1a : SUCCESS in 11m 00s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/cda4e20c99f8484f97e0338390f3419e : FAILURE in 16m 18s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/a720230fbce745c187b8c6fec79ca71d : SUCCESS in 6m 41s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/ff77d0dc864648a58948c9452e3612ef : SUCCESS in 17m 00s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/b67cbcb8f765424eac577fe2af8abaaa : SUCCESS in 7m 05s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/27d1646d0b4f41cf8b43da0e657c5507 : SUCCESS in 24m 03s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/a86530b23e944388b96994de02b86c27 : FAILURE in 13m 23s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/1ff34c4de5a24f9d9818b6d705e23c77 : SUCCESS in 8m 04s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0dd4161e1f2e4601a2bd7c4e06e9e169 : SUCCESS in 6m 32s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9255a68d64d245d0907ecc32d0f16f5c : FAILURE in 22m 01s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/c1866e4444eb4ece8202227c44cd0d92 : SUCCESS in 8m 01s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/6b11cea089ae471fa836885b6b4e5bd6 : FAILURE in 5m 47s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4dc4398ff3e74bf5aa296000f18d3384 : SUCCESS in 8m 47s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/ec3021df8ec74c1dbb7da40585715d35 : SUCCESS in 14m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/030fb15a70e246889f5b87f89baee124 : POST_FAILURE in 21m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/df3e28ac20e04306935d7fa098d2944b : FAILURE in 18m 21s","accounts_in_message":[],"_revision_number":1},{"id":"07f4119510cf052daf723a5216827647a13f7743","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-12-15 18:48:14.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"52b03976ee9f1969cce0d63f60d01835759ee396","revisions":{"52b03976ee9f1969cce0d63f60d01835759ee396":{"kind":"REWORK","_number":1,"created":"2021-12-15 02:48:44.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/95/821795/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/95/821795/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/821795/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/821795/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/821795/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/95/821795/1"}}},"commit":{"parents":[{"commit":"e74f91e250d127ff825dad5219527d62efaeb740","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e74f91e250d127ff825dad5219527d62efaeb740"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-12-15 02:48:43.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-12-15 02:48:43.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ie14c48c76b9edc594facfa4c156b6e979bb9250f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/52b03976ee9f1969cce0d63f60d01835759ee396"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/52b03976ee9f1969cce0d63f60d01835759ee396"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
