)]}'
{"id":"openstack%2Frequirements~821859","triplet_id":"openstack%2Frequirements~master~If1354895383e2e98fda99006df662a69e9a7e4ee","project":"openstack/requirements","branch":"master","topic":"new-release","attention_set":{},"removed_from_attention_set":{"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2021-12-15 18:46:25.000000000","reason":"Change was submitted"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2021-12-15 18:46:25.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If1354895383e2e98fda99006df662a69e9a7e4ee","subject":"update constraint for neutron-lib to new release 2.18.0","status":"MERGED","created":"2021-12-15 15:49:36.000000000","updated":"2021-12-15 18:48:21.000000000","submitted":"2021-12-15 18:46:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"821859-new-release","meta_rev_id":"a8f241181f93d56833c74d435f51b37268f301ea","_number":821859,"virtual_id_number":821859,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-15 18:46:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2021-12-15 15:51:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2021-12-15 17:04:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-12-15 16:18:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2021-12-15 17:04:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-15 15:51:50.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-12-15 16:08:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-12-15 16:18:23.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-12-15 16:37:28.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-12-15 16:37:28.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2021-12-15 17:04:51.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2021-12-15 17:13:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a72ff7f88227f319c2f89574d0f26ad2bcf76004","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-12-15 15:49:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"92df8f0b92594cee928319d4813d321ec05e9ccb","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-12-15 15:51:50.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"46fc6500896ee5dd4909b5c117b044f1b4adaa28","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-15 16:08:04.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/4ff6e8445fcd4d75b9bfc3f1f70d00b8 : FAILURE in 12m 11s","accounts_in_message":[],"_revision_number":1},{"id":"900f13c1e897e04136d24cf79ac60566785fd328","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-12-15 16:18:23.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d1e541b12699826137ac5362f94528549e8e0d60","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-12-15 17:04:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7a6607e57ed275f1a695b94c3b2811f90bfbf5e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-15 17:13:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/425fbd4ed800473aad208dc0bb80a469 : SUCCESS in 12m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71aef7eda0b645d49c0e7d374877bc50 : SUCCESS in 4m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/665dca9c3beb41199d45f001a03a5879 : SUCCESS in 4m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d2492adb6c0e432aaac520a663ba9935 : SUCCESS in 4m 14s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b83f488054f849a6a8e193145f664f01 : SUCCESS in 4m 28s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2cd51bf2459e4511b001584bf2b0d69d : SUCCESS in 3m 56s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ec74b1a29a604740b4d78241abb1de35 : SUCCESS in 10m 36s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1cf9b2b93ccd4228965d96a7fcecf214 : SUCCESS in 10m 49s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/85e1ccc7e92f4fffa6a53879379c9d30 : SUCCESS in 7m 02s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/9fc5a9d687984efc9f9ff82b8ffc6b05 : SUCCESS in 11m 49s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/20f3a43c35bd4253b2f0c5fccddf389a : SUCCESS in 6m 39s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/71f61f0b682547f5aa1568166a3177ae : SUCCESS in 11m 14s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c7705dec1307447291e22455b311dfc3 : SUCCESS in 8m 27s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f89cf9d18f0042398aec768e74f3b4a1 : SUCCESS in 8m 46s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/07a56310218140eabf23bf5f02b0e9b0 : SUCCESS in 34m 27s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/437c9d2cd2d447288b5a5c885882d399 : SUCCESS in 5m 32s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/99dc4f4af9dc499cabb0f02f3442925f : SUCCESS in 16m 33s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/447bb0c4c8144d788351bdee1c94c5aa : SUCCESS in 5m 44s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/3b6f2b9351084cc687058bed73861975 : SUCCESS in 24m 22s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/60e638079f144649a0ca3cca225670b3 : SUCCESS in 16m 18s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/3462c673fc1a4677b63f45ec4c01c180 : SUCCESS in 6m 28s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/81a46cdaa4444fbe967877f04ff1fb75 : SUCCESS in 4m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ac0877daddf149c4a31ccbcbe3fd9306 : SUCCESS in 19m 58s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/3c16d89d896244bcb90de5cfed3fd900 : SUCCESS in 7m 04s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/da768873fbd24cd2b6f89e019ba8f7ee : SUCCESS in 6m 59s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/96b765705f91461284a1765c31fe552f : SUCCESS in 6m 48s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/0d168d85d10845318f014c3b225b89f2 : SUCCESS in 13m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2bad8f7dc69d4e0583b656707eb31d2d : SUCCESS in 48m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d1a9937e82a64938947bfb090d94f9c6 : SUCCESS in 1h 19m 00s","accounts_in_message":[],"_revision_number":1},{"id":"1812ca0bc970c3fecde45226fd450af705b7404c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-15 17:14:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"52824718006f7bf79c440cc0441e25e3e81017d5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-15 18:46:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc5bdd9e3f974125a597fa7028ec93a9 : SUCCESS in 5m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2e74cfd2e594b5e9ec0295a681eb07b : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fda597e0a8b248d3bf88a44812a913b6 : SUCCESS in 4m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/96b3b43b164f4425bb46b7231bf8c931 : SUCCESS in 4m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7e1e1f32ee624acdb43a57495c2838d8 : SUCCESS in 4m 47s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/17b67ea0f9ee47fe953a6dc4fadb7857 : SUCCESS in 4m 52s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ee7a94ee1b5845f4839f6e359a5af0be : SUCCESS in 10m 01s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a13cf25b2fc3433d9c031c65ba12e3c5 : SUCCESS in 10m 00s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/92a5bf2ea4d445fda1e600432898402c : SUCCESS in 11m 37s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/2798256f6251420f97ec5594a271e804 : SUCCESS in 6m 31s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/38f93b612b364981a60600b270c6f179 : SUCCESS in 10m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/15f5b0ec6a8b451faf7a291147f5cd3c : SUCCESS in 9m 40s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/91ef4fc7a20f4548b7034046fb24e0f9 : SUCCESS in 9m 31s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/b80d447da49c492d84cd82a26179c075 : SUCCESS in 13m 39s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/501db9313df84469b9759657fe1702f1 : SUCCESS in 5m 33s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/bc103ffb97f54a08bc37cebcfcd6b8fb : SUCCESS in 17m 00s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/5a3ca4cae73a49feb84ada2f43afb119 : SUCCESS in 8m 48s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/bb22a3cd0e874573a1f2df7cc61efdeb : SUCCESS in 24m 33s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/e517f227bc464b51a9ee6d11a01828b0 : SUCCESS in 15m 17s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/1f4629caaf0f4a49b9ce978e142c5810 : SUCCESS in 7m 01s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a702819a883b4e33adcf2ac8d9e4f3a6 : SUCCESS in 4m 45s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a1869401c47848e09a2cdb8b3e96ce6f : SUCCESS in 19m 29s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/215a3ffb7fe64216be6dbe2ac1d6e156 : SUCCESS in 7m 54s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/51b5b2de922948d58f5506cc430abdde : SUCCESS in 6m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5443d573f10a4c2ea6d5f81855a35764 : SUCCESS in 8m 52s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/c72e5883b5d64cff9986259d006ee5df : SUCCESS in 14m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7eab1c9f5bf947abab3d2ec94cb1519a : SUCCESS in 56m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1778d2b936a540c79052b6f5faf3fc3d : SUCCESS in 1h 24m 22s","accounts_in_message":[],"_revision_number":1},{"id":"cfd5ebf43c71f55b10dc9d03f294ce1605737dac","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-15 18:46:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a8f241181f93d56833c74d435f51b37268f301ea","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-15 18:48:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a32ebc8bacc4423696a84176748ac567 : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"134bfe059a4f78a51f95342f5598c0c7f8fe2a40","revisions":{"134bfe059a4f78a51f95342f5598c0c7f8fe2a40":{"kind":"REWORK","_number":1,"created":"2021-12-15 15:49:36.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/59/821859/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/59/821859/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/821859/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/821859/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/821859/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/59/821859/1"}}},"commit":{"parents":[{"commit":"e74f91e250d127ff825dad5219527d62efaeb740","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e74f91e250d127ff825dad5219527d62efaeb740"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-12-15 15:49:35.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-12-15 15:49:35.000000000","tz":0},"subject":"update constraint for neutron-lib to new release 2.18.0","message":"update constraint for neutron-lib to new release 2.18.0\n\nmeta: version: 2.18.0\nmeta: diff-start: -\nmeta: series: yoga\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nmeta: release:Commit: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nmeta: release:Change-Id: I554ef91c88636df51baad1efb1d0effcdd146d81\nmeta: release:Code-Review+1: Lajos Katona \u003ckatonalala@gmail.com\u003e\nmeta: release:Code-Review+1: yatin \u003cykarel@redhat.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+1: Oleg Bondarev \u003coleg.bondarev@huawei.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: If1354895383e2e98fda99006df662a69e9a7e4ee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/134bfe059a4f78a51f95342f5598c0c7f8fe2a40"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/134bfe059a4f78a51f95342f5598c0c7f8fe2a40"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
