)]}'
{"id":"openstack%2Frequirements~821972","triplet_id":"openstack%2Frequirements~stable%2Ftrain~I530461a3bc23de4c021eadc00e1c1b851b0cefa4","project":"openstack/requirements","branch":"stable/train","topic":"bug/1955010","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2021-12-21 10:54:09.000000000","reason":"Change was submitted"},"14288":{"account":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"last_update":"2021-12-20 15:19:16.000000000","reason":"removed on reply"},"21798":{"account":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"last_update":"2021-12-17 12:23:56.000000000","reason":"Bernard Cafarelli replied on the change"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2021-12-20 15:40:34.000000000","reason":"Slawek Kaplonski replied on the change"},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2021-12-21 10:54:09.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I530461a3bc23de4c021eadc00e1c1b851b0cefa4","subject":"Pin \"lazy-object-proxy\" \u003c 1.6.0 on Python 2.7","status":"MERGED","created":"2021-12-16 12:02:08.000000000","updated":"2021-12-21 10:54:09.000000000","submitted":"2021-12-21 10:54:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"821972-bug/1955010","meta_rev_id":"00033a05f52b7e5a6b0c47efb8d572668774e168","_number":821972,"virtual_id_number":821972,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-21 10:54:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-12-20 16:18:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2021-12-20 12:01:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2021-12-17 12:23:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2021-12-20 15:40:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-12-20 16:18:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-16 13:42:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-17 10:32:26.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2021-12-17 10:32:26.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-12-20 11:33:27.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-12-20 12:01:33.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2021-12-20 12:01:33.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"317a15b593ff60b29af49a211bda9fce2d6e6621","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-12-16 12:02:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"238f27788e150e1829fdb01dac703902f996909b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-12-16 12:22:22.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"ac74b078f09d271724c5639fe479e0442f9d3efd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-16 13:42:19.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0da50884aaa04f16a4ac681863e43659 : SUCCESS in 4m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7296dc461b3a4c1a8dc5f0862002ca04 : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/abed2163ea474bff805b788a80106249 : SUCCESS in 4m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8747769937bc486da0f3536d24db2f56 : SUCCESS in 4m 52s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fa446907bb204c599867d9dd25c80901 : SUCCESS in 3m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/08b521e0d8a94fddb6218bf13cb2c734 : SUCCESS in 3m 50s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/c7f84e96bec24e38b4d35a5493ab8621 : SUCCESS in 8m 54s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c37c5ffc32974f7ba83d0fd6c0138708 : SUCCESS in 9m 00s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/0587504739ca4f5ebb0dadb00acc4e99 : SUCCESS in 10m 09s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d46ac282a72649ecbb0cf3f3c0fd942c : SUCCESS in 9m 47s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/4f3dab95b14c447cbeca75ef4ba82bd2 : SUCCESS in 5m 01s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/3afaa4dab22649a2bdfcfc5f4cc8164c : SUCCESS in 5m 14s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/1926467ba284404cb494a9eabb35596d : SUCCESS in 10m 42s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/44d310fb688a46f984f0de838a9c5918 : SUCCESS in 9m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/26e569df80ea4c0ca96b531d456e652e : SUCCESS in 11m 07s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/deeaa4f291a94b6c834ce2d3ab05acfd : SUCCESS in 13m 35s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/8c0c3d05ba0f46fea598e3f1fd758f75 : SUCCESS in 12m 27s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/62cca352b19245e4a44055f53b539190 : SUCCESS in 25m 31s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d3dd162b42ec427e9e643eaab2c63785 : SUCCESS in 21m 32s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/5be5bd5be5be49ce9c245b075cf95ac7 : SUCCESS in 12m 51s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/2709818570f9408c89747920d749e09a : SUCCESS in 14m 01s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/75be74cea70b4ca8a7fa365d98104106 : SUCCESS in 17m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b9635403d1224c32b9c63105264ce56c : SUCCESS in 20m 14s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c67b14c55bd64e5c8eb713a9bdc108af : SUCCESS in 5m 43s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/261a983e675440c2b6dcc191d387e2f3 : FAILURE in 9m 50s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/48da6fb18af84cc1bb35b6704aa93e3b : FAILURE in 12m 16s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a1c6e3d5102740f4ac7f3bb2e7ca4f4c : SUCCESS in 54m 03s\n- tempest-full https://zuul.opendev.org/t/openstack/build/67a9346f287a4229b58451dfd02288c3 : SUCCESS in 1h 17m 04s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/130a598300ac463e8c8a2098a1c9eeda : SUCCESS in 50m 13s","accounts_in_message":[],"_revision_number":2},{"id":"944686423373c95b3daeb05074d4fc7efa9824e4","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-12-16 14:46:58.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4be5a9fc010218e0e4d1ea85addf606e53da7ef8","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-12-16 18:09:07.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f36e1bddb8d69103ea4aace83373d66838e903b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-16 19:33:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d65ad899d604e50b3443c73aefb4983 : SUCCESS in 3m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/dbe360378e0a484fb0fe767bc3908a56 : SUCCESS in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/21cf9c1392594e71b402bd993e7b2d11 : SUCCESS in 4m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/98316c50f6034dfb91787787c7b63420 : SUCCESS in 4m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e9fa17b6e68049af9edbacf237b07328 : SUCCESS in 3m 51s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8574d1d1b26b494dae347f1c09639b2b : SUCCESS in 7m 54s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/7dfe8989f58849a889e181d151301c19 : SUCCESS in 11m 12s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9cfe8b9714224894a17ef1190c28367f : SUCCESS in 10m 24s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/47570f31fb7f47ab9fe51502e5535424 : SUCCESS in 9m 11s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/80b69485042749ec84c0b631308ea634 : SUCCESS in 9m 33s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/2d9685d9db3e4fb7a6cd69e18569dbc9 : SUCCESS in 5m 04s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/17491e553cbc4a659b40ccadf6c68e8c : SUCCESS in 5m 24s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/0a838ddbf7fc44e98b7c40c4e5c2d106 : SUCCESS in 10m 55s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/8e77826fd497450fb365e6887cd4bcf3 : SUCCESS in 9m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1ee38c23be2b4a87a0fba1382fc5c98d : SUCCESS in 11m 17s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/c2628bbeac8140a5a6a60396b105aca3 : SUCCESS in 15m 08s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/9fd6f6fa9a83425a8b7d322ccda4fb24 : SUCCESS in 13m 19s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/e4c25520307c44d3a18100e47af593bc : SUCCESS in 18m 50s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/2ed571e26aa440b2ade19acbae0bc4e5 : SUCCESS in 20m 31s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/47fb6d20b1f14a39bd85c7573b7458f3 : SUCCESS in 16m 17s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/eaa0559bd486449f8f047d546c0e71c7 : SUCCESS in 11m 33s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/4f673d9252b4438a9d817b0d28058567 : SUCCESS in 13m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d9d24440d37d431bab4bfed2bf29e42d : SUCCESS in 18m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/667942a836f04ce6bc606841b9704cb2 : SUCCESS in 5m 34s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/44f71dd3b204482ca0eb868e3d1b6d11 : SUCCESS in 11m 14s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/9cfec00530784568886e9e090c42b707 : SUCCESS in 11m 47s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/72b662c7ed3643cd93fb625a02c2c91b : SUCCESS in 57m 31s\n- tempest-full https://zuul.opendev.org/t/openstack/build/71cca99e69534e94a50898150e79c072 : SUCCESS in 1h 20m 52s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/dc80ff2990464220b0c605412aac5c7d : SUCCESS in 1h 02m 36s","accounts_in_message":[],"_revision_number":2},{"id":"aa0b196454bcf7f1eeb56a5a6aa60f96f3a83837","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-12-17 12:23:56.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"e8bb288030f62ab4808897ead9d836051752acb4","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-12-20 12:01:33.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"34b8849a322cd1cda45040790cccb2781ab5bdee","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-12-20 15:19:16.000000000","message":"Patch Set 2:\n\nwhat is this impacting?  iirc openstack considers train eol now (not extended maint)","accounts_in_message":[],"_revision_number":2},{"id":"82d34773d7fb160aa12e57ecae103f331e3950ad","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-12-20 15:40:34.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3d2de3b27896259ecc5a7baae313ce070abe169b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-12-20 16:18:03.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nheh, misread the releases page (wrong column), thanks for the pointer","accounts_in_message":[],"_revision_number":2},{"id":"f055ace2b7ec0f5f59ec825ed2a905082b1dec87","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-20 16:19:41.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ca25e215fb0ed1f85c85d46133796daf3cac569c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-20 19:32:18.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dbbd8465b7804aac8c29c4d817faa7c2 : SUCCESS in 4m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/48af3674ba454e2095ed2be5df2aeba5 : SUCCESS in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3f0edb8fd8cd4f32a00bf0b9d26eef45 : SUCCESS in 5m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7bc33d879c06447896850f938219b6d9 : SUCCESS in 4m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/63431de965624283bf3e42577725c22d : SUCCESS in 6m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/db25cae88c9d45769defa9375e5a8a3f : SUCCESS in 4m 27s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/5dcdeaf1f4924c9f89e4f0591f63b482 : SUCCESS in 8m 59s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2247e151d29f49f99e98120ffdf88a08 : SUCCESS in 8m 16s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/450017f0d644471db87720ffe80b43ab : SUCCESS in 8m 34s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d6075a6d77da44b1aee94ce587d56cae : SUCCESS in 9m 06s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/e69968110b2149aea35e9420e6d31eb7 : SUCCESS in 7m 36s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/8611145606f94f6bbff1eba3205a28f8 : SUCCESS in 5m 40s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/f719c10c816b44f9b225ea68cc168f9d : SUCCESS in 9m 26s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/223db9860fc7482eb33d3da9c8791ac1 : SUCCESS in 9m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2a60fc50fc6d4aed92aa86b0f1a50b47 : FAILURE in 11m 15s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/4cb9356eb7e34561822dd8e6450f5921 : SUCCESS in 12m 02s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/43b5445f58c74a0889fad6b8aca7677a : SUCCESS in 15m 26s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/e0897b53beb54d108bceda3d2ec26f7d : SUCCESS in 20m 03s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/6056a7bedd594d78858ca0025f62d158 : SUCCESS in 17m 29s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/04206bcf76654006921c2eb42497f089 : SUCCESS in 17m 37s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/bd5528f2627247a1b782472b2e3b328e : SUCCESS in 13m 52s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/46544b44248d4fc291e8be364802ee25 : SUCCESS in 17m 43s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d5f6bc689e1541ee89e4793bd3217f88 : SUCCESS in 18m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/709021283956487a9add05bf3b8263a4 : SUCCESS in 6m 45s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/beed682af9c142df9176c8a094093981 : SUCCESS in 14m 00s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/570127c8d3e24c2bbe682119f242230f : SUCCESS in 14m 11s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/703cfdec225e4beba3841ff1b655da5a : SUCCESS in 58m 27s\n- tempest-full https://zuul.opendev.org/t/openstack/build/624ea7a49ca341c399af5bc745763e9a : SUCCESS in 1h 21m 15s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/6159662f4ebe44e4920614cf40c09bb7 : SUCCESS in 44m 20s","accounts_in_message":[],"_revision_number":2},{"id":"931e2efa491b775e6f0c64289c2d77eab165aae1","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-12-20 19:38:06.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"f3df85fa2c21b0dec4ba227e8ffef1a5d0c882ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-20 21:21:09.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb7db6964bd74ea292ae7a567aabaf27 : SUCCESS in 4m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fda4039df1834236b05836b3ca0b3d53 : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5415680a4ec9444989c9ef6518fd7ee2 : SUCCESS in 5m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9603197f28324110a47d3855f02a34c3 : SUCCESS in 4m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/be8b0a327521452190cb8b719cec4421 : SUCCESS in 3m 50s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9045b8a373dc4a2596fa09ad2285dc6b : SUCCESS in 3m 40s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/aa88e296b4994298a63819f0b59d81bf : SUCCESS in 7m 13s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6fabbeb38423408e8fa18c186c211808 : SUCCESS in 7m 36s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/00890ba133e04f63a579b62cb36790bd : SUCCESS in 12m 49s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/2e2e9a40190f460c8084e83f4aacc7f9 : SUCCESS in 8m 50s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/b68793cf1ccd4df99293ad3a485840ed : SUCCESS in 7m 00s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/eabd081ab0f649b88c2504c732b0c166 : SUCCESS in 4m 44s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/98d3d1170a084baa90d60fbfa70a5356 : SUCCESS in 9m 24s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c3c8c3a900c14974a1255aec32531f51 : SUCCESS in 8m 19s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0e6d6107b0654860ba3cc855b30ccdc2 : FAILURE in 13m 00s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/6c09137666da40a1b8f70dfdddabe8a8 : SUCCESS in 12m 49s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/8c935ad6eee34fb3ae3f3e06f66f07b0 : SUCCESS in 12m 20s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/5d2bc6f8173940f7b389c75baa6250c6 : SUCCESS in 24m 28s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/bce343896d2e456b914e663454ee6f19 : SUCCESS in 19m 40s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/482a218e0e144918a5f872b405d98790 : SUCCESS in 14m 06s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/840df6c6caf848f09d15063c6d532f63 : SUCCESS in 12m 49s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/526ab8c18e82458e9bc026cd9bf5c64a : SUCCESS in 14m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7c1c54f100f04687bd2105a7ba568ccd : SUCCESS in 18m 52s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/78756a224674452f960dd8ceb5e3b92a : SUCCESS in 5m 20s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/02761c742af648ac93d442a7cf94d4fa : SUCCESS in 10m 10s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/9dace4e7c248487d8a98ee8cdaca511e : SUCCESS in 12m 06s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/affa2fce8b3845faa387516152d9e3b7 : SUCCESS in 1h 00m 20s\n- tempest-full https://zuul.opendev.org/t/openstack/build/1b4bb072751b473a875d3d2dd7b8a3a4 : SUCCESS in 1h 29m 29s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/ee53588461624e91b6d0d89ab3bb1e07 : SUCCESS in 45m 49s","accounts_in_message":[],"_revision_number":2},{"id":"c0c03c8f693424cf88b2fa37fd283b1896ae39ba","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-12-21 08:09:02.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e5b0a60fbb63d6516501104c0d0a05d600753d8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-21 09:20:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/235c0cbfc53b4440a26b639e47eaa246 : SUCCESS in 4m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/101f051bff144a2ab37dd32c94f6ef93 : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3ae1142e9d194aae9b29d45628232a36 : SUCCESS in 5m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/05a93d6a190e4bb4b9c84e882ee4452c : SUCCESS in 4m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cdbab72b03da42eda5e9e35b197a9b75 : SUCCESS in 3m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b7b48443d10e49faa52dbda5bdd5e66b : SUCCESS in 4m 49s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/5af497b8e5de457e9808e6bc2f5a102e : SUCCESS in 11m 34s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3499749e11e7401c9c4ca70a49aa02c7 : SUCCESS in 8m 32s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/d41128b10a314521b3e2e6fc5fb79e21 : SUCCESS in 9m 41s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/7fdec3b5e4764ceda7b469d64125047d : SUCCESS in 8m 30s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/381d642d285043418ecf923f4b03f63d : SUCCESS in 4m 59s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/7fef92dcbf7f45eca8f544fe98297efa : SUCCESS in 5m 00s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/52cbbbc47d484a5597a7f19200473854 : SUCCESS in 10m 02s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3207620be7be45eaadf633fa48659fb5 : SUCCESS in 7m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5ff1c8c74fa7478b8cb95bac56d0b89c : SUCCESS in 14m 05s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/9a50bc4db82b40f79df651744de3adb9 : SUCCESS in 10m 00s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/65ebd8f09efb4d9ea0751fb684aabb96 : SUCCESS in 14m 08s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/c8f575eb44bc4c37a09d857773eb48e7 : SUCCESS in 21m 11s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/580dfce1dd614b3c82e1127bcbf2411e : SUCCESS in 17m 22s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/2926ecb7a7a54fdd9cdc0d85f4d778de : SUCCESS in 13m 56s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/46ea52ae2c4a4db1b347d0f8b43624a3 : SUCCESS in 12m 15s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/49e9bb82061c4139a126c7dc1dba7185 : SUCCESS in 11m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c580b44233304264bbfb86388fc9bdf4 : SUCCESS in 18m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3c112e2579524871a59eb5f4e9aae51b : SUCCESS in 5m 16s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/5d1ab184f39c454c868131992ad874df : SUCCESS in 9m 44s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/22ca4442dde842f3b3ff143cef7c186a : SUCCESS in 11m 18s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/57057649904b467eba27d0f0fdf89cd9 : SUCCESS in 48m 52s\n- tempest-full https://zuul.opendev.org/t/openstack/build/6edc509517c849c99735bb48ab8ff241 : SUCCESS in 1h 08m 55s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/f30eec30be8d43f0bad99803b860a58f : SUCCESS in 44m 26s","accounts_in_message":[],"_revision_number":2},{"id":"9a230ef7931eba603c41c90ac366fd64e25f667c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-21 09:22:01.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"87379f4e6c884f3c2db96e4512142eee0218c0e8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-21 10:54:09.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17c317664cf74f929811a5bbdf47929e : SUCCESS in 4m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/879d0669bcd6400c83832da3d2614153 : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b5bd99540495478dbd65bce3a19f0e1b : SUCCESS in 5m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/923b9ecb1cd4471b936de7e2303ba2fe : SUCCESS in 5m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e4f3838b546f4fb88b73801b39379810 : SUCCESS in 5m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/70c035f24eb847e3a2b4e84f703e1c51 : SUCCESS in 4m 03s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/f66f8da5325941958cb8a565714039b4 : SUCCESS in 7m 43s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/326a9526996a4bd5a54bfe24fce1642c : SUCCESS in 8m 08s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/d97b1d9255b84847a5ea8c189a2af9e8 : SUCCESS in 9m 56s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/f8d8a3acb4d1440b879b8b19dee59d0a : SUCCESS in 9m 11s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/70842a1f2e524ad0bdaffd8728f4c970 : SUCCESS in 5m 00s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/87dac68790bd4c13a39130af3535dfb8 : SUCCESS in 5m 11s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/a7f17fe61fba4ff0b4583ebb838f6e27 : SUCCESS in 10m 32s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/aea0c9494e964fee8e916d286dc33576 : SUCCESS in 9m 39s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0e99c6ce8fb549bda78b58e42be2e38b : SUCCESS in 9m 28s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/f595a40129784753af0543a6e9381d06 : SUCCESS in 13m 22s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/339917b144a64b4083491b8078c23495 : SUCCESS in 13m 59s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/5a374b3a999343fbaaec6a9867d75ca8 : SUCCESS in 20m 22s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/fd267fd6547a4b9daf9d8986cbec96dc : SUCCESS in 21m 20s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/b7708bd0dcbf40e8bd304a8796b4b9ee : SUCCESS in 17m 00s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/cd75b6b09bfd471badb3b5d5f0eb8406 : SUCCESS in 11m 57s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/933d018bc1d346609b52233a2085babf : SUCCESS in 13m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/026d7fc0e10948e18bcaa5d3e224d418 : SUCCESS in 16m 59s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/49003ec0eb874cc485968c4f716253c2 : SUCCESS in 5m 45s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/96a2fa5700ad45bd88ace6c37b75d903 : SUCCESS in 9m 04s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e384fb15d8a746709489a7785bb2506d : SUCCESS in 10m 12s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/55ec9d54fdff46ad8e4815322d14235c : SUCCESS in 44m 25s\n- tempest-full https://zuul.opendev.org/t/openstack/build/b6b8c7c0d09f4b2aad57dcf97856b368 : SUCCESS in 1h 29m 53s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/cafe33700efc4663b32779b419e3a086 : SUCCESS in 1h 01m 27s","accounts_in_message":[],"_revision_number":2},{"id":"00033a05f52b7e5a6b0c47efb8d572668774e168","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-21 10:54:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9c6194159c30c757ddb23297dd0cbb271d05f9ce","revisions":{"35269e2e63af57e9a076110fa11717349238b709":{"kind":"REWORK","_number":1,"created":"2021-12-16 12:02:08.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/72/821972/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/72/821972/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/821972/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/821972/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/821972/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/72/821972/1"}}},"commit":{"parents":[{"commit":"9874f188560ba932c7fadc185caf8da3ed75b17f","subject":"update constraint for keystoneauth1 to new release 3.17.4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9874f188560ba932c7fadc185caf8da3ed75b17f"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-12-16 12:01:40.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-12-16 12:01:40.000000000","tz":0},"subject":"Pin \"lazy-object-proxy\" \u003c 1.6.0 on Python 2.7","message":"Pin \"lazy-object-proxy\" \u003c 1.6.0 on Python 2.7\n\nThe library lazy-object-proxy has dropped support for Python 2.7\n[1][2] in version 1.7.0.\n\n[1]https://github.com/ionelmc/python-lazy-object-proxy/issues/61\n[2]https://github.com/ionelmc/python-lazy-object-proxy/commit/16db739035116ed74af16aff4a8f2cfeab6fce80\n\nCloses-Bug: #1955010\nChange-Id: I530461a3bc23de4c021eadc00e1c1b851b0cefa4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/35269e2e63af57e9a076110fa11717349238b709"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/35269e2e63af57e9a076110fa11717349238b709"}]},"branch":"refs/heads/stable/train"},"9c6194159c30c757ddb23297dd0cbb271d05f9ce":{"kind":"REWORK","_number":2,"created":"2021-12-16 12:22:22.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/72/821972/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/72/821972/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/821972/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/821972/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/821972/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/72/821972/2"}}},"commit":{"parents":[{"commit":"9874f188560ba932c7fadc185caf8da3ed75b17f","subject":"update constraint for keystoneauth1 to new release 3.17.4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9874f188560ba932c7fadc185caf8da3ed75b17f"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2021-12-16 12:01:40.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2021-12-16 12:22:22.000000000","tz":0},"subject":"Pin \"lazy-object-proxy\" \u003c 1.6.0 on Python 2.7","message":"Pin \"lazy-object-proxy\" \u003c 1.6.0 on Python 2.7\n\nThe library lazy-object-proxy has dropped support for Python 2.7\n[1][2] in version 1.7.0.\n\n[1]https://github.com/ionelmc/python-lazy-object-proxy/issues/61\n[2]https://github.com/ionelmc/python-lazy-object-proxy/commit/16db739035116ed74af16aff4a8f2cfeab6fce80\n\nCloses-Bug: #1955010\nChange-Id: I530461a3bc23de4c021eadc00e1c1b851b0cefa4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9c6194159c30c757ddb23297dd0cbb271d05f9ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9c6194159c30c757ddb23297dd0cbb271d05f9ce"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
