)]}'
{"id":"openstack%2Frequirements~821992","triplet_id":"openstack%2Frequirements~stable%2Fussuri~I6efffb447d998fd1d7edb65bfa689ed435f85602","project":"openstack/requirements","branch":"stable/ussuri","topic":"use_2to3","attention_set":{},"removed_from_attention_set":{"3153":{"account":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"last_update":"2021-12-16 20:13:23.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6efffb447d998fd1d7edb65bfa689ed435f85602","subject":"Cap setuptools to \u003c58.0.0","status":"MERGED","created":"2021-12-16 15:39:24.000000000","updated":"2021-12-16 20:13:23.000000000","submitted":"2021-12-16 20:13:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"821992-use_2to3","meta_rev_id":"67c9cd5b05db9e1441cb3a64059c14507ab3b3ab","_number":821992,"virtual_id_number":821992,"owner":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-16 20:13:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-12-16 18:45:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-12-16 18:45:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-16 18:14:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-16 18:45:08.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"f4d03611545cbc2bb77c3a0e5c5f676a27889c94","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2021-12-16 15:39:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0c7f950be0082224569d2c67d805cb453039152b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2021-12-16 16:07:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1a78a45107d7c71cd76436a84ecd16de581967b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-16 18:14:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c54f1c3a14e2485f9e2eace8a25ee26c : SUCCESS in 5m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fc894d689515439cafa56139faa2665d : SUCCESS in 5m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a9327e9ab239496085e08bd9f9977a42 : SUCCESS in 4m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3f6c254bcf1843c1bdbafb4b321c5045 : SUCCESS in 4m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/33799ff5bbf849db9b7dd706671b5363 : SUCCESS in 47m 36s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/91a96434627043d88854a4e607d8da6b : SUCCESS in 1h 13m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d08785b862ea4ccf98511558a991cf15 : SUCCESS in 2h 02m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/30550c3d7e294634954297becd1eee49 : SUCCESS in 1h 09m 02s","accounts_in_message":[],"_revision_number":2},{"id":"d3cfd6c394b3836b69a14cb7b0b967bc8c984250","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-12-16 18:45:08.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8b3d77cda68b9e7cc355b71b7d797e230afcdd66","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-16 18:45:50.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8e591ae420031445bfd0f9dbe2769f56c4adb92f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-16 20:13:22.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45f45495b5e846028474613e88fb89e5 : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/faff64f4b1d14036aee166c466d4b30f : SUCCESS in 5m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c1671b9a8ecb4397a2f9610fe2d6d5d2 : SUCCESS in 5m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e7e88ec23232456cafa2a10176d83654 : SUCCESS in 5m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7ebba941f40743acba9c5a46d673aac4 : SUCCESS in 51m 29s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/bc37e1e79596451890bd5e0f162f1872 : SUCCESS in 1h 04m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fe0aa3f3793b47b8b07abf56cf735ccf : SUCCESS in 1h 27m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/ce27d9819c6c48f4892d1514591a3279 : SUCCESS in 1h 11m 56s","accounts_in_message":[],"_revision_number":2},{"id":"67c9cd5b05db9e1441cb3a64059c14507ab3b3ab","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-16 20:13:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6d3c282da968938ef234546797a4f4dfee2aca8f","revisions":{"a4cb8d239d1b95bacfb65611ab208c8773fdda8a":{"kind":"REWORK","_number":1,"created":"2021-12-16 15:39:24.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/92/821992/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/92/821992/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/821992/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/821992/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/821992/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/92/821992/1"}}},"commit":{"parents":[{"commit":"84831a3287aa98a7aa640e36b0c69a03d0dc7534","subject":"[stable-only] pin virtualenv","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/84831a3287aa98a7aa640e36b0c69a03d0dc7534"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2021-12-16 15:38:45.000000000","tz":-300},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2021-12-16 15:38:45.000000000","tz":-300},"subject":"Cap setuptools to \u003c58.0.0","message":"Cap setuptools to \u003c58.0.0\n\nSetuptools 58.0 (bundled in virtualenv 20.8) breaks the installation of\ndecorator 3.4.0 which is still being used in Ussuri and Train.\n\nChange-Id: I6efffb447d998fd1d7edb65bfa689ed435f85602\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a4cb8d239d1b95bacfb65611ab208c8773fdda8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a4cb8d239d1b95bacfb65611ab208c8773fdda8a"}]},"branch":"refs/heads/stable/ussuri"},"6d3c282da968938ef234546797a4f4dfee2aca8f":{"kind":"REWORK","_number":2,"created":"2021-12-16 16:07:49.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/92/821992/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/92/821992/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/821992/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/821992/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/821992/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/92/821992/2"}}},"commit":{"parents":[{"commit":"84831a3287aa98a7aa640e36b0c69a03d0dc7534","subject":"[stable-only] pin virtualenv","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/84831a3287aa98a7aa640e36b0c69a03d0dc7534"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2021-12-16 15:38:45.000000000","tz":-300},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2021-12-16 16:07:44.000000000","tz":-300},"subject":"Cap setuptools to \u003c58.0.0","message":"Cap setuptools to \u003c58.0.0\n\nSetuptools 58.0 (bundled in virtualenv 20.8) breaks the installation of\ndecorator 3.4.0 which is still being used in Ussuri and Train.\n\nChange-Id: I6efffb447d998fd1d7edb65bfa689ed435f85602\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6d3c282da968938ef234546797a4f4dfee2aca8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6d3c282da968938ef234546797a4f4dfee2aca8f"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
