)]}'
{"id":"openstack%2Frequirements~822024","triplet_id":"openstack%2Frequirements~stable%2Ftrain~I6efffb447d998fd1d7edb65bfa689ed435f85602","project":"openstack/requirements","branch":"stable/train","topic":"use_2to3-stable/train","hashtags":[],"change_id":"I6efffb447d998fd1d7edb65bfa689ed435f85602","subject":"Cap setuptools to \u003c58.0.0","status":"MERGED","created":"2021-12-16 21:07:04.000000000","updated":"2022-01-07 02:05:03.000000000","submitted":"2022-01-07 02:05:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"822024-use_2to3-stable/train","cherry_pick_of_change":821992,"cherry_pick_of_patch_set":2,"meta_rev_id":"f6302bb415100a7ea6c5d438f1428c27852b6151","_number":822024,"virtual_id_number":822024,"owner":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-07 02:05:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-12-16 22:53:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-12-16 22:53:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-16 22:36:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-16 22:53:22.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"f92ae4105f3117166d6870bd62bf4402400643e1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2021-12-16 21:07:04.000000000","message":"Patch Set 1: Cherry Picked from branch stable/ussuri.\n\nThe following files contain Git conflicts:\n* global-requirements.txt\n* test-requirements.txt","accounts_in_message":[],"_revision_number":1},{"id":"a82eac3d07705e9d447db6d0605903a90c893d43","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2021-12-16 21:09:31.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"628e94ed3a6cf246d178d5101a631558c85f59be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-16 22:36:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44a0f6b0256a4c7286375238e28a87b3 : SUCCESS in 4m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5e4c20f8c2dc486cb77ec0589876873f : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ffb1bc2dd2144dd5a3b0f64d10b3a0f9 : SUCCESS in 4m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c2b64d91e17a4b85815998b70d4cbd27 : SUCCESS in 4m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5d88c711678b4ae88ef376f1fcd3387c : SUCCESS in 3m 27s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/dfbe7adb2b0840fface7141cc888b4fc : SUCCESS in 1h 02m 32s\n- tempest-full https://zuul.opendev.org/t/openstack/build/c37ee0edc7fc4905ba367b30ebe31b76 : SUCCESS in 1h 23m 19s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/8a280706de474de786c88a28b830b074 : SUCCESS in 50m 12s","accounts_in_message":[],"_revision_number":2},{"id":"09858fb493883e3d4731c199ad49e5da992159fa","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-12-16 22:53:22.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d48b74e0a454baa1ed5aca4c9449bd194a1f1a28","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2022-01-06 19:10:48.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"b92885656a83377e7d992160f50818e5b862cf7f","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2022-01-06 19:12:27.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"312b1a18652d6e762f350424822d16b7f5511522","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-06 21:03:53.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ca2a3c4d4b34de8900ab3fce51c8ca2 : SUCCESS in 4m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7741a3ebcaf2485bb1aaa6f62e3fde72 : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2af65c6dbd1f4c329b535fd06099dcf8 : SUCCESS in 5m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f8a4b9bdec4043b69c87015502e360c5 : SUCCESS in 5m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f9bf03f266064268865cac153c86753b : SUCCESS in 5m 39s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e27d1599f7174eeda6635d9dd87abbc1 : SUCCESS in 1h 07m 10s\n- tempest-full https://zuul.opendev.org/t/openstack/build/8bee70521d44413297293d3e110eabf4 : SUCCESS in 1h 39m 50s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/7b8323c97a1440738797d451e4fac23a : SUCCESS in 1h 02m 26s","accounts_in_message":[],"_revision_number":2},{"id":"edc8df14bb4e2037984a90bc442df74361a387ba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-06 21:05:39.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ead989ad3c6e956bd4352d933e02047c77d2148b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 02:05:02.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29b907d204de41628f5d6e7b3f2d6a2b : SUCCESS in 3m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7e277b8fc31646858c6be27a33f7dcf0 : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/671690b6acf34650b67339a3f1395f6f : SUCCESS in 4m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b70240e0380147e9bdb06a22cb849952 : SUCCESS in 4m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3f74f7fd45c3446992be2e10c806fd00 : SUCCESS in 5m 08s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ed61eef8c5f44ed381ccfe566f5b71cb : SUCCESS in 1h 19m 44s\n- tempest-full https://zuul.opendev.org/t/openstack/build/e1123585043f4e9b955ddbdd4c0a66c1 : SUCCESS in 1h 28m 31s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/8ce1af4153b0420d8a81301da94b547a : SUCCESS in 50m 12s","accounts_in_message":[],"_revision_number":2},{"id":"f6302bb415100a7ea6c5d438f1428c27852b6151","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 02:05:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1500e176c4991a3df62eef49738e14e63f699dbe","revisions":{"68a598ce484d6b2054eed12a1300bd6d838807ba":{"kind":"REWORK","_number":1,"created":"2021-12-16 21:07:04.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/24/822024/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/24/822024/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/24/822024/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/24/822024/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/24/822024/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/24/822024/1"}}},"commit":{"parents":[{"commit":"9874f188560ba932c7fadc185caf8da3ed75b17f","subject":"update constraint for keystoneauth1 to new release 3.17.4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9874f188560ba932c7fadc185caf8da3ed75b17f"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2021-12-16 15:38:45.000000000","tz":-300},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2021-12-16 21:07:04.000000000","tz":0},"subject":"Cap setuptools to \u003c58.0.0","message":"Cap setuptools to \u003c58.0.0\n\nSetuptools 58.0 (bundled in virtualenv 20.8) breaks the installation of\ndecorator 3.4.0 which is still being used in Ussuri and Train.\n\nChange-Id: I6efffb447d998fd1d7edb65bfa689ed435f85602\n(cherry picked from commit 6d3c282da968938ef234546797a4f4dfee2aca8f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/68a598ce484d6b2054eed12a1300bd6d838807ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/68a598ce484d6b2054eed12a1300bd6d838807ba"}]},"branch":"refs/heads/stable/train"},"1500e176c4991a3df62eef49738e14e63f699dbe":{"kind":"REWORK","_number":2,"created":"2021-12-16 21:09:31.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/24/822024/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/24/822024/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/24/822024/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/24/822024/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/24/822024/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/24/822024/2"}}},"commit":{"parents":[{"commit":"9874f188560ba932c7fadc185caf8da3ed75b17f","subject":"update constraint for keystoneauth1 to new release 3.17.4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9874f188560ba932c7fadc185caf8da3ed75b17f"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2021-12-16 15:38:45.000000000","tz":-300},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2021-12-16 21:09:19.000000000","tz":-300},"subject":"Cap setuptools to \u003c58.0.0","message":"Cap setuptools to \u003c58.0.0\n\nSetuptools 58.0 (bundled in virtualenv 20.8) breaks the installation of\ndecorator 3.4.0 which is still being used in Ussuri and Train.\n\nChange-Id: I6efffb447d998fd1d7edb65bfa689ed435f85602\n(cherry picked from commit 6d3c282da968938ef234546797a4f4dfee2aca8f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1500e176c4991a3df62eef49738e14e63f699dbe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1500e176c4991a3df62eef49738e14e63f699dbe"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
