)]}'
{"id":"openstack%2Frequirements~822761","triplet_id":"openstack%2Frequirements~master~Ia7b92d62abdedcb6222cef4159d9edecdf65e0e0","project":"openstack/requirements","branch":"master","topic":"directord-bootstrap","attention_set":{},"removed_from_attention_set":{"7144":{"account":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"last_update":"2021-12-22 22:51:13.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia7b92d62abdedcb6222cef4159d9edecdf65e0e0","subject":"Add task-core to global-requirements.txt","status":"MERGED","created":"2021-12-22 19:27:32.000000000","updated":"2021-12-22 22:53:30.000000000","submitted":"2021-12-22 22:51:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"822761-directord-bootstrap","meta_rev_id":"12e402debbadcf68f4eed2eacd058f1468c24414","_number":822761,"virtual_id_number":822761,"owner":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-22 22:51:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-12-22 21:30:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2021-12-22 19:28:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2021-12-22 20:49:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-12-22 21:30:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-22 19:28:14.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-12-22 19:45:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-12-22 20:29:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-22 20:49:14.000000000","updated_by":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"reviewer":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"state":"REVIEWER"},{"updated":"2021-12-22 21:30:50.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"7b823cb5fdfb8f59f2699c8653876c17bbad98b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2021-12-22 19:27:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ec589deab7065d8bf87f7ea14cc52a45eb1f33c5","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-12-22 19:28:14.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f8fe316c9134a40546f1468413f1861f76e65a21","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-22 19:45:50.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/721c1b417a854f5b8cc5fc436b527559 : FAILURE in 13m 24s","accounts_in_message":[],"_revision_number":1},{"id":"4435a633c163184438b67274bcb41b5feb5486f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-22 20:29:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7b35ee38bab467ab44381d1b4bc6ffd : SUCCESS in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e33193ecca640f6a4cee0b9263d7c66 : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fede8059c38942db8c95dc0dd783ad48 : SUCCESS in 5m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/67d916d9b5114b0990e377394a2cbbc9 : SUCCESS in 4m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a9d17e2243f14a6894391d74b9468157 : SUCCESS in 3m 52s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6604e2969a66452d8c923b8340a43c8b : SUCCESS in 4m 23s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5623bf4278d74ca89a7c6f1cfa50b116 : SUCCESS in 11m 04s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1762c18d3559492ba6cb4c7f2fe136b2 : SUCCESS in 8m 21s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/c1c07524124b4637816094e6243daea2 : SUCCESS in 5m 45s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/b96045ae86384c7087343280a6cbfe91 : SUCCESS in 12m 33s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/8e65a51c70804277a715ee457389f0bc : SUCCESS in 6m 17s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/69b31e1a66aa4e6b8a8aec50a4108970 : SUCCESS in 11m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/23a4581c1e5542a8b49c8f6ee11ed227 : SUCCESS in 7m 50s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/33504daa71d04d28a1aa51beaccfb8f6 : SUCCESS in 9m 42s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/a72bd61232ca4f96b91c8923c6a7dbbc : SUCCESS in 11m 49s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/1dc25ee4dbda40c39d80bde86ded2450 : SUCCESS in 4m 43s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/f3f831a4a9ab4114937cba2a9d54aa76 : SUCCESS in 22m 56s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/91dde2ef6898467c98fd12a4fa57ab22 : SUCCESS in 6m 03s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/a8e07c54e780458382eb8d1ebb236e19 : SUCCESS in 23m 00s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/e1d9fc5b89f143908d28b274b6d9e99c : SUCCESS in 13m 40s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/03480cf53ca4409f81f31a847b85b87a : SUCCESS in 5m 26s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7f33c5322aa44367957663c8a4473d8a : SUCCESS in 4m 46s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/36fdb306a1824c039e9c3354d2084f03 : SUCCESS in 21m 23s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/c3e98631619b43528469e9a26285de5a : SUCCESS in 6m 45s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/57551eec7edc403aba7d12dd30dafaa7 : SUCCESS in 5m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7955287aaaa846649cf4de1189b53fe6 : SUCCESS in 7m 08s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/4c0b17a8df364973967cae6d5b76b63b : SUCCESS in 10m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4fe3344005914839b1b2a08a4780364c : SUCCESS in 37m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8ada6cbd6c004226b775da656be69203 : SUCCESS in 57m 08s","accounts_in_message":[],"_revision_number":1},{"id":"94f735a1ad4efd940410488b4366e1f0f7197cc0","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2021-12-22 20:49:14.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"37262a065563019e39090efac657b59e78b5e15b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-12-22 21:30:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"61afd34f328075ac5451c7500f82ee49f919bd34","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-22 21:31:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0683d357f19abcf3be1abe5ff5bac164836e8d5c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-22 22:51:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed0ac25c9eac402194f2fccd8a96f927 : SUCCESS in 4m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/822de226582742be8355d7a9fa532c9d : SUCCESS in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/48489c9e8b6940a0a18bfdb8c31b288f : SUCCESS in 5m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5d2e8642f06644de83a512b90e642721 : SUCCESS in 4m 05s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/272803b292584dc8b523ed8bc644b4b7 : SUCCESS in 4m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1814b62abe8843bb94282e2616be3742 : SUCCESS in 3m 50s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c9eddde5529a4bc2be247daf4511dd94 : SUCCESS in 8m 51s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b6f4699704bf47a8908cf0860dc155d6 : SUCCESS in 8m 38s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/cbd293ebd0f14f869eb3c588a54b35b2 : SUCCESS in 13m 10s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/06ae0a5718b14b9eb2f918217fd7ba3e : SUCCESS in 5m 00s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/581488ab7c964ca1b525d3843efe6a2e : SUCCESS in 11m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a24461bf84ad4e78a71720b1b1cd299f : SUCCESS in 8m 46s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/a08e755d0c0d46448103c400b7dc6317 : SUCCESS in 6m 45s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/a923264d718c4bd891d00dccddb43df6 : SUCCESS in 13m 40s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/89ffde7a8ad54e05a3655b6c9771294d : SUCCESS in 4m 06s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/4a5156bafb204336aef4902c7ea85c72 : SUCCESS in 15m 05s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/5bb59b71d3154b1db1f7c87186bfdf94 : SUCCESS in 5m 03s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/570921bfb69c48ccb0f42724e841047a : SUCCESS in 22m 47s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/548f5ee4476a44fcb329d1ff3875b939 : SUCCESS in 10m 01s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/ba0774babc7e4d62b8ed9ae6344a9fa2 : SUCCESS in 5m 18s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9930db964aa64da48b0a4640f936d52c : SUCCESS in 4m 36s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a1bc63a3dd514854848e3c593b0fd4a7 : SUCCESS in 18m 35s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/c5872fcb97eb407d8051142a9ce19283 : SUCCESS in 7m 06s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/323373720daa49e484e156eb51df46e3 : SUCCESS in 6m 18s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7693d0013e634ee8807cae9fd8ef31b5 : SUCCESS in 5m 44s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/eced5557c29e4e0a9472e2beefde0745 : SUCCESS in 11m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7e2da39394c64d459f95bfa918994dbd : SUCCESS in 41m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/673e1bbe0b8f4e06a8b3e49c367dedc8 : SUCCESS in 59m 26s","accounts_in_message":[],"_revision_number":1},{"id":"ff9646cb931b0eb6c4fd8cfd1003510d657103b4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-22 22:51:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"12e402debbadcf68f4eed2eacd058f1468c24414","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-22 22:53:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05ece1f0cc884f6bb305e3cf99ad2768 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e5ba688ec6ad0f16b8bbd95e56e86c857dfcd96f","revisions":{"e5ba688ec6ad0f16b8bbd95e56e86c857dfcd96f":{"kind":"REWORK","_number":1,"created":"2021-12-22 19:27:32.000000000","uploader":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"ref":"refs/changes/61/822761/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/61/822761/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/61/822761/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/61/822761/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/61/822761/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/61/822761/1"}}},"commit":{"parents":[{"commit":"6615f437dec636877b8257cc24d671fc52473d5a","subject":"Add directord to global-requirements.txt","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6615f437dec636877b8257cc24d671fc52473d5a"}]}],"author":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2021-12-22 19:24:51.000000000","tz":-300},"committer":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2021-12-22 19:27:24.000000000","tz":-300},"subject":"Add task-core to global-requirements.txt","message":"Add task-core to global-requirements.txt\n\nThe TripleO team discussed and agreed to move forward with adding task-core to\nTripleO at the last PTG:\nhttps://etherpad.opendev.org/p/tripleo-directord-task-core\n\n* Is the library actively maintained?\n\nYes, it is actively maintained by the TripleO community at\nhttps://github.com/directord/task-core\n\n* Is the library good code?\n\nYes, as the authors, we\u0027d like to think so.\n\n* Is the library python 3 compatible?\n\nYes. There are python3 jobs setup on Github.\n\n* Is the library license compatible?\n\nYes. It is licensed Apache 2.0.\n\n* Is the library already packaged in the distros we target (Ubuntu latest / Fedora latest)?\n\nNot yet, but we are in the process of packaging it in RDO:\nhttps://bugzilla.redhat.com/show_bug.cgi?id\u003d2021983\n\n* Is the function of this library already covered by other libraries in global-requirements.txt?\n\nNo.\n\n* Is the library required for OpenStack project or related dev or infrastructure setup? (Answer to this should be Yes, of course) Which?\n\nYes. TripleO.\nThe first patch that would require task-core is:\nhttps://review.opendev.org/c/openstack/python-tripleoclient/+/821624\n\nIf the library release is managed by the Openstack release process does it use the cycle-with-intermediary release type?\n\nIt is not yet managed by the OpenStack release process, but may be in the future.\n\nDo I need to update anything else?\n\nNo.\n\nSigned-off-by: James Slagle \u003cjslagle@redhat.com\u003e\nChange-Id: Ia7b92d62abdedcb6222cef4159d9edecdf65e0e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e5ba688ec6ad0f16b8bbd95e56e86c857dfcd96f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e5ba688ec6ad0f16b8bbd95e56e86c857dfcd96f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
