)]}'
{"id":"openstack%2Frequirements~823104","triplet_id":"openstack%2Frequirements~stable%2Fxena~I0953b5c1f0b784f03c427b5670706a5211c44760","project":"openstack/requirements","branch":"stable/xena","topic":"bug/1949964","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2021-12-31 05:23:17.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0953b5c1f0b784f03c427b5670706a5211c44760","subject":"Update amqp version","status":"MERGED","created":"2021-12-28 17:29:45.000000000","updated":"2022-01-27 16:46:57.000000000","submitted":"2021-12-31 05:23:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"823104-bug/1949964","meta_rev_id":"44fec7491470f4a499a211255121d126dd27f204","_number":823104,"virtual_id_number":823104,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-31 05:23:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-12-31 04:22:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2022-01-27 16:46:57.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-12-31 04:22:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-28 17:42:28.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2021-12-28 17:45:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-12-28 19:11:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-31 04:22:55.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2022-01-27 16:46:57.000000000","updated_by":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"reviewer":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"state":"REVIEWER"}],"messages":[{"id":"b3b301901e7aa6efb2994939a9c4813c19a7cfd7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-12-28 17:29:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"95eb865c7e0175f7f7a0906fea34375bac4d1b40","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-12-28 17:42:28.000000000","message":"Patch Set 1:\n\ncould you link to a bug that describes the issue in the commit message? (openstack or uptream)","accounts_in_message":[],"_revision_number":1},{"id":"006e1b88ee96d28ed767346c42c3b7a14423dc13","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-28 17:45:38.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/c21c2d8bb3634c03b1d8af156aa424ae : FAILURE in 15m 21s","accounts_in_message":[],"_revision_number":1},{"id":"272fa68309d6478e223b13139c381bd201fee66f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-28 19:11:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25a23958119d43c7a7b45454f84290cb : SUCCESS in 3m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0cc6ca3d848d4ffb867460d98b3cb012 : SUCCESS in 3m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fef13793dbf44717aa9ffb29d389f562 : SUCCESS in 3m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/acfd59527e664d63bdd14624c7e86bce : SUCCESS in 4m 09s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bd7e8feb44cc424192b4fbd22d3e111b : SUCCESS in 3m 32s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/dc663ac233c64feeaea06680a74f73d9 : SUCCESS in 3m 29s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/82bd761e5f2a4fedbf3bf260b5330b66 : SUCCESS in 13m 28s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6d06706b51ab439f9bf17c747978c292 : SUCCESS in 9m 43s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/6f1858b25f0d46b3a3fd42cfb663a851 : SUCCESS in 5m 49s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/5943f68080ac4435b95d0f7f9dfccbc0 : SUCCESS in 8m 54s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/af03f111add24152a11b0295a328f5fc : SUCCESS in 5m 24s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/23179bca1cbc492a987f24fa0c502544 : SUCCESS in 10m 57s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/808f2c7ca5224f948ed55fee9d352729 : SUCCESS in 9m 17s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/d4e92884e16944eab192686810ad5a8d : SUCCESS in 8m 26s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/86d55002502949d58904907bcf45c03b : SUCCESS in 14m 10s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/ef7c979ec2514a6f93a9e497c9ba0e23 : SUCCESS in 8m 30s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/071408c3e6764900b77bfae2f6ce7a2d : SUCCESS in 4m 49s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/df2be0d875344c9ebbd1b2c52a073255 : SUCCESS in 24m 16s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/a05da2104b2c44d380f23edb1c4acf0a : SUCCESS in 11m 29s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/f502b558ca7c41fa9b6f1de4d17d96c2 : SUCCESS in 5m 15s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/63fcfa6a11324858bc84464e77fecc3b : SUCCESS in 3m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/44732b3ad24640f6b6faba02cbbe5356 : SUCCESS in 18m 33s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/065acaab15ac4a969edbe231a9545fec : SUCCESS in 5m 18s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/ae72ca14a10641ba9837a85c7ddeabcf : SUCCESS in 5m 32s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6b7d6d244f4246548decd97d43792db6 : SUCCESS in 6m 43s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/6b757f26202b47669671b7ba2831d9bf : SUCCESS in 12m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bbc97986f12041e99bfe75940e755819 : SUCCESS in 47m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d4b68b8f6260416b9266e6393b1a45e1 : FAILURE in 1h 39m 15s","accounts_in_message":[],"_revision_number":1},{"id":"0eaa5b11759cb33e0cf7d13b39027394dfbd9d6b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-12-31 03:18:12.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"01d7802917ca840d736433c31181ef74d38b8e2e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-31 03:34:06.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/f601e591ee9d4eff816beb6ff0dede44 : FAILURE in 14m 02s","accounts_in_message":[],"_revision_number":1},{"id":"09f12cbd6beb7ecee95f5a5e51f6bcd57fe82f2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-31 04:17:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0485665946f74431b3927b88213e3ee4 : SUCCESS in 7m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/43e613cee1694d54b004671352a9e490 : SUCCESS in 6m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c4970768fff942b6b82084d45dacd744 : SUCCESS in 8m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0110fbbeabc544df9d7ea1f704e43b94 : SUCCESS in 8m 07s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/99605127dcf74307bd725bf0b43d21c9 : SUCCESS in 5m 47s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c89ccf98208d49a9bd26fa6001b44b1f : SUCCESS in 4m 24s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/12fdc74457c24e7792a51282dc49c50f : SUCCESS in 9m 05s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/cfc929ca2a694f06926b801ca1630ca9 : SUCCESS in 8m 52s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/340329a146764393a0eec926dcd0f09f : SUCCESS in 5m 33s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/aa4a2a59f2b64e5e864000decc1391b4 : SUCCESS in 10m 01s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/2a895f1030c34e87801bb7c5c430bbbf : SUCCESS in 5m 53s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/db5c73940a574e839938c2bdae720289 : SUCCESS in 9m 39s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7959f494cb9d46529cfba9861f1c2cae : SUCCESS in 8m 17s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/119ddfc9754f494fb535a457fb2d93d3 : SUCCESS in 7m 10s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/46136ac5a5664e7fbc827a7ba2956c95 : SUCCESS in 11m 07s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/f7b99fbb1dda448b81078a3fc67c03f1 : SUCCESS in 11m 45s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/c976e236868f4e6688c0a9ff5c78ba11 : SUCCESS in 5m 40s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/50e3f9dc6c354c2cb5f09eae6b0d42c6 : SUCCESS in 20m 22s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/a0dd9ca2d03340499064be42aafc05d4 : SUCCESS in 13m 42s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/d856a72f51644a9bb0903145f297d860 : SUCCESS in 6m 12s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/dc3b970ec8f54c9486f2d620c676a3fe : SUCCESS in 4m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c4458c3d053647b3a2385ebaaf0c889a : SUCCESS in 18m 29s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/434b96ffa35e4114af4073f968afd18b : SUCCESS in 5m 29s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/c7f83c83703d4a82b965c7c52cce7d87 : SUCCESS in 6m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fc3729aea79b46129314afa65bd9216a : SUCCESS in 6m 58s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/6851a416ad2e4dfaa79b9ab0190a9bfa : SUCCESS in 14m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cc36b209a03e46a59ee0471918636f6b : SUCCESS in 43m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/114e1baeb0014f3884304526a34742dd : SUCCESS in 57m 26s","accounts_in_message":[],"_revision_number":1},{"id":"b13903970b0e1f874fe7a457a1af759cfcabfc92","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-12-31 04:22:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2048e53f78bb82120f7d16fa44814a7b40ce3b2e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-31 04:23:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"86e6e95df13c1ed164ad11774d0b3b782ea53d37","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-31 05:23:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f1c9354a9eb4613880b6b2aa1d3c719 : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/14ceaa27f0ff4a0cb1df8580b4b166eb : SUCCESS in 4m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d6646fc490fb4369ab8b96bf30bd2f1d : SUCCESS in 4m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/35e496caedab4232a4a2e5b9667d459c : SUCCESS in 4m 50s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0e5e493b1f74440290bdbfac03b480f2 : SUCCESS in 3m 06s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c05c98439d08469aaa3ef238ecad612f : SUCCESS in 10m 21s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/611efe997a7e473c895807b1d5948d99 : SUCCESS in 8m 35s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/b636c9bd6785414e84c1962e813b97c4 : SUCCESS in 8m 42s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/0ba6ae74321745a29fbf177c28c709b3 : SUCCESS in 4m 41s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/2601f8c66b464bfc9db07cda103661be : SUCCESS in 7m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/bc1bcd6f4aa64c10bc3483d619d4c40a : SUCCESS in 8m 56s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/5f175f8b3dd343d2b50f81aa3564e766 : SUCCESS in 8m 15s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/1123ddf0acf24959b3ae935d7e709ad8 : SUCCESS in 34m 10s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/2a9622fd05c94322b4b98ae5571252f1 : SUCCESS in 13m 16s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/bd2afc0bde0c40ec8c0ba4109c80fc02 : SUCCESS in 7m 07s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/fcd236f9e6824c81ba417b3572d8cb7a : SUCCESS in 19m 25s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/0967977ac361402ca0777dd53a9153e6 : SUCCESS in 9m 03s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/e9f9845e4e2e46bf9fea66fa241acca1 : SUCCESS in 5m 43s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/af3652387b1d455c8229b6ce6cd2f11d : SUCCESS in 4m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/84b8e19afa054a5bb687d932a27f87f8 : SUCCESS in 18m 19s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/f4d59ddffa564a29a923192560f2de3a : SUCCESS in 5m 44s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/bf7bbe3db28b43f18d56334610585a50 : SUCCESS in 5m 20s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c739d31cb54741059a13ef9ee3f27edf : SUCCESS in 9m 08s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/d2a918a33e63489e903327c08cfdeafa : SUCCESS in 14m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c1bcdce4c3e64f4183570ae7108fd28a : SUCCESS in 32m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/97f1b33804764556898750fec5b2d44d : SUCCESS in 58m 57s","accounts_in_message":[],"_revision_number":1},{"id":"7b9ce0b10e58e41d3e33fa39f556567e686ed069","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-31 05:23:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"44fec7491470f4a499a211255121d126dd27f204","author":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"date":"2022-01-27 16:46:57.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f243f496e015d7904b2ca1973ab484bdc02d9458","revisions":{"f243f496e015d7904b2ca1973ab484bdc02d9458":{"kind":"REWORK","_number":1,"created":"2021-12-28 17:29:45.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/04/823104/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/04/823104/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/823104/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/823104/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/823104/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/04/823104/1"}}},"commit":{"parents":[{"commit":"f5dcb1d64534feb5720400c81089bd2b0765e2c3","subject":"update constraint for metalsmith to new release 1.5.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f5dcb1d64534feb5720400c81089bd2b0765e2c3"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-12-28 17:29:12.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-12-28 17:29:12.000000000","tz":120},"subject":"Update amqp version","message":"Update amqp version\n\nAmqp package that is being currently used supposedly leads to\nleak of file descriptors.\n\namqp bumped to version 5.0.8 should fix the mentioned issue\n\nChange-Id: I0953b5c1f0b784f03c427b5670706a5211c44760\nRelated-Bug: #1949964\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f243f496e015d7904b2ca1973ab484bdc02d9458"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f243f496e015d7904b2ca1973ab484bdc02d9458"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
