)]}'
{"id":"openstack%2Frequirements~823122","triplet_id":"openstack%2Frequirements~master~Ib00a6f22991030333930a6cb013e5c58c1c75c00","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","attention_set":{},"removed_from_attention_set":{"14288":{"account":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"last_update":"2021-12-29 20:14:38.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"Ib00a6f22991030333930a6cb013e5c58c1c75c00","subject":"Updated from generate-constraints","status":"MERGED","created":"2021-12-29 02:57:37.000000000","updated":"2021-12-29 22:39:53.000000000","submitted":"2021-12-29 22:38:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"823122-openstack/requirements/constraints/noclob","meta_rev_id":"c8c45b9c1fc361e475f387982e01fb8b54966ea2","_number":823122,"virtual_id_number":823122,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-29 22:38:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-12-29 18:37:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-12-29 18:37:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-29 03:10:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-12-29 03:24:36.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2021-12-29 05:10:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-29 08:19:38.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"CC"}],"messages":[{"id":"31d640beae071db2a695726dc5ad88dd871a4d0c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-12-29 02:57:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"69c71dc0d1a8e86dd8862ebfdaa3d5db95efbd17","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-29 03:10:20.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/274dbee818c441ea85a1355ef2333070 : FAILURE in 7m 55s","accounts_in_message":[],"_revision_number":1},{"id":"b3daf010f47c01b41d0da8f3b3195cc561387426","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-12-29 03:24:36.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e4408baa28dc8f11a887f3326531ab46e60c00de","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-12-29 03:24:49.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":2},{"id":"f305cfbcec0c2c7eeb4f653205756111707a6ad9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-29 03:56:20.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/3d6c19f740364949af8116fdbc4582a6 : FAILURE in 17m 35s","accounts_in_message":[],"_revision_number":2},{"id":"a71c1019226e3dbde88a558b2bb1a02dfe10d368","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-29 05:10:29.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/895f5d02b1704411bac49040091118d7 : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8186063b6cf2466ba267a196e1bcd745 : SUCCESS in 3m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d078808673cf41929253c4940c979748 : SUCCESS in 4m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d39d3c7408ab480388f165b2d6e34b49 : SUCCESS in 4m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/50e956d368784c708e7da499808efc44 : SUCCESS in 3m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4ffcea35244946a19e8236c49486ff5f : SUCCESS in 3m 28s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a509369c645343148b4716cc1e0dda6e : SUCCESS in 11m 54s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/70276a4c54a84899bbec4f228e6ae5ae : SUCCESS in 10m 16s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/36f8489ce0f34748a4b3e61750fc0412 : SUCCESS in 6m 01s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/b7b0ccd22b0e4e98857b4133686404ab : SUCCESS in 9m 49s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/c70d627c6fa2420ba2ca11f942bf2fb0 : SUCCESS in 5m 24s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/0917f71161a543d2887700fb05b948ed : SUCCESS in 10m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/dd1178a12f3c4cbf83eaeb5fe191fd98 : SUCCESS in 7m 07s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/a403062b24a54b86bcb956c33b415e75 : SUCCESS in 8m 39s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/92d6d2358d2845d291bbd9eb11c2df1a : SUCCESS in 18m 17s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/0cdee9af2e6744749cde580bfac33f47 : SUCCESS in 3m 56s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/9772918fa3844045bcdc53983e4da364 : SUCCESS in 15m 51s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/6bfaee6b15a6490aafb0d85fead55e6a : SUCCESS in 5m 41s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/f5d62662350d4cc1890b6b6a71112378 : SUCCESS in 24m 59s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/5c38e5c511134166a4460b65ef194f69 : SUCCESS in 10m 23s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/40795d4f8e1349b2b844957dfb308d24 : SUCCESS in 5m 17s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/08ffdbb3e94342dfb03288c16a430507 : SUCCESS in 4m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8f6849b8f9804154b3459144db412340 : SUCCESS in 18m 11s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/991bdbca88294ffa84cb3f10ff4745e3 : SUCCESS in 6m 35s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/d27a9d407664451b86ec02325271f337 : SUCCESS in 6m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/817cb4d9a0054e4eb1a1e3853a68b640 : SUCCESS in 5m 20s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/2934185b9e8b4f6a867d1db8905381a6 : SUCCESS in 9m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a1b1226b99f9425096f6a7b7a35c917d : SUCCESS in 44m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/98bb1421e559421c860f539281bc2d70 : FAILURE in 1h 41m 12s","accounts_in_message":[],"_revision_number":2},{"id":"dd077248247e06a3a308a507bdd28900dd964066","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-12-29 05:27:38.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"60cedaa282bc166d25526f4b1c768bfcea087b5d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-29 05:48:46.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/d3d6bafdbd74468493ddd5bdfd4ca4ed : FAILURE in 18m 58s","accounts_in_message":[],"_revision_number":2},{"id":"111c96c74aeb1f69e1c3b71c5869019cb2b9e13f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-29 07:05:24.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f3c90d2c7134a2d9ef2896c98da5180 : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c314f3f4575487aab6336c3a1a4e454 : SUCCESS in 3m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d688a89e030c4d61860be9f6a8686194 : SUCCESS in 3m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9891ee4d576b4a2c9ca9c5021e156671 : SUCCESS in 4m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/06ecae6ffe054b3d811eeb1c7055724a : SUCCESS in 4m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c5482b16d9d4463f8d9bf4fc53211c4d : SUCCESS in 4m 11s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1536736c62e64b03b2e81cd6c3988ae9 : SUCCESS in 9m 37s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/926389a897f34cdfbe5cbbcddfbeedb1 : SUCCESS in 10m 17s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/363b5209fb3e41b9bd0de5288f4aa4ec : SUCCESS in 6m 00s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/a1032cc83f134fe5833ffec5c3ba6a26 : SUCCESS in 10m 41s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/e970f16865654487a824aa1e7e06c188 : SUCCESS in 5m 27s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/745a561ea1d94dfb8ec4721b7740e573 : SUCCESS in 10m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a3149d1b7238417eb5b1908d4e970ef6 : SUCCESS in 8m 10s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/3917bd85651a4b63a34f7da622f83a5f : SUCCESS in 8m 35s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/4f86a73ff8d84fb2834e4ae0da4a1a30 : SUCCESS in 16m 02s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/6e6143d544da4d738943fbd2c24bc4aa : SUCCESS in 4m 27s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/caf150612c7b4968aae613c2b4a2fe04 : SUCCESS in 16m 04s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/e8a4b85af80f497e981ee3aec78cfd88 : SUCCESS in 5m 52s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/a942d92884ac4a15bf5bea6c60b1fdf8 : SUCCESS in 24m 08s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/c5efde0b51c442e69b5177488f47f432 : SUCCESS in 12m 50s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/2368cd9106674ad281b694b883007e86 : SUCCESS in 5m 49s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3850ce869aa3498fb0206919a7981af5 : SUCCESS in 3m 57s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a58ad40b4ac640f7aa7846e12c340c8c : SUCCESS in 18m 50s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/c56565e947e541eeb36d2379176247f2 : SUCCESS in 6m 22s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/ce565b38c44345f0ac53f76af9759f25 : SUCCESS in 5m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/144c6eab0280450ba85a8945818c2fc2 : SUCCESS in 6m 50s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/800113aec94a4ad7b75d7844b89191df : SUCCESS in 12m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0896df482032462282afdf3beeddf594 : SUCCESS in 46m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/880a7d8f5f434c01bb38ad8558cdf1c5 : FAILURE in 1h 35m 49s","accounts_in_message":[],"_revision_number":2},{"id":"4826cc1b59d0e1f499eecd65a0cb332cd6b35ce6","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-12-29 08:19:38.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6cdcd4790cd31309d675a75fe0995d93023db297","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-12-29 16:48:02.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"70bf9301f3851f36c5baa3100a57a999431097ac","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-29 17:06:20.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/5df1e98fc52f4e628af280aa5630d3ad : FAILURE in 16m 13s","accounts_in_message":[],"_revision_number":2},{"id":"b950791055aac3ae4058e0fabd7327ddaa1b4c6e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-29 18:36:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58d39cd2dfe94db5aeccbc849ff23368 : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf322bc4466e472c94f163304659bb37 : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a6c0be4fb9c8477c9ad52627518879f1 : SUCCESS in 4m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2b4f7f0df7d7490eb0422f189e941180 : SUCCESS in 5m 03s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/96be21a344c94873a82fb47510de5764 : SUCCESS in 4m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d0fdc05e53ac4075a1abeece5183d4cd : SUCCESS in 4m 32s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c69d70523375482a997ffbca753bc378 : SUCCESS in 13m 44s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/85f04bd80bc245bc819efa28d40bad2a : SUCCESS in 9m 58s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/863ae92fdb6d40d988a36e17355cdefe : SUCCESS in 5m 45s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/7145248302b540c49d66c9c0d3c4bd96 : SUCCESS in 11m 53s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/0598d399d0f643ff8e5e20bd80203fc3 : SUCCESS in 6m 16s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/569f274b65964785b9c0d8170cabb78a : SUCCESS in 10m 07s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0b2acf309631421991323a63f20c8234 : SUCCESS in 9m 21s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/078902afafc140e48cf90b59286dd3cf : SUCCESS in 9m 22s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/52fb628be9f74dddb2c2a18426ebccee : SUCCESS in 13m 24s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/f338c92013f24caeaafdcd688a0a923f : SUCCESS in 4m 53s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/1aa9c6d3236d494cb45968ec88ad0120 : SUCCESS in 16m 30s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/9e7a4348fe7d498ab84d7fac4249a0f5 : SUCCESS in 5m 41s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/d0f7bf2ea72b4be89d6d5d3b94b957fa : SUCCESS in 22m 31s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/5243be0cdb7f48c081805dd2d19136f5 : SUCCESS in 11m 54s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/414d6da8911447f1a7d9e4f8c17d02ca : SUCCESS in 6m 29s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c8593db32ae34f9382144932a7f438ff : SUCCESS in 4m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2307ec84ecca4681a2431492e7b5b0f9 : SUCCESS in 19m 12s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/270e1ff6492e4f72b96df24db772892d : SUCCESS in 6m 21s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/18639705b8b54970811d10a4b6755271 : SUCCESS in 5m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/aa33283f1b4e4322bdad735be859dbdd : SUCCESS in 6m 56s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/1e3c6d764e0d4882b763f58d7d498124 : SUCCESS in 10m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b5fcf3d67b8946ccab9d62154869bb5d : SUCCESS in 40m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5a2c613888274fc384105b1956167380 : SUCCESS in 1h 43m 04s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"638a8053422da7f2208b8bea7f29a0c50a8a0b56","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-12-29 18:37:13.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"6da44e285b57640177cb2da847668b030e9b77ce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-29 18:38:08.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"cdda4a04329e311257564c51d28f9150b210016e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-29 19:30:05.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff989ed6ceed41df81003cdfbb24f1c2 : SUCCESS in 4m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33ba7bbcfccd4352b56e1479865f6a78 : SUCCESS in 3m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c3b7f281399246a488390ef2939e384c : SUCCESS in 4m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b3dd339ec08d499f9d6cf205f77f43e0 : SUCCESS in 3m 34s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7357658d2d584042afabb6dcc998bc20 : FAILURE in 2m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/16c91a103d734465a5d0253bc564e506 : SUCCESS in 3m 19s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2db3a63480814684b6e091ae2bcdff3b : FAILURE in 4m 30s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ee19296fcb844539a178c3e7f04d98b4 : FAILURE in 4m 31s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/a581db649bef4ad2aa78fe2a729e5321 : FAILURE in 3m 23s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/4555d0550a2a4e39847c84572a7fdfdd : SUCCESS in 5m 43s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/e36505bcc3ac44929cacbbd13de6876d : SUCCESS in 10m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b28349748895491a908574459fba8593 : SUCCESS in 8m 07s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/8ba6872a70374e25a2e2e4ba374de7a7 : SUCCESS in 8m 26s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/69df70fc0f674c42a4d3102c663510ea : SUCCESS in 13m 21s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/36251bbf1d334ac29b495796b9a1bec6 : SUCCESS in 3m 42s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/e7dafb897e35476a92f0ee86644170fa : SUCCESS in 24m 43s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/b824c287a2fb47e791ee192cd3128d54 : SUCCESS in 4m 42s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/b99001fe0bed449980deff25fddc938b : SUCCESS in 24m 22s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/5657387c26cc4eedae9db4932c62a7c5 : FAILURE in 4m 53s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/c99f93484bdb4136a84117488416e817 : SUCCESS in 4m 49s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0644e4171f8449b1bab18dd73196ecca : SUCCESS in 3m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8952a0f1e2c54482a311d7d935bb9d46 : FAILURE in 5m 35s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/8bd94bf6bb344230aa9afc41c1bf285a : SUCCESS in 5m 14s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/dce2bc5dcf104d0a98afe83f8e4f034e : SUCCESS in 5m 34s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a2f5b1816962405e9335333e79f65871 : SUCCESS in 6m 30s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/c08135da7cc846ac920f2aebbeb5da36 : SUCCESS in 12m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d0972cb933294974aaf4cd1eef4975e9 : SUCCESS in 48m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b0fb96623b2240e68c9c8b9f6f8e9ef6 : FAILURE in 35m 17s","accounts_in_message":[],"_revision_number":2},{"id":"27feaedfa26da9095552eb520f427aa697e302f4","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-12-29 20:14:38.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"8e30224a96dd2e5614438e35a0fc42ef5fcbe24b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-29 20:31:31.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/f25e750c389d46e1b85226982a2bc833 : FAILURE in 15m 05s","accounts_in_message":[],"_revision_number":2},{"id":"b765a243f59c8ba889417f3b75e706fc5cdc71df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-29 21:22:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/373b311b89674ec29cba41749aaf5bec : SUCCESS in 4m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8bc13311d854837ae992fb2adf16072 : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/af98dc66cba24c91af972d6506f21bad : SUCCESS in 4m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c706ddbd8791435bb2b462c4e5d9c0d6 : SUCCESS in 4m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/800d51e47d084d7d886c4daed5664b53 : SUCCESS in 4m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/35b29f28e3bd46ec9eee5c293b84c37c : SUCCESS in 3m 33s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/441c499ac8f04c83b569001b80b1b060 : SUCCESS in 8m 18s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5e2bd24e13344436bc13cd6045b473c8 : SUCCESS in 8m 20s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/c457c53b9e884022bf010a15016d9c90 : SUCCESS in 4m 46s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/e2430f3cba3b4bba8ffd269e73a1d007 : SUCCESS in 9m 14s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/91b902597c1542499944fb17e0cc5e82 : SUCCESS in 5m 02s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/fda2b0a217074dda940ec6eee5262460 : SUCCESS in 8m 40s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/355dada9923c4531bc27334fa927bd61 : SUCCESS in 6m 57s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/9e05cdcf1a6144089bfe538142326533 : SUCCESS in 7m 55s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/fbf864fca5b1482c9e32023e938b487e : SUCCESS in 13m 30s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/863d247ade804397950da9eda833fa62 : SUCCESS in 5m 05s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/12ddb8b640ca45ee973f898a55d7f176 : SUCCESS in 14m 38s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/13da45b3b9674bc88c89873dba4ae4a3 : SUCCESS in 6m 06s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/8038f63e899543b49b595e14b325b76d : SUCCESS in 31m 04s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/ec9d1218aac74f69b2375e4915a2e2af : SUCCESS in 11m 04s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/3479294c2e2942d189feff57e400000e : SUCCESS in 6m 45s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4964ccdd3ef2422ca700ee54c458f2ab : SUCCESS in 6m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/94e445d979ca4f91a0122d4645fa9e5b : SUCCESS in 17m 16s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/c345a821171a4ed1a9ef23331bd74182 : SUCCESS in 5m 58s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/edfa7e4c89864bf79e65c26a510d7077 : SUCCESS in 4m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/833c0531413b4e3da99e4b5e4b9a5491 : SUCCESS in 6m 29s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/6337395bb96043c6b77af55022bb8d53 : SUCCESS in 8m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dd95ad6ff5344e9ebb63f9c18ac258c3 : SUCCESS in 50m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/508f32cc80b940f6b1def7b2ccd07bcc : SUCCESS in 1h 05m 41s","accounts_in_message":[],"_revision_number":2},{"id":"9247a4ee0af6a7cd72173529d22619ea01d9b584","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-29 21:23:52.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6dc00967f5b0488ab59d1a7526a0eff77d2c07d2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-29 22:38:35.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c16c020761d4ac18f34daf0b3f28dd7 : SUCCESS in 6m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/099c26bbfb694ebcbcc37fbc605a65b0 : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/36183a6e2bfc46f88d910e747871652a : SUCCESS in 5m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0676d4055c1c4151b0300a774b19312e : SUCCESS in 3m 48s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f6e2bc1dba6a47f59bbf1c4f540b3f6e : SUCCESS in 3m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/909afca76f9f4d7db5fe082b8ec8594c : SUCCESS in 4m 09s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0a40279ce9094914bd9fc2b4a5f0a953 : SUCCESS in 8m 19s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/878266805b50478cb975f11c6bf48ff8 : SUCCESS in 8m 12s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/358d052b3fe84afeb1555493cf21b914 : SUCCESS in 9m 34s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/572a2e259bb8467a9027c80714b29843 : SUCCESS in 4m 50s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/235ffdf1e593467f9abd8176fff58d2f : SUCCESS in 9m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/039dbcd680d3475187573b5bf107921e : SUCCESS in 7m 36s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/2d68ad66240e4c8084d0503b09d8559b : SUCCESS in 10m 18s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/98a023bafd464db1ac4f9abf71f5f64b : SUCCESS in 11m 57s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/80c44f1825bb49a5a817e8314b826e70 : SUCCESS in 6m 07s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/d4a430beb9fb4e9492eaa45e1140e459 : SUCCESS in 17m 19s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/543ab78c06594fa9a0d09023ebf999e1 : SUCCESS in 4m 27s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/931e12cddfbe4da59969364d09c673ea : SUCCESS in 19m 15s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/ab5c9e36dbb54bb89a56f9d6488c9abf : SUCCESS in 10m 59s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/fdf65ec8045542f48ca852423722e434 : SUCCESS in 7m 03s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/91cebbf3c02e4c3790da43f3465c218f : SUCCESS in 4m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/db0bd3eca5664ea89bab10622f6e8602 : SUCCESS in 20m 28s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/9d66fc125b6d493391a9cc13d2f12b08 : SUCCESS in 8m 08s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/b00dc193c74040bf951bbb25a3bcfbdf : SUCCESS in 7m 07s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9caf607d716c4ca5934db04d869399c9 : SUCCESS in 7m 44s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/f7f98514011f40c08c25bf91430dc026 : SUCCESS in 10m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7a62c0afaeb94b44b2b51f759655069a : SUCCESS in 44m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/86d6b71109504ae5980118c69dddb98a : SUCCESS in 1h 12m 02s","accounts_in_message":[],"_revision_number":2},{"id":"bb20035a29a39a7543ab0a344882cd85b3f5c2f6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-29 22:38:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"c8c45b9c1fc361e475f387982e01fb8b54966ea2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-29 22:39:53.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b9d30520ce94499b8008921535ed43a : SUCCESS in 55s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b3a7ef0b94e8e749c013231eea60b253c5b9dbfb","revisions":{"26731d0cd198c1b197757c3ece383c33f34d6db4":{"kind":"REWORK","_number":1,"created":"2021-12-29 02:57:37.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/22/823122/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/823122/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/823122/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/823122/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/823122/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/823122/1"}}},"commit":{"parents":[{"commit":"1288a8180ac70e7788bfc89e3e237e3e03b23cfb","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1288a8180ac70e7788bfc89e3e237e3e03b23cfb"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-12-29 02:57:33.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-12-29 02:57:33.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ib00a6f22991030333930a6cb013e5c58c1c75c00\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/26731d0cd198c1b197757c3ece383c33f34d6db4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/26731d0cd198c1b197757c3ece383c33f34d6db4"}]},"branch":"refs/heads/master"},"b3a7ef0b94e8e749c013231eea60b253c5b9dbfb":{"kind":"REWORK","_number":2,"created":"2021-12-29 03:24:36.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/22/823122/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/823122/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/823122/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/823122/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/823122/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/823122/2"}}},"commit":{"parents":[{"commit":"1288a8180ac70e7788bfc89e3e237e3e03b23cfb","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1288a8180ac70e7788bfc89e3e237e3e03b23cfb"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-12-29 02:57:33.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2021-12-29 03:15:57.000000000","tz":-360},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ib00a6f22991030333930a6cb013e5c58c1c75c00\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b3a7ef0b94e8e749c013231eea60b253c5b9dbfb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b3a7ef0b94e8e749c013231eea60b253c5b9dbfb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
