)]}'
{"id":"openstack%2Frequirements~823175","triplet_id":"openstack%2Frequirements~master~I1116112bf1fbe152fd7616eaf984db9934c2ceeb","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","attention_set":{},"removed_from_attention_set":{"14288":{"account":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"last_update":"2021-12-30 09:35:57.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1116112bf1fbe152fd7616eaf984db9934c2ceeb","subject":"Updated from generate-constraints","status":"MERGED","created":"2021-12-30 02:51:51.000000000","updated":"2021-12-30 09:37:20.000000000","submitted":"2021-12-30 09:35:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"823175-openstack/requirements/constraints/noclob","meta_rev_id":"941f755e444bf881d5a98f885c6900607aac7b08","_number":823175,"virtual_id_number":823175,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-30 09:35:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-12-30 05:22:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-12-30 05:22:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-30 03:11:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-12-30 03:42:39.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2021-12-30 05:19:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-30 06:59:59.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"}],"messages":[{"id":"4c774d96572f8d47cb2e64a7e5f39cfca1fc5538","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-12-30 02:51:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b58a27e1ce7f17e3a664ab49958047ae2d7d6282","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-30 03:11:52.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/e515fddbb18e4bb38c3d3cac28874160 : FAILURE in 11m 22s","accounts_in_message":[],"_revision_number":1},{"id":"096d7c8d50e4ec44e8a7ee3c487f4e2d9fa6c9a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-12-30 03:42:39.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"164749188c56a2943529dba9e4b82591df0a619e","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-12-30 03:43:00.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":2},{"id":"ab51ea6bc827fd774029c1a585e93944ca3896b7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-30 04:02:17.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/e94bbe2b581f49909323e99cb0a78b38 : FAILURE in 17m 39s","accounts_in_message":[],"_revision_number":2},{"id":"a510bc2026134c55df884de11828d3e00c485239","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-30 05:19:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c172b12aeb504c8ea9f03e74a85dbd3a : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d940320aeca4449b7abe459941153be : SUCCESS in 3m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/884db1d5625b46f881147a4ba03de339 : SUCCESS in 4m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9f35b1cc2dab4074bd99f3a6ae397093 : SUCCESS in 3m 52s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/12764607ad754c92aac5afbeed963ad7 : SUCCESS in 4m 26s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/62be270e301e442291598e492bb7b109 : SUCCESS in 4m 10s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/de3202280a684d1ab272fe14b59beaff : SUCCESS in 8m 10s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f9831b10a88d42478d963d7dbb14c1d3 : SUCCESS in 9m 47s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/13b20fe4929a4f6c8f512e1fd11fc5cd : SUCCESS in 5m 05s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/c7f742a73e43488f80c61754519d8cfb : SUCCESS in 10m 40s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/5f4f8609d2fc4d828cddfb41f814e7ba : SUCCESS in 5m 34s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/ba026e22400a49d2a7ce12970fd2989d : SUCCESS in 10m 45s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f7199e458ebf4d448fc61cecc9a5de63 : SUCCESS in 8m 43s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/176f19e73e2d4f919b8dbe72fa5322f4 : SUCCESS in 8m 11s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/4d618d4f5ebc413998deafc281a7d63e : SUCCESS in 14m 36s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/0cc03131d61c42ae9e81d2802988ab15 : SUCCESS in 4m 45s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/3ddfc16971e1431d9480588d4d2eaf8c : SUCCESS in 17m 11s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/05651ba551204e65ba176f5c4d92f133 : SUCCESS in 5m 20s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/9e69b44ec97142d4b2a98bb7ce817a1a : SUCCESS in 27m 23s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/cca73520110740c19043cf47a2fd9270 : SUCCESS in 12m 27s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/95e7c936390641ffa432e0618ae051b9 : SUCCESS in 5m 31s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/153a5384353e4d9e84c7613e32b6a3e5 : SUCCESS in 4m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1b9d3d20c4694dcaaf65afef01e1f584 : SUCCESS in 18m 17s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/29c130369efa4ff78076cf26f2ab1752 : SUCCESS in 7m 43s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/e8e7d7d712ee438e98688a07309189dc : SUCCESS in 6m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b442d2fa0dbf44b8b22dc202e97a66d8 : SUCCESS in 6m 46s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/3fae988912d941c68958006c0b9e5f67 : SUCCESS in 12m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bb08339a769c471db94d9b89973aa6a1 : SUCCESS in 40m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/da8c6ef08c364f7c9837582e18df0881 : SUCCESS in 1h 31m 35s","accounts_in_message":[],"_revision_number":2},{"id":"e4374308d5ebc49921124862265c3918adbeb79f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-12-30 05:22:04.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c6be0b63d9b08424feb565d0a616256b633cf272","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-30 05:22:48.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8a5da8f33fa6b7e69b4b5ec737108f24f7ce2bf4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-30 06:38:25.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49e1a6281fb94769aa1304326255dc2f : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab9f094f03214f66a767fffedc4975b7 : SUCCESS in 3m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cf18ec363c41408e8d4c81172ae0c8b4 : SUCCESS in 4m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/98b76fdfa4924972b9d8299c50f435f2 : SUCCESS in 4m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b49156f13d9e427fa6775b5e9bc26e40 : SUCCESS in 4m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/466fa4f892694187a413914965305e93 : SUCCESS in 3m 44s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/27d2fc03dc134a7b803c1844210d2a4c : SUCCESS in 8m 34s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8c9bd27dd8fc4381ae8b929ad0b380ae : SUCCESS in 12m 59s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/0deb2af99ef249b986e036b4a19aa6ac : SUCCESS in 12m 12s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/a978b69cc03b4ef09918b761e1a39148 : SUCCESS in 5m 19s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/8138c416607542c5a675c85da55043ca : SUCCESS in 10m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/24a5394e387c484bb3cb6f4e3c533287 : SUCCESS in 7m 15s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/27fd057e8ab1430e85fd0be4fabe2f99 : SUCCESS in 7m 12s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/5538284bb8b640faac0b786ece50094a : SUCCESS in 15m 40s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/5d0033994be44d54a2d3ad4cb1680a78 : SUCCESS in 3m 53s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/b7801c03c6ce4e50ad03cd01dff562c5 : SUCCESS in 23m 10s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/69bce6b54e1c4cd480a51373b739ff91 : TIMED_OUT in 41m 23s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/456d5d7627de4bdf94250476ee0d972c : SUCCESS in 18m 58s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/5e3db0739b60454dbf62b7c9d3990fd1 : SUCCESS in 11m 12s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/91e7e4912d0d4aec9cd8abe213be9a4d : SUCCESS in 5m 09s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e4e2accd5576402ca6fba8e16615b234 : SUCCESS in 4m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a5b5901cb081436a90ad4daa2ba6dad4 : SUCCESS in 17m 06s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/51abedab40914e6fad91becdb8d4e030 : SUCCESS in 6m 19s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/8c7b8045b2ca4a8aacd68880bfbabcda : SUCCESS in 4m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7864321ea2bf462bb8180cd1c1678cc0 : SUCCESS in 6m 24s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/9543a593b6a24c6991f62c1b6302ec36 : SUCCESS in 9m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/993ef87eba734564a656ebf6106fafd7 : SUCCESS in 34m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a21621b59822472985996a515f0b44a8 : SUCCESS in 1h 14m 59s","accounts_in_message":[],"_revision_number":2},{"id":"e8e5c0cb9f930dd63c53133eb81e87188d8e623a","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-12-30 06:59:59.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"56199d0af42180d77a001df172a6aad614322696","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-30 07:15:12.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/5daf4d785adc4b52be2c4e27dfd00d2e : FAILURE in 11m 18s","accounts_in_message":[],"_revision_number":2},{"id":"16574af92e61f69a3cf989b369483f63eac617e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-30 08:22:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b20f3a6064ef4c8c916f9a17b0485ece : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d61dfa746f14a688559634e1a4b283b : SUCCESS in 4m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b453722bf0df40c3894395201e0e30cf : SUCCESS in 4m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c5d630724bb14f5bb8b814aa0c589732 : SUCCESS in 3m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/81acd54733264b4f8de866f92c1c9c7a : SUCCESS in 3m 12s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b76868923d8f48668c64822a2e1b788d : SUCCESS in 3m 33s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4dec8d1c175f4907ba0bd006a1191ab4 : SUCCESS in 12m 37s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/21bcbb626ef249f2ac7efca5a16a07ca : SUCCESS in 11m 24s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/f6f93c5bb7a74f3fa8bdffdd81330e9c : SUCCESS in 4m 54s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/75691e8aebe945bd994426e0c5236b0b : SUCCESS in 9m 21s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/e7a0e039e2b048f3b5a94b268937d011 : SUCCESS in 5m 17s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/e977e58d9c694f4f9bcf6390f6eab61b : SUCCESS in 11m 52s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/dfae6986120944c782be36c041433ca1 : SUCCESS in 7m 40s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/033856d5cc9345e2b870ff5ae54c8025 : SUCCESS in 6m 47s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/4defb15ff26049cb860b03ff31c0fba0 : SUCCESS in 14m 44s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/8b2be1d49d864ebb93535fc486a63bfe : SUCCESS in 3m 41s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/fb3164285b7c4a70b513bb1880530726 : SUCCESS in 17m 06s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/40356dbc24fa4cd4b815a36a9112b703 : SUCCESS in 4m 59s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/fef1625e97bf47a294f15e64e2b3d93c : SUCCESS in 26m 41s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/e574775fa45e4e30977b2bf280718ce7 : SUCCESS in 10m 30s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/cf4945ca5b3b499491d785a6a35023a1 : SUCCESS in 4m 51s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/83504da5559146598dbfc5d7c8ed2f02 : SUCCESS in 3m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/41a0cb1d26444bdbb438273ebfdce3ed : SUCCESS in 17m 28s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/d5f6b2a888764e83bb80a2056e6468fc : SUCCESS in 6m 23s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/9351976933234594a5756eec425c8727 : SUCCESS in 4m 34s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1e1714094845460091d7220a78bed1d9 : SUCCESS in 5m 49s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/64c49f54f1dc4c6590981b70b48fa4e6 : SUCCESS in 10m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/29acff5e2ed14459b8c09b1e77c28863 : SUCCESS in 39m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5337e53e3265482dab468139cb0baf41 : SUCCESS in 1h 20m 45s","accounts_in_message":[],"_revision_number":2},{"id":"0f1d01da8f8d2f58bac4c2f83f9f1d39edd22d68","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-30 08:23:21.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e1a9cc31b14ea6f2f0c3379aca116e83e5e1c3d7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-30 09:35:57.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a474473d903746bea18539e9258668f6 : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/482432d300b24fe3b9ab5c7ab8581243 : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/65094b6e095b498f87bacf96708716b7 : SUCCESS in 3m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4aa9106feab04220809ea4beab389ca6 : SUCCESS in 4m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3af3e8acc66240339807659b38f30077 : SUCCESS in 3m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b2995c5ce87e4ca7b56fa123790ff3a4 : SUCCESS in 4m 25s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/90cb5fff0132490ba4e256f99bcd05f2 : SUCCESS in 8m 20s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c390d65f59c74e17b4033f052be1d353 : SUCCESS in 11m 38s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/557a0bfd833845ca8409c3eaf68bd509 : SUCCESS in 9m 02s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/a81120ba242d40fe86df3eb948cb8bc3 : SUCCESS in 6m 28s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/9f748ee0ceb846f79df10cb294cd96fa : SUCCESS in 8m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/db124e7811ed4317982bc1b4f2add025 : SUCCESS in 8m 43s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/a71ac03d40924bab8c722767de761bd9 : SUCCESS in 7m 19s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/e5233cde1ebe447193522486533ff401 : SUCCESS in 19m 37s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/e30b57498c894c6e9b655dc8e7a117f1 : SUCCESS in 3m 50s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/1f23be8bc71f490fbe1de26ba6428252 : SUCCESS in 22m 46s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/cf9cf2bd4b8d442aa2d8a59c00adb8e2 : SUCCESS in 4m 50s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/0cf4f4c0630442e78dd079292a16e375 : SUCCESS in 21m 56s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/aa3b7994e13d4c54b829cb1b4576751b : SUCCESS in 11m 07s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/8ea9dbd3767e42bdbfe3ad86c5b066db : SUCCESS in 6m 08s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/cfd23f0e37c34444bc99da4cd4bea4c4 : SUCCESS in 3m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7430eded5a9d471d8cc817ca509ac6a3 : SUCCESS in 17m 19s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/3811e7cc8a5d46a0a651b64f5647db9b : SUCCESS in 6m 44s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/deb2920c9efb465abd8684633e8bce47 : SUCCESS in 4m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/40ba764743884fdabd5083be4c373b51 : SUCCESS in 6m 29s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/40e07c0289a34b3b939cbdb840565272 : SUCCESS in 11m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/592491bff4c147cbb9e6d07edd49043f : SUCCESS in 33m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3f1a431b276542a8aae0d72909e82d96 : SUCCESS in 1h 11m 22s","accounts_in_message":[],"_revision_number":2},{"id":"80dff7599426b2ec784d6e5a717b913a287fd952","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-30 09:35:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"941f755e444bf881d5a98f885c6900607aac7b08","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-30 09:37:20.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a3cbed9ec3b499eafc8193fd41b1243 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9fdc8d1e641e4e5c84a6805baa1deca308c98508","revisions":{"164e4282095757f9a40e0eb47d1ab82f2b89b2dc":{"kind":"REWORK","_number":1,"created":"2021-12-30 02:51:51.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/75/823175/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/75/823175/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/75/823175/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/75/823175/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/75/823175/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/75/823175/1"}}},"commit":{"parents":[{"commit":"0f255c347e768aaa3dfa02fc41f45056a7182cdf","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0f255c347e768aaa3dfa02fc41f45056a7182cdf"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-12-30 02:51:50.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-12-30 02:51:50.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I1116112bf1fbe152fd7616eaf984db9934c2ceeb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/164e4282095757f9a40e0eb47d1ab82f2b89b2dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/164e4282095757f9a40e0eb47d1ab82f2b89b2dc"}]},"branch":"refs/heads/master"},"9fdc8d1e641e4e5c84a6805baa1deca308c98508":{"kind":"REWORK","_number":2,"created":"2021-12-30 03:42:39.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/75/823175/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/75/823175/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/75/823175/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/75/823175/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/75/823175/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/75/823175/2"}}},"commit":{"parents":[{"commit":"0f255c347e768aaa3dfa02fc41f45056a7182cdf","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0f255c347e768aaa3dfa02fc41f45056a7182cdf"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-12-30 02:51:50.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2021-12-30 03:42:23.000000000","tz":-360},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I1116112bf1fbe152fd7616eaf984db9934c2ceeb\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9fdc8d1e641e4e5c84a6805baa1deca308c98508"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9fdc8d1e641e4e5c84a6805baa1deca308c98508"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
