)]}'
{"id":"openstack%2Frequirements~823219","triplet_id":"openstack%2Frequirements~master~I7911d252e71147f026c43df62a41d3569bb1278c","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","attention_set":{},"removed_from_attention_set":{"14288":{"account":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"last_update":"2022-01-03 17:58:35.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I7911d252e71147f026c43df62a41d3569bb1278c","subject":"Updated from generate-constraints","status":"MERGED","created":"2022-01-01 02:51:38.000000000","updated":"2022-01-03 19:31:59.000000000","submitted":"2022-01-03 19:30:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"823219-openstack/requirements/constraints/noclob","meta_rev_id":"182234a405821e9f882e46379227e2e59f8c0109","_number":823219,"virtual_id_number":823219,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-03 19:30:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-01-03 17:58:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-01-03 17:58:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-01 03:08:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-01 04:10:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-03 15:52:57.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"3200265a3692c72c0698c0b7c33a1cb34f41b844","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-01-01 02:51:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aa2b615eaece8058c2bcf79b8cad38ef8d73a086","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-01 03:08:22.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/de7463d70c824296ac89ac7a6c8ae382 : FAILURE in 11m 45s","accounts_in_message":[],"_revision_number":1},{"id":"17c0bee0dea06591d0e635d52cf6afc75151526b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-01 04:10:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/402cfaddab2e42c38e8d3120022b5d57 : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0ae4a65bde44eef818568abda1f5f3b : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/76488d8e63804f3f971d2e3cf48ba973 : SUCCESS in 3m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/af6a6d17977a4cb896d0748fabc44589 : SUCCESS in 4m 03s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b1de3f36bd64451bb5b38bf513e80a52 : FAILURE in 4m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1c69b589571f4f6fa2f90b7b0191bc6a : SUCCESS in 3m 51s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/396881483ba4415e9840b9cd8fa88861 : FAILURE in 4m 53s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/07d19018332b4ea3ad894be03c808f65 : FAILURE in 4m 12s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/d354b7ec948147c78ed391300f0c6024 : SUCCESS in 7m 05s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/5f06e5bfc07f47a4a0c61d554e115361 : SUCCESS in 10m 11s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/e06d8c5e05294e0a8c5ddc01ceab1214 : SUCCESS in 6m 29s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/da17195304a248958eb6a3a2fa4e1004 : SUCCESS in 8m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ee609e51366b495c8c6c88fdf0c1112d : FAILURE in 7m 36s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f963f933b3354457925b5a6634fae213 : SUCCESS in 8m 33s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/b7a7ba06218a4211a1fa075a31704f66 : FAILURE in 17m 55s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/29cb06c637064f8782bd659c591de556 : SUCCESS in 4m 41s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/2f144482f70c47fc923c8bf22c097090 : SUCCESS in 15m 34s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/20437a04cc2c43c78adc455bab0ac15b : SUCCESS in 5m 48s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/11cb8bb4c20a4394a6aa358b0cc16186 : SUCCESS in 20m 44s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/750347d5c9a04248b8e1a2dde4c3812d : FAILURE in 12m 48s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/1ad1e704e66449edabca5bcce796b3ab : SUCCESS in 6m 36s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/974b5fb65dfc48bea78313e9662bccef : SUCCESS in 5m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1ce197e387174122b10bfe78f838bb30 : FAILURE in 24m 10s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/49596d016c8c4dfaa060c0327de9da5a : SUCCESS in 7m 40s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/482a3fe2cfff4c5496c44a99018f4d43 : SUCCESS in 6m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a66bd4cdf9804237ae7df55526b7d464 : SUCCESS in 9m 28s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/75fa135402cb466a83e0396d9cf1f45c : SUCCESS in 12m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5d3a5d9a0ee6469199915a7e53b6f277 : SUCCESS in 43m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e74c60aa36f9494fbbc42983d27a79eb : SUCCESS in 1h 14m 22s","accounts_in_message":[],"_revision_number":1},{"id":"1c572082f01b5ce3fdb56bdc67502e638f063965","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-01-02 02:56:32.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6dd0ebf688a750ed7b72154891e6efdc63d28631","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-02 03:11:17.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/014fd7cfdcfd44f985600dbbb1a259b0 : FAILURE in 12m 05s","accounts_in_message":[],"_revision_number":2},{"id":"e1d96cf3d6013bc10707d113a10c47c1488101d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-02 04:03:58.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6b08f12d5514892baeeebe5b331c21a : SUCCESS in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c035020838264808ba7c37b940499789 : SUCCESS in 3m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/be226e6f2753437ca91c56337c950ea9 : SUCCESS in 5m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7372d08b273c4b0097141d3cd6234055 : SUCCESS in 5m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1073213c92c14b02aba9b13de732b8ff : FAILURE in 4m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6c965afe9c874d0aa4709080634df3ac : SUCCESS in 5m 38s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/25b105b2c7b04eeb9160905ec336bd30 : FAILURE in 5m 56s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f94150ce45304df9aa54c3578bd59779 : FAILURE in 6m 02s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/fdc485bb3ad447d8bf1158d13a8b72a5 : SUCCESS in 8m 24s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/f03ed570e1184257bcc039698dc8a17e : SUCCESS in 11m 38s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/4d498535412442bb8e04ded74bf69bab : SUCCESS in 6m 37s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/956bd6dbd6794fbf9f1944dd517d028e : SUCCESS in 11m 58s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9031de5e43ff4160b960fb24d0580132 : FAILURE in 8m 14s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/c609942fd6f042e1a3383db528a3d279 : SUCCESS in 7m 57s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/e566a6c9561d4754bc0036c1f4f21538 : FAILURE in 38m 03s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/79bb482fa16843b89cd3f93746affa23 : SUCCESS in 4m 02s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/99509c4930564285a56bfed1ba255d96 : SUCCESS in 19m 51s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/9d91303578074b2e829c01c5e44e2814 : SUCCESS in 8m 26s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/bd5fe0aa973a46868aef952b63d32556 : SUCCESS in 18m 51s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/5f72d9b0ccaa4b70afb6b0bc98e63f0d : FAILURE in 11m 12s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/e6644a763198406a9e10b71f02042875 : SUCCESS in 5m 43s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5d6f870e352d44fd8903f0fcd0c5c79c : SUCCESS in 4m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fc5387aa28da4aa482eae8c400683a38 : FAILURE in 18m 12s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/8d757277c15f47d39c92558a410ab802 : SUCCESS in 6m 22s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/70a2e71f3a034477af831854ae0a51ac : SUCCESS in 5m 25s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b4dc63d6d51947af971127fba4027875 : SUCCESS in 6m 25s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/011bc4e2cc044477bddaf2d30458132d : SUCCESS in 11m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0cd390d4601a48c787dd34ee6a214b49 : SUCCESS in 49m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/43649135f8a94971b8a276d7c38f1416 : SUCCESS in 1h 05m 35s","accounts_in_message":[],"_revision_number":2},{"id":"8a07412efa1df2e2036c08406e0c8417562291c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-01-03 02:57:12.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"019209c5dbd1378fed831dd8b3c9dc91451e5b58","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-03 03:10:17.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/f6a706e38815450fae7e2b68fcceb829 : FAILURE in 8m 21s","accounts_in_message":[],"_revision_number":3},{"id":"4fd5ccd759102dbfb33d88b6a9a4d01094d5e497","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-03 04:04:12.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/050d7b265a8644789ffdec91e7604521 : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0da1b75e9fd42beb186efaf430c1299 : SUCCESS in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2d4a3eeccb874653ae4d2027e23647ab : SUCCESS in 6m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/400ac20c22ad4d23a4b51e2278956605 : SUCCESS in 6m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a7f5f453d51c47369f7820b6c82e1993 : FAILURE in 4m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a9f82685e7d341cf8e594d44406d6508 : SUCCESS in 5m 19s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c19f4f796e0845acb684184b6168a8c9 : FAILURE in 5m 56s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/44bf5d28195644b4a4eeaa1d94ff11b5 : FAILURE in 5m 07s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/d583df8f3e8b494eb8b7862f7853de49 : SUCCESS in 8m 25s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/ff6cdfd85fa24dc180d2324872b65856 : SUCCESS in 12m 46s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/fcf39ade263148e2b82736559350cf8a : SUCCESS in 8m 06s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/15731772b8e44cc38df72569d300d824 : SUCCESS in 9m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/42877c1de1e0443b8071c89715ad1061 : FAILURE in 8m 52s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/34f2576b141c40f6991a45aef3d7b450 : SUCCESS in 10m 49s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/ec4426f9af34423dadf2ec5a356e979d : FAILURE in 14m 20s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/8ad1ed2d4e31492ba1eec1ef6b01d939 : SUCCESS in 5m 59s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/77c6f717c62b4848a2ec9d16effe1b52 : SUCCESS in 16m 13s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/eefedcc16c7a4a3eb5ca42ff97483c07 : SUCCESS in 6m 43s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/ea9368822ff140f9866766011eb1e0a6 : SUCCESS in 54m 08s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/9b37d6299fbd49f5896a73fba4646505 : FAILURE in 12m 02s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/c1cb4ec808f04707a36da7f2e637a95b : SUCCESS in 5m 55s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e6a9aad884de49dd818c775c472ce5c0 : SUCCESS in 5m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/294bb058531247398e07ea81843fe638 : FAILURE in 21m 13s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/7648586ac91b499bbb35e849b2a2505e : SUCCESS in 5m 54s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/c7ba594a39e64e7784d7b9b1734113f0 : SUCCESS in 5m 23s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/94617d01ecb84309bfabfb819c6cce15 : SUCCESS in 7m 31s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/367c111af1974baa916c916f9caa1f20 : SUCCESS in 12m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0dcf8248ff35464cbe6a6536e1cebf10 : SUCCESS in 41m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0a0d73dd17464c409ce5f2fcf1914afb : SUCCESS in 1h 02m 53s","accounts_in_message":[],"_revision_number":3},{"id":"d70f74ce337c3779b4d538df4bfc44cce5da33aa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-01-03 15:52:57.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"fa1574bfd17c00447095d6c8f682c2154f3d76f6","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-01-03 15:53:01.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":4},{"id":"cd1c8d11bdf64ed0b290d6cc0b748bde8cf2e7ab","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-03 16:10:11.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/8a802d9ff85b483789468a41aed37577 : FAILURE in 15m 09s","accounts_in_message":[],"_revision_number":4},{"id":"29bce743fb012066e3c1d16c9739abd9aaf3031c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-03 17:24:26.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38d52cc453b840a4ae296aefd07cd2a6 : SUCCESS in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c5ccbc4b85e42beb5eed2881733bf0b : SUCCESS in 5m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/69823ff78be3478e8e01dbf897b12cae : SUCCESS in 4m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b338e12a9d174d0da59fa89349b5d83a : SUCCESS in 6m 12s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ad2aabf0ab24444d88dd88ad763b0fec : SUCCESS in 3m 47s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/51b40dfd8a7d4707b8ec7c8de4be284a : SUCCESS in 5m 18s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d8ac05d9aab04ba4a45239ad7372e563 : SUCCESS in 16m 02s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d2a7d126836c45a6a86fb13c1eeb196a : SUCCESS in 13m 38s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/85f902b0c76649b581dff7133346962a : SUCCESS in 6m 02s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/f59d4379e3624140931a15a760516c81 : SUCCESS in 11m 01s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/e9e5063c06e7417d8a684b7dd337eaf9 : SUCCESS in 6m 06s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/650179f7f8b64f3cb2358a94146563f6 : SUCCESS in 11m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/01746256a661454abbf351e14977e706 : SUCCESS in 8m 20s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/4931ce982e114d5da8e6733065f750a0 : SUCCESS in 7m 49s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/2628e50874934a0386d81abbeaf49e80 : SUCCESS in 15m 13s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/303927c3d8d143a3a64879483cee5134 : SUCCESS in 5m 28s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/6fcbbf7c8f6d4f9a99eb92b1629f3e98 : SUCCESS in 19m 39s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/22662d498dbe47268a5ddc0b4b413cf8 : SUCCESS in 5m 29s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/f2d0fd8f638a49b0a47247aaded1114e : SUCCESS in 24m 41s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/f6645a0eda144de5a23d69f3e730c678 : SUCCESS in 12m 21s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/30f8483706694d698df519998429bb97 : SUCCESS in 6m 10s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0585f86a8dc64c548cb9b8be27ad2e5b : SUCCESS in 3m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/debc1893ae4c4db792c6a2b0c4cd6376 : SUCCESS in 19m 24s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/6e5c162575a742bc944d364357746931 : SUCCESS in 6m 06s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/9b4dbd2ff2e2475596cb330f3bd0ca2b : SUCCESS in 5m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/02ed16c0f18f489ba9a7805cad4804ee : SUCCESS in 6m 45s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/e91bdfc0b9054e83baa580fbcd912136 : SUCCESS in 11m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3236bff1239248669cb809a8bc8799b3 : SUCCESS in 49m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dbea6621352349df806f9b9ae370b64c : SUCCESS in 1h 25m 53s","accounts_in_message":[],"_revision_number":4},{"id":"bb8b7eea3b562ab60eb8bd5edb2318ef6950373e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-01-03 17:58:35.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"4061ee3f9d8d4dbd72096ae03749f08eb5e3f2e0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-03 17:59:30.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"dbf7293506942aedef93b634d20b9e62a85e8a20","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-03 19:30:36.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57ef751ef69e44b48665fd3656873298 : SUCCESS in 5m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4836aaa5e10543b297443a1623487f07 : SUCCESS in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/989ea9c89b894c1e975033ddfa08e2ba : SUCCESS in 4m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b008d8e4f8d54bdd9367f3e4796ddff3 : SUCCESS in 4m 41s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3c602485664e402ebae864b3017fef1c : SUCCESS in 4m 45s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c2354d8875774538b93ff8771ae90309 : SUCCESS in 4m 11s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/53b3bcda900741c88669aeaf532c2303 : SUCCESS in 10m 47s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5249a75d8c4a4dfea7de499b18a3fe7c : SUCCESS in 11m 14s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/7641e81cf26b48908290a6384a859dec : SUCCESS in 11m 58s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/b221381643754050ac0ac91d862d92fd : SUCCESS in 6m 29s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/f657e292c96d4e01a4efeba26ab820f5 : SUCCESS in 11m 14s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/af1b9ce38b7e4e259747e49bc46e6b2d : SUCCESS in 8m 49s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/1a7edf9ca3a74605a4011ffb7a252ba9 : SUCCESS in 12m 51s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/4d222f6c3f0c4c3889fd794a1ac89143 : SUCCESS in 15m 37s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/ace0cf2cb14141d19462fa49b8eac953 : SUCCESS in 4m 32s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/275cfc84783d446c9b2085948351d447 : SUCCESS in 16m 46s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/d2316d7ceff34eecbee73677a8d8f68a : SUCCESS in 6m 32s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/a4f985612d2c4c368301e30b910d00a4 : SUCCESS in 25m 43s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/762ccfe990ac4d72999256223547eb05 : SUCCESS in 12m 32s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/6d2786d0174f44d6af4df63fcfdec3a1 : SUCCESS in 6m 59s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4ba7db7d2e7f429382300405101ff314 : SUCCESS in 4m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4f5a5ffec1b5438d85d4f2e1a52a96ea : SUCCESS in 20m 09s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/c583b41dd6c34f04bee5177e882b4e09 : SUCCESS in 7m 47s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/979d0ae5e2f24f718a43f43debe817a6 : SUCCESS in 5m 53s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b555043892e344c0ad4da99b29364181 : SUCCESS in 11m 02s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/49201af2ed1d424da134d5a38a1c6560 : SUCCESS in 17m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bc25413c12fe49409f4b0232e719648a : SUCCESS in 52m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d451e8ca849842c5b75a0fbb66f39390 : SUCCESS in 1h 29m 50s","accounts_in_message":[],"_revision_number":4},{"id":"fc846f21f5afb52b19ce4f65ad83552b731988eb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-03 19:30:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"182234a405821e9f882e46379227e2e59f8c0109","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-03 19:31:59.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7c5e0e6018348eba9ff5161027be63f : SUCCESS in 53s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"8473e174791f2b3a8a693de0c6ffcdbacfc22431","revisions":{"8d36f43417ba57bc0f2757637962afe7c0a17297":{"kind":"REWORK","_number":1,"created":"2022-01-01 02:51:38.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/19/823219/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/19/823219/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/19/823219/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/19/823219/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/19/823219/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/19/823219/1"}}},"commit":{"parents":[{"commit":"9fdc8d1e641e4e5c84a6805baa1deca308c98508","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9fdc8d1e641e4e5c84a6805baa1deca308c98508"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-01-01 02:51:34.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-01-01 02:51:34.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I7911d252e71147f026c43df62a41d3569bb1278c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8d36f43417ba57bc0f2757637962afe7c0a17297"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8d36f43417ba57bc0f2757637962afe7c0a17297"}]},"branch":"refs/heads/master"},"2bcedcd7bbde572c65c3f9a04dda5cf4b791bcaa":{"kind":"REWORK","_number":2,"created":"2022-01-02 02:56:32.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/19/823219/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/19/823219/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/19/823219/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/19/823219/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/19/823219/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/19/823219/2"}}},"commit":{"parents":[{"commit":"9fdc8d1e641e4e5c84a6805baa1deca308c98508","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9fdc8d1e641e4e5c84a6805baa1deca308c98508"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-01-02 02:56:30.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-01-02 02:56:30.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I7911d252e71147f026c43df62a41d3569bb1278c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2bcedcd7bbde572c65c3f9a04dda5cf4b791bcaa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2bcedcd7bbde572c65c3f9a04dda5cf4b791bcaa"}]},"branch":"refs/heads/master"},"1019af74e4fb24b49c315ef847939726de82b413":{"kind":"REWORK","_number":3,"created":"2022-01-03 02:57:12.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/19/823219/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/19/823219/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/19/823219/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/19/823219/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/19/823219/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/19/823219/3"}}},"commit":{"parents":[{"commit":"9fdc8d1e641e4e5c84a6805baa1deca308c98508","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9fdc8d1e641e4e5c84a6805baa1deca308c98508"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-01-03 02:57:11.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-01-03 02:57:11.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I7911d252e71147f026c43df62a41d3569bb1278c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1019af74e4fb24b49c315ef847939726de82b413"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1019af74e4fb24b49c315ef847939726de82b413"}]},"branch":"refs/heads/master"},"8473e174791f2b3a8a693de0c6ffcdbacfc22431":{"kind":"REWORK","_number":4,"created":"2022-01-03 15:52:57.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/19/823219/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/19/823219/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/19/823219/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/19/823219/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/19/823219/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/19/823219/4"}}},"commit":{"parents":[{"commit":"9fdc8d1e641e4e5c84a6805baa1deca308c98508","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9fdc8d1e641e4e5c84a6805baa1deca308c98508"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-01-03 02:57:11.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2022-01-03 15:52:37.000000000","tz":-360},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I7911d252e71147f026c43df62a41d3569bb1278c\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8473e174791f2b3a8a693de0c6ffcdbacfc22431"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8473e174791f2b3a8a693de0c6ffcdbacfc22431"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
