)]}'
{"id":"openstack%2Frequirements~823264","triplet_id":"openstack%2Frequirements~stable%2Fxena~Ief1b9bef9f1b309910f1c4b16fb3082c723abb35","project":"openstack/requirements","branch":"stable/xena","topic":"new-release","hashtags":[],"change_id":"Ief1b9bef9f1b309910f1c4b16fb3082c723abb35","subject":"update constraint for aodhclient to new release 2.3.1","status":"MERGED","created":"2022-01-03 10:25:00.000000000","updated":"2022-01-03 18:50:17.000000000","submitted":"2022-01-03 18:50:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"823264-new-release","meta_rev_id":"794a4567be8cb6eccb7ece3af6f2f540d4836abb","_number":823264,"virtual_id_number":823264,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-03 18:50:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-01-03 15:50:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-01-03 11:05:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-01-03 15:50:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-03 10:43:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-03 11:05:25.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2022-01-03 11:50:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-03 15:50:30.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"5198fb9d2d2634057d7072484d1800a05791d50b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-01-03 10:25:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e9eb040ce4dd3ede4a7a1f55f4460e856ec59c0f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-03 10:43:33.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/449f8ee6f4f646b0a90a68f6e810ecda : FAILURE in 13m 50s","accounts_in_message":[],"_revision_number":1},{"id":"e03f3eee1401a660a38e285685d1cbee6b719c40","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2022-01-03 11:05:25.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"399c2e0ae1af7b4b118256b0f584aff46bf4366f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-03 11:21:42.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/a3097bbe1db04272ae6196cf0e467ef7 : FAILURE in 14m 13s","accounts_in_message":[],"_revision_number":1},{"id":"c011d56e3cb2aed33b0e60644e2919bed7509c37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-03 11:50:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c2d830f41464cf2830ce3d66fe9151a : SUCCESS in 3m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/556a80b9b67a4d0b82544761f6f265af : SUCCESS in 3m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2df5d0b64fb341319d06eeaf9566da48 : SUCCESS in 3m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f02387b4d97a4e54860a91c52222dfbe : SUCCESS in 4m 21s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f0ad0f8f3d944df9aeeea4b9900e28cb : SUCCESS in 4m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fb659f4d13394fe3af31225bac5906ca : SUCCESS in 3m 29s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d88c0ae43d7e4978a7f6f311eb575b1a : SUCCESS in 9m 32s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/095276d5e4544b0d9f91e3179162616c : SUCCESS in 9m 35s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/8ff1a2036d40449eaadaa5b11ab3ddd0 : SUCCESS in 5m 58s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/2d5a9f13a3ca47e58a0a141609539928 : SUCCESS in 9m 24s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/e9c99bbcc7534a4a81781ebad4a64ea9 : SUCCESS in 6m 53s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/58c526c4a92c42b9a59d505ca6332138 : SUCCESS in 10m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/29630315061045caab232a9ed0fef18b : SUCCESS in 7m 42s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/c9e588fcb93649a3a64c7ef9f84c8b78 : SUCCESS in 9m 11s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/2746b38ea02a4cf2890f06d3eec76172 : SUCCESS in 31m 45s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/1f6a68d523c241ac98fc6d224ccb101a : SUCCESS in 8m 58s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/6c98a3086b274c85a020f63f8229df84 : SUCCESS in 4m 45s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/972f2b17d3e744aba82333f2dc314ee7 : SUCCESS in 23m 36s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/19377606380b42d7bfb8a5c9a3f5cf9f : SUCCESS in 12m 25s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/14c800527d86483e88b0438485d83afe : SUCCESS in 5m 01s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e1ad1234b7fd4914a6710f440b2f0523 : SUCCESS in 4m 02s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ecd78f7d94e84ef5b3bfa934f21ab594 : SUCCESS in 18m 00s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/f9e2c1d40a08460bb80f9a0941b4bb62 : SUCCESS in 5m 45s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/a34d634c57dc41a2a3732b4f43a7ffd5 : SUCCESS in 5m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/574ce14849fb4b5da66f9f0ad9791b17 : SUCCESS in 6m 17s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/19d68dae3d194cd2b389d784426631d9 : SUCCESS in 11m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/aefb675d28b54606b853da1c606ec4fa : SUCCESS in 46m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7a55bfda5fad4adf94c699981261700e : SUCCESS in 1h 20m 45s","accounts_in_message":[],"_revision_number":1},{"id":"95fd7e75d4e96fb81fee4464615bf3939051a3a2","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-01-03 15:50:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4a43f71657490b6c206df5eb8b55f29ee6032e76","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-03 15:52:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dc1665418b3dd15e097660c3d51be90bacc6e984","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-03 18:50:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c5c28119e1944a4af13b7df93bea3d2 : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ffb13c5e58834ba5bdeb5aebbb390e46 : SUCCESS in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/244ea598ff75449a9b84f04d4e9cb116 : SUCCESS in 6m 20s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8865b5532a2343e1af5fc21a0437ece2 : SUCCESS in 5m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ee91afbce70b46e2b4eed11b1f65160b : SUCCESS in 5m 52s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a6994d40a160475fb155f49a320942d2 : SUCCESS in 7m 24s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ec20ccf6fdcf4fb6b2a1f12873e65c1c : SUCCESS in 9m 47s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/1b28b74c910f405cb0dcaeca99fa975f : SUCCESS in 10m 45s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/fa28b5369c82443db70bc04f84e7bd74 : SUCCESS in 7m 44s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/6d4f12fd9f454127b7124a7fd34a4098 : SUCCESS in 11m 14s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/16f48883a8bf440785c970b7e6d22ede : SUCCESS in 9m 38s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/5c0b8099a72c46dd80448454e22d543b : SUCCESS in 10m 18s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/ac4f1bb000204c059755af92e2c0f9fd : SUCCESS in 14m 14s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/dec0d5cb876d41c18a4403a7f0aa7fba : SUCCESS in 10m 21s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/5e05a05a40f74ae48b3761f587addcb9 : SUCCESS in 6m 22s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/a7f5c8049942436a85ecebd2d287c928 : SUCCESS in 21m 54s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/c8880902837e447eb97fc9cf6f8709a7 : SUCCESS in 12m 30s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/4caa328300e146e6bc0fd3b47471520b : SUCCESS in 7m 09s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b06ed80c771841388011cac865ee5d86 : SUCCESS in 4m 49s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c8d4caf7819347248f6b75419bf7a334 : SUCCESS in 19m 08s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/cb24e5ab859440b3abfd24db3b369504 : SUCCESS in 7m 56s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/1074a402d28b4272bd7351e515d69510 : SUCCESS in 6m 04s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6641a87260c441e5b06c83f212b156bc : SUCCESS in 9m 14s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/5124fc8043fc4d0eaa7730fa735db019 : SUCCESS in 15m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3094a0f5035a4af4a097ffd66c9334f5 : SUCCESS in 43m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e667f57571ec4159a2763116cb8f9771 : SUCCESS in 1h 14m 27s","accounts_in_message":[],"_revision_number":1},{"id":"794a4567be8cb6eccb7ece3af6f2f540d4836abb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-03 18:50:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4e8983fc1ad9d96202791ec7ab93128191061926","revisions":{"4e8983fc1ad9d96202791ec7ab93128191061926":{"kind":"REWORK","_number":1,"created":"2022-01-03 10:25:00.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/64/823264/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/64/823264/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/823264/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/823264/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/823264/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/64/823264/1"}}},"commit":{"parents":[{"commit":"f243f496e015d7904b2ca1973ab484bdc02d9458","subject":"Update amqp version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f243f496e015d7904b2ca1973ab484bdc02d9458"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-01-03 10:24:59.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-01-03 10:24:59.000000000","tz":0},"subject":"update constraint for aodhclient to new release 2.3.1","message":"update constraint for aodhclient to new release 2.3.1\n\nmeta: version: 2.3.1\nmeta: diff-start: -\nmeta: series: xena\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Eyal \u003ceyalb1@gmail.com\u003e\nmeta: release:Commit: Eyal \u003ceyalb1@gmail.com\u003e\nmeta: release:Change-Id: I5db4ddba7a6ffbc6ddd95dcb62d3f9cea3198ba1\nmeta: release:Code-Review+1: Eyal \u003ceyalb1@gmail.com\u003e\nmeta: release:Code-Review+1: Matthias Runge \u003cmrunge@redhat.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Ief1b9bef9f1b309910f1c4b16fb3082c723abb35\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4e8983fc1ad9d96202791ec7ab93128191061926"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4e8983fc1ad9d96202791ec7ab93128191061926"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
