)]}'
{"id":"openstack%2Frequirements~823337","triplet_id":"openstack%2Frequirements~master~I51547efb797dae50a2a600be0c5724cf5784526d","project":"openstack/requirements","branch":"master","topic":"new_release","attention_set":{},"removed_from_attention_set":{"19134":{"account":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"last_update":"2022-01-04 19:24:07.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I51547efb797dae50a2a600be0c5724cf5784526d","subject":"update constraint for aodhclient to new release 2.3.1","status":"MERGED","created":"2022-01-04 08:43:34.000000000","updated":"2022-01-04 19:29:11.000000000","submitted":"2022-01-04 19:24:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"823337-new_release","meta_rev_id":"11d3aca8dd93b94720724a42d8329ba46dad991d","_number":823337,"virtual_id_number":823337,"owner":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-04 19:24:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-01-04 17:40:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-01-04 17:40:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-04 09:02:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-04 10:15:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-04 17:40:29.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"e4b64ca7660c8dc932d338b11515e44515fe5d1a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2022-01-04 08:43:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"00752bf0a1f076abea54782d2431143533f64a59","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-04 09:02:52.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/d569d79cfcee410d88bf56219daf76fe : FAILURE in 17m 20s","accounts_in_message":[],"_revision_number":1},{"id":"f43fb04c5da7abd7c468a0920446b9a441b3fc4d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-04 10:15:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc08bb8040674835bd216c044d135970 : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82b8df66720c47feb97cae457cf32b8f : SUCCESS in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f6ce78c176c14e47a1fdf97c820820e7 : SUCCESS in 4m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8139a6048e1f46e5af2e4466634f06f6 : SUCCESS in 4m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/352d8cda7cff41c09b88adab4b9fcccc : SUCCESS in 4m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d5631519628c49e1803bdab17c54a821 : SUCCESS in 4m 21s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b97ab21972704e86957dcccac0f85ee2 : SUCCESS in 9m 40s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c626a975e5e543beb85260c918defc01 : SUCCESS in 10m 09s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/532ad079f9934bc1b6b9c05ffe195a1a : SUCCESS in 5m 51s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/82cf080a2b34485080089470f4ac77e5 : SUCCESS in 11m 38s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/6076c7a2d882458a8eef22ceed620ff8 : SUCCESS in 6m 29s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/3f6529510b774c5193a4ad5b06feb035 : SUCCESS in 10m 45s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d730513734d748bf90d246ec9b6829c0 : SUCCESS in 8m 03s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f2c2c5aac8714101b9618a7c67a32f91 : SUCCESS in 9m 05s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/8e5b55985e384166a31d51bf7f325237 : SUCCESS in 14m 22s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/9b728c6998c3406696c718132248bc35 : SUCCESS in 4m 43s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/3fc232a5f09d443a8e713c9002a6615d : SUCCESS in 16m 31s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/7f3f861247484ada96d7d2c438229b60 : SUCCESS in 5m 57s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/15262117660143219f03f579fbf7b57b : SUCCESS in 27m 29s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/1a8a834081144a648b5c2f64ac0940f0 : SUCCESS in 12m 24s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/0ffe04017ace4d2dadcb84323cf4408e : SUCCESS in 5m 59s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b8d82ecdae234170b113c2ffc61907e7 : SUCCESS in 3m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6ffafcd245834e049f56afc0e5c42bea : SUCCESS in 19m 34s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/0972ca683601404c89fe0e3ff23b71bf : SUCCESS in 5m 59s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/6a68ac707f944822857d14015fe2deeb : SUCCESS in 5m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/95165e3e30f84456a9177647a06f5ffb : SUCCESS in 6m 59s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/1eaffdf6492d4407a75aa1aa3bcd4667 : SUCCESS in 12m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/38759010255144b6ba846ee65671653f : SUCCESS in 1h 05m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d7be980bcb274d159bc42f9afe760fb3 : SUCCESS in 1h 26m 24s","accounts_in_message":[],"_revision_number":1},{"id":"41995bb47ea59b1080a5fc963dc7e6b176d023bf","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-01-04 17:40:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"dc9f4d922575ed5ecd7074f72c6e260e64c4dbfd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-04 17:42:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"94947aa5211ce8f265273e184a59a40ce3559f05","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-04 19:24:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d439dc6ef76410ba93358c78cb313f5 : SUCCESS in 4m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a53b5bd411f3417ebc2cce9dec756a35 : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ae71068a355e4087b115c9c2ff866ebd : SUCCESS in 4m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/195fd0dc836344169e53ab5fcdd3acc0 : SUCCESS in 5m 14s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1a4446f2be68425682520a0ce0509ede : SUCCESS in 4m 42s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cbe161fefa954e90b9c45b3fa9ec48df : SUCCESS in 4m 47s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/61c0e532c0bb4a029b86a68427f8d979 : SUCCESS in 11m 00s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/216e7fbc81874a4eba39278b117df290 : SUCCESS in 11m 43s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/9d0690ab233140b49852f5c055dd6fa0 : SUCCESS in 11m 22s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/8fc93c44723c48989c5c590a8adc09bb : SUCCESS in 7m 52s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/d4e092bbc0dc42909466c66c026a1c18 : SUCCESS in 11m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7555322648d74ef992fc1b7a4167ef7e : SUCCESS in 9m 19s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/89e2b42331ae46b68b407073f59cda63 : SUCCESS in 9m 11s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/533d859cab804a5491444f42c242ff31 : SUCCESS in 14m 42s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/12665c4e271948ba98f5c846ad31c634 : SUCCESS in 7m 04s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/5d173903c7c744d794de74f5937792bd : SUCCESS in 16m 32s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/d31dfd96113b415e818f81a654218e54 : SUCCESS in 8m 12s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/f9450cb3b07848f484824ee4177e6ab4 : SUCCESS in 37m 03s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/f6d0d23414994146938d9f78474c56ae : SUCCESS in 13m 02s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/eb2f8ce95d984419bf98ea46c1f26bc0 : SUCCESS in 8m 30s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/15a560102e364a70aa312c822f5c0313 : SUCCESS in 5m 46s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f9a5a4ed560d49f3a29075d9a9360bf5 : SUCCESS in 22m 48s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/09ada413d27d4aeebd114dacb08c0530 : SUCCESS in 8m 19s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/3fd376c826274380bedeeff07e41dc5a : SUCCESS in 7m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b08d0a000c7746f4a6cc2501ed9148ba : SUCCESS in 9m 30s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/8400d7adb55f439ea11650cc4a71b099 : SUCCESS in 15m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/56bfa2249764453e90af4707ec623ba0 : SUCCESS in 50m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/99c30d4b1c6a48799eaca8b8623e2725 : SUCCESS in 1h 29m 15s","accounts_in_message":[],"_revision_number":1},{"id":"896422366c992504669db4a679efb52c5517818c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-04 19:24:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"11d3aca8dd93b94720724a42d8329ba46dad991d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-04 19:29:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd38770dcd954336ac33078f93968b2e : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"47499c4a52359cf58d3a869e81b4ec5bb7dc0c26","revisions":{"47499c4a52359cf58d3a869e81b4ec5bb7dc0c26":{"kind":"REWORK","_number":1,"created":"2022-01-04 08:43:34.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/37/823337/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/37/823337/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/37/823337/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/37/823337/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/37/823337/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/37/823337/1"}}},"commit":{"parents":[{"commit":"8473e174791f2b3a8a693de0c6ffcdbacfc22431","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8473e174791f2b3a8a693de0c6ffcdbacfc22431"}]}],"author":{"name":"Eyal","email":"eyalb1@gmail.com","date":"2022-01-04 08:43:02.000000000","tz":120},"committer":{"name":"Eyal","email":"eyalb1@gmail.com","date":"2022-01-04 08:43:02.000000000","tz":120},"subject":"update constraint for aodhclient to new release 2.3.1","message":"update constraint for aodhclient to new release 2.3.1\n\nChange-Id: I51547efb797dae50a2a600be0c5724cf5784526d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/47499c4a52359cf58d3a869e81b4ec5bb7dc0c26"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/47499c4a52359cf58d3a869e81b4ec5bb7dc0c26"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
