)]}'
{"id":"openstack%2Frequirements~823449","triplet_id":"openstack%2Frequirements~master~I2fd61816159dba06ce5107fbe15cd89c1e0e4862","project":"openstack/requirements","branch":"master","topic":"things-that-break","attention_set":{},"removed_from_attention_set":{"14288":{"account":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"last_update":"2022-06-16 11:25:48.000000000","reason":"Change was abandoned"},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2022-06-16 11:25:48.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I2fd61816159dba06ce5107fbe15cd89c1e0e4862","subject":"DNM: things that break horizon","status":"ABANDONED","created":"2022-01-05 03:11:44.000000000","updated":"2022-06-16 11:25:48.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"98de176bee446291047918c9dfa94c6cd2998320","_number":823449,"virtual_id_number":823449,"owner":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-06-14 08:32:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-05 03:35:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-05 05:00:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c8bf46483c20afbe55c43182c3c387ef0df22f08","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-01-05 03:11:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ba14b9e60c155edf36f657a6925ee201ee67177f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-01-05 03:13:18.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"df1d83d4e451759a6534c044c09f6e1d70098366","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-05 03:35:59.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/75ea0b94d7e64125a38dcd01dab226a4 : FAILURE in 12m 19s","accounts_in_message":[],"_revision_number":1},{"id":"2c5cba878e4cd9e2a7db354b01061d3c20c0df8a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-05 05:00:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c3a0fa3354843eb98812e5755b451fe : SUCCESS in 6m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0a6732e07cb417a912024903c2e117b : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ced9a4f2aade4581a86ba562b9f4d2de : SUCCESS in 5m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/11ffa95e5e2e4224ab3d16ecdc665d38 : SUCCESS in 5m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fa0bb4a1925b4daaac1a5058273278c6 : SUCCESS in 4m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a47697f7a122459a873e3090fc41e231 : SUCCESS in 4m 24s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6e332e8e0f8c4cbd991d1f249490dd79 : SUCCESS in 12m 35s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/bc694d26fe5a4ffcbf6831dd11f6df6b : SUCCESS in 11m 44s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/ea8bff83bfe141f5bd25b1da410b3a93 : SUCCESS in 8m 41s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/4e7c70b6f2b043f9b731714e75981fcc : SUCCESS in 13m 46s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/afc2134bd300455da26d425007fc4a42 : SUCCESS in 7m 01s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/869ac2d13b4542bd9d9a2241011444b2 : SUCCESS in 12m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2dbb9460e60644868725bfa3ae057d81 : FAILURE in 9m 25s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/85ad870f7aae48699a1f2af03cda6da7 : SUCCESS in 13m 32s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/397d7312930c482d9db7650ee8858f59 : SUCCESS in 21m 18s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/f5bc0ecfb17f4d008316adcdd4f74a0d : SUCCESS in 5m 17s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/fad24c7a18ae44dc996de871e4cc9a20 : SUCCESS in 18m 48s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/86c4455e75e448dcbe3b8eba18f01db6 : SUCCESS in 8m 18s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/3c32a510591e4f9bb90c1b8b57df3ba8 : SUCCESS in 25m 18s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/27afb1fb4f6348aabfba15b106ae86bd : SUCCESS in 14m 03s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/45c837f376bb441b85bb1fc283522569 : SUCCESS in 6m 14s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/339bcccd16064031a7526da518c70af0 : SUCCESS in 5m 09s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/03f4638f0e1a4b00bcbafd75a63cd737 : SUCCESS in 21m 04s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/ba7d726ef99547c5a75323cfeef3b8fb : SUCCESS in 7m 02s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/56440101a5e5423ebf24df8a43116b25 : SUCCESS in 6m 02s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/40afd93e13164c0fad7779f3b43f44c8 : SUCCESS in 9m 23s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/b7b5041a6d354d408b272eb465f3774a : SUCCESS in 12m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5ac426eb266744c1aa3dd3bcf165c934 : SUCCESS in 51m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ec9d204fa6c04f9c98ba64c8b5efb100 : SUCCESS in 1h 35m 59s","accounts_in_message":[],"_revision_number":1},{"id":"293c90f6f850f0fc93c8e00cbac013c9046e49f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-06-10 06:40:57.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"bf6707da613da7152a8c861b87f6eb6c4e5694d0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-10 07:03:43.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/89689d134e7640caa342eb677a41ba7c : FAILURE in 16m 37s","accounts_in_message":[],"_revision_number":2},{"id":"171df161124035157a07a3cb57607d35d01fbf96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-10 08:08:14.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf56a47c2197412ea7d0be4ac0fd1ecc : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/199aee855dc9499faa8c912716e6e0ed : SUCCESS in 3m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4469804655404be6826ad367ceec8269 : SUCCESS in 4m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/10c77d5ba7884170b0264188c1acbc00 : SUCCESS in 4m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/870876036d5a47b8babef21b582970d8 : SUCCESS in 4m 47s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c1ef90b59844491fa7ce3ad99799c38d : SUCCESS in 4m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1dc8b223bd70445e8a7bbed398383965 : SUCCESS in 4m 36s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a7ba2323b1634cb6946df282fb361330 : SUCCESS in 9m 41s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/5321a7f7977a405a9fc387f06608c388 : SUCCESS in 14m 34s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/af9592884a434f50a740a9215e27357a : SUCCESS in 8m 38s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/5de7df55098244af872b0c83abd02987 : SUCCESS in 5m 17s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/f57fae9db0af42b5bd078e3786366c3d : SUCCESS in 6m 22s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/50af5bbe59f341be9eb2858b00b0dc7f : SUCCESS in 10m 13s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/6ef727c337ee4d32b8d4f8288a740f29 : SUCCESS in 5m 17s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/c94196593d3743a5b71dc2c896dfab6c : SUCCESS in 7m 19s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/cb4f211f034c49749f3c2063cfd16cda : SUCCESS in 7m 47s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/6e59a99eab914d20bb478bd06754533f : SUCCESS in 8m 12s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b03a2ea068cd416c81cba50c2c8ea283 : FAILURE in 7m 16s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/55af52bd5ce04f9792857301062b291e : SUCCESS in 8m 05s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/67675fd49d744cdb9f96a04a6f15291d : SUCCESS in 14m 01s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/dc2fca31a4084e4ab23ccd155cfe0ee0 : SUCCESS in 3m 48s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/384787e7b91740deabc08af765933331 : SUCCESS in 10m 49s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/3811bae49dd340249434dd4268dabf8d : SUCCESS in 5m 32s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/09cb11e0003c418b88f8b8b5168d9676 : SUCCESS in 25m 57s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/8b9e5d7a37d64987ba5dcea6224b4e15 : SUCCESS in 10m 46s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/f55ee84ebecc47d4aa7eae11cd116689 : SUCCESS in 6m 17s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/972a4c8b50024027b6fa0c813ca05821 : SUCCESS in 3m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9fd9ab7c95e64c42baed6a789130d5e3 : SUCCESS in 21m 49s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/06336ea8c1034a32a31e31c993657ca0 : SUCCESS in 5m 54s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/d6fff4d352f4499a859708d4f24c2119 : SUCCESS in 6m 13s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8aa16c999f03476bafa51040a61a25bf : SUCCESS in 7m 24s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/ec0f5e418c4a4fab9aa760e243a49b01 : SUCCESS in 11m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/99842ac1af954bb8a1b084595c0ced75 : SUCCESS in 40m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/75945b554b9a49b6818a76043495a064 : SUCCESS in 1h 24m 26s","accounts_in_message":[],"_revision_number":2},{"id":"ab99a35d5ecc4d542ba3dcb722397909dd03085f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-06-14 07:12:38.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"23921a960bbe14a8fbd9c28308bc0ef73c27872c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-14 07:24:50.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/a60dff32a797456e912b295a495e9f48 : FAILURE in 8m 32s","accounts_in_message":[],"_revision_number":3},{"id":"c797a343a93a1ecd4b6b93472914d0859db0d675","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-14 08:32:54.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f2c46bb7e6f4744a6ded94d36a33a05 : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/451d0e666cb841ccb5ee8e333cd3405d : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9e69e81389d6436ca60945bb617af828 : SUCCESS in 4m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0af76d8287f440a49ba77fecf7f98704 : SUCCESS in 4m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/afdb3a558feb49828af24815ae0b27d2 : SUCCESS in 4m 16s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b3093d18e72f479fb40581f2b832ca0d : SUCCESS in 4m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fd4fc36daa7b44e68683faf80f5627ba : SUCCESS in 4m 18s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c39add05ccaa4a939950c3705cbbd36b : SUCCESS in 9m 11s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/358fefcf946a4bbca38704347773a75d : SUCCESS in 12m 09s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/3624860bd57041a4837e3a26a3f8ec06 : SUCCESS in 8m 18s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/0102e9ac021644c6895860a516a0bc65 : SUCCESS in 5m 29s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/b11ef26acee04023b59f2ee8267230c6 : SUCCESS in 5m 52s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/1683ea690e804715bda330b39c9ca183 : SUCCESS in 21m 45s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/9cc267a4f9614d68808e2a6ca8cb7cd1 : SUCCESS in 5m 37s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/1c747ff54fd6482689c05cf4ebaf5019 : SUCCESS in 5m 28s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/f3fa5ace1845408f922c95528783762b : SUCCESS in 9m 03s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/37400dbb0d2349b38afae0c24901502a : SUCCESS in 10m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a774f9e612c245cc9888cdf28d835612 : SUCCESS in 13m 52s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/748ea34c428246248d96934d51e99a0e : SUCCESS in 10m 12s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/63e6406e40cf4560b059421ac97805dd : SUCCESS in 11m 54s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/574b86b838d4410d8aa9fcec20d6356e : SUCCESS in 6m 04s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/d13b3945edd74d649e522a9502f54af8 : SUCCESS in 8m 46s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/c3f5cefcf7eb4b9caa13e7324d661f8a : SUCCESS in 5m 18s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/9807851f03bb4a2ab242320a52a942eb : SUCCESS in 32m 56s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/a01ca9cf64864c1aa4b0528cfe3b8724 : SUCCESS in 21m 39s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/b3f043c1102044f0aafd5b86e2b00a81 : SUCCESS in 6m 02s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/fc753fdaf0da413ba1576a67b0fa3dde : SUCCESS in 4m 05s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b1e31b91ecf64758967500b6fc0d4ce3 : SUCCESS in 21m 12s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/386345c219514d2fb3d28b28faafb13b : SUCCESS in 5m 59s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/85e1915f67f7447792b1df283b26181f : SUCCESS in 6m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4e96206cd98742c0aa76a4bc95e85515 : SUCCESS in 6m 22s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/f4a9f8135e804ace96fb429d26aee940 : SUCCESS in 11m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fc8f32a8c40d44edab5afb154d2fbdc0 : SUCCESS in 44m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e9527437b9ef4579ba97110b82560f56 : SUCCESS in 1h 18m 22s","accounts_in_message":[],"_revision_number":3},{"id":"98de176bee446291047918c9dfa94c6cd2998320","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-06-16 11:25:48.000000000","message":"Abandoned\n\nMerged as part of https://review.opendev.org/c/openstack/requirements/+/845871","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"34238c0008330bb75577808d30f140bf313fe48f","revisions":{"f7aa8fcb293cbae877e9251213c1ef91be8e135e":{"kind":"REWORK","_number":1,"created":"2022-01-05 03:11:44.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/49/823449/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/49/823449/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/49/823449/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/49/823449/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/49/823449/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/49/823449/1"}}},"commit":{"parents":[{"commit":"47499c4a52359cf58d3a869e81b4ec5bb7dc0c26","subject":"update constraint for aodhclient to new release 2.3.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/47499c4a52359cf58d3a869e81b4ec5bb7dc0c26"}]}],"author":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2022-01-05 03:11:27.000000000","tz":-360},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2022-01-05 03:11:27.000000000","tz":-360},"subject":"DNM: things that break horizon","message":"DNM: things that break horizon\n\nChange-Id: I2fd61816159dba06ce5107fbe15cd89c1e0e4862\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f7aa8fcb293cbae877e9251213c1ef91be8e135e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f7aa8fcb293cbae877e9251213c1ef91be8e135e"}]},"branch":"refs/heads/master"},"e909487db4b1a2690eccf831378d85eb52ee7d92":{"kind":"REWORK","_number":2,"created":"2022-06-10 06:40:57.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/49/823449/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/49/823449/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/49/823449/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/49/823449/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/49/823449/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/49/823449/2"}}},"commit":{"parents":[{"commit":"e9011ef27ade290123f7ddf735395d2b0fa72efc","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e9011ef27ade290123f7ddf735395d2b0fa72efc"}]}],"author":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2022-01-05 03:11:27.000000000","tz":-360},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2022-06-10 06:40:37.000000000","tz":120},"subject":"DNM: things that break horizon","message":"DNM: things that break horizon\n\nDepends-On: https://review.opendev.org/c/openstack/horizon/+/843346\nChange-Id: I2fd61816159dba06ce5107fbe15cd89c1e0e4862\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e909487db4b1a2690eccf831378d85eb52ee7d92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e909487db4b1a2690eccf831378d85eb52ee7d92"}]},"branch":"refs/heads/master"},"34238c0008330bb75577808d30f140bf313fe48f":{"kind":"REWORK","_number":3,"created":"2022-06-14 07:12:38.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/49/823449/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/49/823449/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/49/823449/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/49/823449/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/49/823449/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/49/823449/3"}}},"commit":{"parents":[{"commit":"e9011ef27ade290123f7ddf735395d2b0fa72efc","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e9011ef27ade290123f7ddf735395d2b0fa72efc"}]}],"author":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2022-01-05 03:11:27.000000000","tz":-360},"committer":{"name":"Dr. Jens Harbott","email":"frickler@offenerstapel.de","date":"2022-06-14 07:12:38.000000000","tz":0},"subject":"DNM: things that break horizon","message":"DNM: things that break horizon\n\nDepends-On: https://review.opendev.org/c/openstack/horizon/+/843346\nChange-Id: I2fd61816159dba06ce5107fbe15cd89c1e0e4862\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/34238c0008330bb75577808d30f140bf313fe48f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/34238c0008330bb75577808d30f140bf313fe48f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
