)]}'
{"id":"openstack%2Frequirements~823679","triplet_id":"openstack%2Frequirements~stable%2Fwallaby~Id32adfcf27cd48a179dbacced47b2990fcf628e6","project":"openstack/requirements","branch":"stable/wallaby","topic":"VF-versionbump","attention_set":{},"removed_from_attention_set":{"11491":{"account":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"last_update":"2022-01-06 22:12:26.000000000","reason":"Change was submitted"},"16515":{"account":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"last_update":"2022-01-06 22:12:26.000000000","reason":"Change was submitted"},"27427":{"account":{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},"last_update":"2022-01-06 22:12:26.000000000","reason":"Change was submitted"},"15895":{"account":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"last_update":"2022-01-06 22:12:26.000000000","reason":"Change was submitted"},"32926":{"account":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"last_update":"2022-01-06 22:12:26.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id32adfcf27cd48a179dbacced47b2990fcf628e6","subject":"Revert \"Downgrade validation lib to 1.1.0\"","status":"MERGED","created":"2022-01-06 12:24:12.000000000","updated":"2022-01-06 22:12:26.000000000","submitted":"2022-01-06 22:12:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"823679-VF-versionbump","meta_rev_id":"3454bae2d4bea29d8f53244a900e2852f6428725","_number":823679,"virtual_id_number":823679,"owner":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-06 22:12:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-01-06 16:58:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-01-06 16:58:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"}],"reviewers":{"REVIEWER":[{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-06 12:38:42.000000000","updated_by":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"reviewer":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"state":"REVIEWER"},{"updated":"2022-01-06 12:38:42.000000000","updated_by":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"reviewer":{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},"state":"REVIEWER"},{"updated":"2022-01-06 12:38:42.000000000","updated_by":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"reviewer":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"state":"REVIEWER"},{"updated":"2022-01-06 12:38:42.000000000","updated_by":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"reviewer":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"state":"REVIEWER"},{"updated":"2022-01-06 12:38:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-06 13:58:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-06 16:58:56.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"1e4527431fffea40153028adbcabb56e1f533998","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2022-01-06 12:24:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"509e04c8ac14f39644b89496401bf975bbbf0df9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-06 12:38:56.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/974ec7a2c57b44aea18dae9bfadc8c26 : FAILURE in 10m 27s","accounts_in_message":[],"_revision_number":1},{"id":"82915479ad2f1016006e1d6227cc7752782c4d94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-06 13:58:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7fa318ab428b40f99c01636f97c0a60e : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b26f54b65a7b403eba04e4a771147c1a : SUCCESS in 5m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5dc954074321410baf3904bb1ae4fbbe : SUCCESS in 5m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/034eb06f9d2840ffbe68148c6bd24395 : SUCCESS in 6m 04s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4d4a42b3b2bc40c68476ad59c81a2a79 : SUCCESS in 4m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fad1e71e8b294584b178968770e5ffa3 : SUCCESS in 5m 07s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b586971320124ea4ac061187ac1476ad : SUCCESS in 11m 09s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1aa5805dd72b4bf7aec64cae99c3e4cb : SUCCESS in 10m 06s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8b65b091ce82478d969a9e1249f736d1 : SUCCESS in 11m 00s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f8a4ada727ac445bb280bf7f0558313f : SUCCESS in 6m 12s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/8fb128d54f19472babf557278f5ff367 : SUCCESS in 11m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/756574ec0a4c41a3b6bff3f4f94ba98a : SUCCESS in 9m 14s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/e0a01c79b9234b6b813509e1908b2bec : SUCCESS in 9m 18s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/4a3387946e35411e81dc33ed75feac38 : SUCCESS in 18m 46s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/1805baea582a46d88c0fe6dbbf7af5cf : SUCCESS in 4m 56s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/3dd26b87b4e7427a8aaf148817d00469 : SUCCESS in 29m 52s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/eebbef2eb5d7434d88a299999afa42ab : SUCCESS in 11m 44s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2d359800c04741c5becbdd080023ff6d : SUCCESS in 3m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c76dd13350b54d359283953d9857d405 : SUCCESS in 17m 22s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/e57fb056c2e444ae99cb096a2c031f18 : SUCCESS in 6m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/82c76176897b4c02ad773afac7245183 : SUCCESS in 7m 31s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/367bdeb8fbcf4e56b8894c5a7010444d : SUCCESS in 13m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/704a3062120d489abddcebbe72fdfa06 : SUCCESS in 55m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/eb4a96ec80e740f187462bb06054e8ad : SUCCESS in 1h 27m 48s","accounts_in_message":[],"_revision_number":1},{"id":"7e6ceca39d4fd18f76594f36c99263b30e01f58c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-01-06 16:58:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"52490a9c7f49a9cb19fa494d3f7fc3ed55417ca5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-06 17:00:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ea30bed859a76fa5dd0856a3a2466de4ce1a72f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-06 22:12:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05bbaca5c3e74e26ba5d6c9624b61700 : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a9328f728c1e442896dbd627ae20c995 : SUCCESS in 6m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f55f3de0a03d48bd99523db8bd2b790d : SUCCESS in 6m 23s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/08457ceaf5cb4254882eabd4a24a8fac : SUCCESS in 7m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e79055b46b9c4dc4a42a8883bab781da : SUCCESS in 6m 06s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9632eb3fecf6487e890fe63841a09c6b : SUCCESS in 11m 26s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3cc82f44dbd2441db535be24df65f20d : SUCCESS in 10m 33s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/dd2c0468f5d24b48a620e9be995d6f61 : SUCCESS in 13m 02s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/c1779dee3ab64ce497855d94c19c5a2e : SUCCESS in 8m 34s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/707e3ac1d715463086aeb6b8c955c73b : SUCCESS in 12m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/07be004436f4462293a242e5a8133789 : SUCCESS in 9m 20s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/637836c9cf1b4f5e9ed69cc6286d3f3a : SUCCESS in 9m 44s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/3797033d84c34d879abbca7443d0cace : SUCCESS in 14m 37s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/207c078e849948878f1cdb3e456af88a : SUCCESS in 7m 50s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/5374130ebb544a678c0d698c5864e8ab : SUCCESS in 31m 25s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b4f2f0ee68a249cbbde07cf4d6157e18 : SUCCESS in 16m 08s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/57af082383f847ceb6afb7a9a4a2c5a3 : SUCCESS in 7m 47s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d73cc30cf71742a1ac4fc83558d870eb : SUCCESS in 20m 51s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/de5d03cf09c34015aaa316f20184c4b3 : SUCCESS in 7m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/32e1a0ae43934b6da66346f468a19f47 : SUCCESS in 9m 18s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/1196846a442f4a57b5ad7612938bcf0c : SUCCESS in 12m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/819d95a2ebad4bcbb1679882900e5604 : SUCCESS in 57m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5524a69007fe4d6b81c1df1208ba8889 : SUCCESS in 1h 44m 06s","accounts_in_message":[],"_revision_number":1},{"id":"3454bae2d4bea29d8f53244a900e2852f6428725","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-06 22:12:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a0fc6afb6a1b88ddd7f89375435aeefd481cdc9c","revisions":{"a0fc6afb6a1b88ddd7f89375435aeefd481cdc9c":{"kind":"REWORK","_number":1,"created":"2022-01-06 12:24:12.000000000","uploader":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"ref":"refs/changes/79/823679/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/79/823679/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/79/823679/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/79/823679/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/79/823679/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/79/823679/1"}}},"commit":{"parents":[{"commit":"6a95003769be77dcb2d636d58684612e1824fd44","subject":"Merge \"update constraint for tripleo-common to new release 15.3.0\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6a95003769be77dcb2d636d58684612e1824fd44"}]}],"author":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2021-08-06 14:23:56.000000000","tz":120},"committer":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2022-01-06 12:23:18.000000000","tz":60},"subject":"Revert \"Downgrade validation lib to 1.1.0\"","message":"Revert \"Downgrade validation lib to 1.1.0\"\n\nThe fix was merged some time ago and appears to be stable.\nFurthermore, versions 1.2.0 and higher introduce number of\nimportant fixes and changes that are now being left out.\n\nThis reverts commit 960c75301338423c2b78c06742dca09ed30b07bd.\n\nSigned-off-by: Jiri Podivin \u003cjpodivin@redhat.com\u003e\nChange-Id: Id32adfcf27cd48a179dbacced47b2990fcf628e6\n(cherry picked from commit 552c96db9676b04f64d92f5acc4cd7384aed90cd)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a0fc6afb6a1b88ddd7f89375435aeefd481cdc9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a0fc6afb6a1b88ddd7f89375435aeefd481cdc9c"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
