)]}'
{"id":"openstack%2Frequirements~823680","triplet_id":"openstack%2Frequirements~stable%2Fvictoria~I7273dd978f4a259d721762b7d770deeedadacf87","project":"openstack/requirements","branch":"stable/victoria","topic":"VF-versionbump","attention_set":{},"removed_from_attention_set":{"11491":{"account":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"last_update":"2022-01-06 22:12:29.000000000","reason":"Change was submitted"},"27427":{"account":{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},"last_update":"2022-01-06 22:12:29.000000000","reason":"Change was submitted"},"16515":{"account":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"last_update":"2022-01-06 22:12:29.000000000","reason":"Change was submitted"},"15895":{"account":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"last_update":"2022-01-06 22:12:29.000000000","reason":"Change was submitted"},"32926":{"account":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"last_update":"2022-01-06 22:12:29.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7273dd978f4a259d721762b7d770deeedadacf87","subject":"[Victoria-only] Bumping upper constrain of openstack/validations-libs to 1.2.0","status":"MERGED","created":"2022-01-06 12:38:23.000000000","updated":"2022-01-06 22:12:29.000000000","submitted":"2022-01-06 22:12:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"823680-VF-versionbump","meta_rev_id":"2a8a8c0d00b04839a257c5ab5d69078335cd3311","_number":823680,"virtual_id_number":823680,"owner":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-06 22:12:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-01-06 16:58:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-01-06 16:58:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"}],"reviewers":{"REVIEWER":[{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-06 12:38:52.000000000","updated_by":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"reviewer":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"state":"REVIEWER"},{"updated":"2022-01-06 12:38:52.000000000","updated_by":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"reviewer":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"state":"REVIEWER"},{"updated":"2022-01-06 12:38:52.000000000","updated_by":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"reviewer":{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},"state":"REVIEWER"},{"updated":"2022-01-06 12:38:52.000000000","updated_by":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"reviewer":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"state":"REVIEWER"},{"updated":"2022-01-06 12:58:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-06 14:41:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-06 16:58:57.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"3e5bd6f7c163a35d5237efab0d372a1e6c6db2a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2022-01-06 12:38:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"900114aa1e9406c2326154276e69189322c0908a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-06 12:58:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/131b62f0877344cfb0567f80aa90a6ce : SUCCESS in 17m 36s","accounts_in_message":[],"_revision_number":1},{"id":"7d5f170b91f13aecc31e7ef15e52aabc440f2883","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-06 14:41:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a81607fc1d6e44708da0862d8e575ddb : SUCCESS in 6m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f1830df46d594b21a0294b2e71a73c2d : SUCCESS in 4m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/21d25048923e4c139a6ba4019c80be36 : SUCCESS in 4m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/eb4e29dae9594633b1f63775b96077a3 : SUCCESS in 5m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/210cdd980a234071979d9a9a91d4c826 : SUCCESS in 5m 16s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/99a0d3730b774a099b057692f33077b7 : SUCCESS in 11m 10s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/05d62a3c7b494d288bad96b07853b0e9 : SUCCESS in 11m 13s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/52747123a5d443a4a1c3d44c751464fa : SUCCESS in 10m 32s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e20f5f397fc84580ad89e96b28f0fc2e : SUCCESS in 6m 01s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ec20c5316b5641cb8b82b72900afc64b : SUCCESS in 9m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6cf23b2ab3be499c999b941f5b2b15af : SUCCESS in 9m 43s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/dd48da1888474fa1a60fc2ebab39186b : SUCCESS in 8m 46s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/30e540e7917647efb5b26fd11a65e7e2 : SUCCESS in 15m 55s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/b27f2c00034543a9acedc42420733c1f : SUCCESS in 21m 13s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/e1e218ac13a94b518127cb80febfa095 : SUCCESS in 15m 53s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ae99e3644f6c43209c1396dc219514a3 : SUCCESS in 6m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/911c6827f55b42e7bc445e98cfce3473 : SUCCESS in 16m 13s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/fd03d0cb039b46a680450197a979527e : SUCCESS in 7m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/be464a0c88624256947e026e91513acb : SUCCESS in 6m 29s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e81155c657fb4fd792d8d23eed1de21e : SUCCESS in 13m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1af38b305e7c4d80b586aafb1a1469a9 : SUCCESS in 52m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/30094fa56e20498a9bc86ed1248bf6e1 : SUCCESS in 1h 56m 44s","accounts_in_message":[],"_revision_number":1},{"id":"08a0071fa92686710b44a32e34a909fca8bc8b5f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-01-06 16:58:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ecf54f1080b7cf9332b550c1e263963b87a2b625","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-06 17:01:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5a28755d989587948e3b9b2ed030b1179289a072","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-06 22:12:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/727fcb25e10d424cbdc094cd620fb09f : SUCCESS in 6m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3702160d90284ec4b8611476e7d4a9d7 : SUCCESS in 8m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f7e252db28ff46c3862a5aa618ce46e6 : SUCCESS in 7m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1769082a02f44a97925aad02f939714e : SUCCESS in 6m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c07eb8f6de0245d68152ae2e65c2e082 : SUCCESS in 5m 56s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/17138ee8bc2d4b83be3fa0b95798743a : SUCCESS in 11m 10s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/74ebcc5b53da48b9bd62045aa9b7f71e : SUCCESS in 10m 47s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/810df180017d464a976c7e42bd3fa1be : SUCCESS in 9m 50s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/80a02368f4284f70942055c380d06595 : SUCCESS in 7m 58s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/5c31cd123c784ff9874da47f43e14837 : SUCCESS in 11m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/56f415835234478baa1f99ad819a9f96 : SUCCESS in 9m 59s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/774dcdce2b15421b80c6d6e04b4efbad : SUCCESS in 9m 35s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/6e468041b5ff441aa77bd3d6108bdb1d : SUCCESS in 17m 48s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/395c119f6ce844f1bcee783d2acda1fa : SUCCESS in 27m 56s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/c9536891d03d479cae61c2bb512c1ad9 : SUCCESS in 17m 29s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ed08794972534bf8aca6338e384b4e17 : SUCCESS in 6m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/40be79c11314461290ad48721cf81b56 : SUCCESS in 18m 12s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/6ec637cb34354c58a3938769702857bd : SUCCESS in 7m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bf5e784ca2994ceda5871ffd73cf8654 : SUCCESS in 10m 08s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e1a2e49f2e784593bdef86facab4b4dc : SUCCESS in 13m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/907e5de3b88744baab73f41c1f8ad86d : SUCCESS in 53m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c32a2d7fde914db89dbcb549396f4fee : SUCCESS in 1h 22m 18s","accounts_in_message":[],"_revision_number":1},{"id":"2a8a8c0d00b04839a257c5ab5d69078335cd3311","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-06 22:12:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0578a9d88db12e254dc7cfc393d1414158336f1b","revisions":{"0578a9d88db12e254dc7cfc393d1414158336f1b":{"kind":"REWORK","_number":1,"created":"2022-01-06 12:38:23.000000000","uploader":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"ref":"refs/changes/80/823680/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/80/823680/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/823680/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/823680/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/823680/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/80/823680/1"}}},"commit":{"parents":[{"commit":"533601b020d5ef571d85f51af84362625c39a100","subject":"update constraint for oslo.context to new release 3.1.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/533601b020d5ef571d85f51af84362625c39a100"}]}],"author":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2022-01-06 12:35:14.000000000","tz":60},"committer":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2022-01-06 12:35:54.000000000","tz":60},"subject":"[Victoria-only] Bumping upper constrain of openstack/validations-libs to 1.2.0","message":"[Victoria-only] Bumping upper constrain of openstack/validations-libs to 1.2.0\n\nAs a number of newer features is being backported[0]\nto earlier releases the constraints have to be updated accordingly to\nensure that CI is capable of installing the necessary packages.\n\n[0] https://review.opendev.org/c/openstack/python-tripleoclient/+/823604/3\n\nSigned-off-by: Jiri Podivin \u003cjpodivin@redhat.com\u003e\nChange-Id: I7273dd978f4a259d721762b7d770deeedadacf87\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0578a9d88db12e254dc7cfc393d1414158336f1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0578a9d88db12e254dc7cfc393d1414158336f1b"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
