)]}'
{"id":"openstack%2Frequirements~823808","triplet_id":"openstack%2Frequirements~master~I5e6ae53e993f8aa314514b9ca7a4cb89fec379bf","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I5e6ae53e993f8aa314514b9ca7a4cb89fec379bf","subject":"update constraint for python-keystoneclient to new release 4.4.0","status":"MERGED","created":"2022-01-07 14:43:18.000000000","updated":"2022-01-07 19:28:30.000000000","submitted":"2022-01-07 19:23:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"823808-new-release","meta_rev_id":"3f6300721362651144060bea3a1dadd634f651e0","_number":823808,"virtual_id_number":823808,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-07 19:23:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-01-07 16:30:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-01-07 15:46:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-01-07 16:30:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-07 15:01:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-07 15:46:26.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-01-07 16:21:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-07 16:30:17.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"7b3f87ae652da0cb466a58a4afcd2d4bae5e7d79","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-01-07 14:43:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e9e1872c229f3e840fd763bf57cffdd630a10d04","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 15:01:02.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/cc34334cd703430faf107519facf3b21 : FAILURE in 15m 53s","accounts_in_message":[],"_revision_number":1},{"id":"443add0c59dcf29fd3ce3aaa247fa4ea079aec46","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-01-07 15:46:26.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"39254508cf618195b3ce7a0d294dc0f84a541bb4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 16:21:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c0e4442dbee4892b197f5c2ef1d268d : SUCCESS in 5m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dff29b5cff8140eca6f1421674638191 : SUCCESS in 3m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1cc7890a5dd8460daaa01ea769726ccc : SUCCESS in 4m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d05e96d8124e4b7cba59646c357d7fa2 : SUCCESS in 4m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/99def9e1af2d4b3cbbb8ddeb1d10097e : SUCCESS in 5m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/92b584daa04e4f74b1b5c206b6c22552 : SUCCESS in 4m 13s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/01bc0d026bf44e449d5cdf55abc0f10a : SUCCESS in 15m 38s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3e7602c2ee6f4a0f83aeec55d7e7ec1d : SUCCESS in 10m 59s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/992de88e952b46cdb70f5ef838674bbd : SUCCESS in 6m 09s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/8206cce187f44a409285c3529af37289 : SUCCESS in 13m 14s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/117f30e6c86f4895bc1e21087ff1e234 : SUCCESS in 5m 48s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/22bff386af00412586048f980a1b25b5 : SUCCESS in 11m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/689da59d807a4fe99151bf7c84400a81 : SUCCESS in 8m 45s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/316bd6f63d3c4335b555589664c83fa3 : SUCCESS in 8m 50s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/4403f0b7f37c4529bec18042514e8df6 : SUCCESS in 13m 41s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/e7e035464da34f898cf6c7d7d709fe5a : SUCCESS in 4m 47s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/c2283d5021484993980ad69c9dcff846 : SUCCESS in 17m 33s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/be6d8ba1b15f406db8da81c139d37ade : SUCCESS in 5m 48s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/a04dddd7d47d453d98de1c306d359fdb : SUCCESS in 25m 07s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/d7c4d99403d246aeb985fd485d0a6892 : SUCCESS in 11m 58s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/e1bcae7ace8c4397bc72a37d084d2159 : SUCCESS in 5m 39s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4fedb3150c50434c869918c150120525 : SUCCESS in 3m 47s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/27cfd23d72474eee9310124bbdea2e83 : SUCCESS in 19m 49s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/a26e3c7ca97b44adb68c8bd4f1cba139 : SUCCESS in 6m 30s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/f1df1c71f0134f6e88d951a825f7ed43 : SUCCESS in 5m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e69d089d53d64bf3b5c3c9581135edae : SUCCESS in 7m 28s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/9424f6cb5bb4404c82e76c815caa2ec4 : SUCCESS in 12m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9f825026fca44e3ca85c99d7d0071114 : SUCCESS in 45m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8ab40b1f90674cdcab4b0c89990cdf4e : SUCCESS in 1h 31m 48s","accounts_in_message":[],"_revision_number":1},{"id":"cdc2015a55ab3b436eb2442168d923562e044990","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-01-07 16:30:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"685b91dbe4c2123dd40fe5360ecadbac472d18a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 16:31:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ba8754d385fb315d24d682d0ee1384dddc28ba5c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 19:23:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/361529f4bb1c4451931b8b4af6457ee9 : SUCCESS in 8m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab172a3602694534b6fb4e3bd22fdd93 : SUCCESS in 6m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c16884e7aab1478fbb92646974803d8d : SUCCESS in 7m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4dde7eeb00674a0bb5d0683e821deb75 : SUCCESS in 6m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9daf5c6b098d43828c1456a8bbded528 : SUCCESS in 6m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2749d35925d34775b05b6ab6ce6fb33c : SUCCESS in 5m 28s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c611a17112a9438fa842f67396556500 : SUCCESS in 14m 33s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/55df10146c0f46a0b453a1edd142b2e4 : SUCCESS in 12m 09s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/6a0b019c416d4e4280a7cbd4a45d37b0 : SUCCESS in 12m 32s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/726adce3fa104daf8b7535221d9a6109 : SUCCESS in 7m 06s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/1e2292f131994d4eb059a3265502824a : SUCCESS in 11m 53s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f34777e65b39446d86dbad8e0398e6fa : SUCCESS in 11m 35s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/161af6ccbb4343c7ba1c650049eae9ee : SUCCESS in 11m 03s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/16728afc906b43a3a43ab453f45df493 : SUCCESS in 15m 12s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/a98214b8b9bf4b3bbdfcf9fbf44d7c91 : SUCCESS in 6m 32s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/839aeb594b8b426fa36657e6ccef8a9c : SUCCESS in 17m 28s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/b1f55338cfa648bebd46119d2ec8440e : SUCCESS in 6m 18s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/5f76d538e5da40b095db0cc06e9b33d8 : SUCCESS in 25m 10s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/2476a48529ef4430821ee6cb17373374 : SUCCESS in 16m 44s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/79481ad7511548ecb7eb1c87ebcaff74 : SUCCESS in 7m 44s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d319445d06d741edb15836afccf80f94 : SUCCESS in 4m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/033537820ad14c96839c65d51831aca3 : SUCCESS in 19m 35s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/2642df0e3ea644d0a6f75d0501655885 : SUCCESS in 7m 31s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/614828452a38406c919bf48856110e3f : SUCCESS in 6m 43s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5bca6c7912924097a61b7cc3bc32dbf3 : SUCCESS in 9m 06s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/9fcc9004d35045dd98c44e7ce511d0a7 : SUCCESS in 12m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c2eff2b85af64d34836a3bcd41712d99 : SUCCESS in 59m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/389df6cb863645c083023ef1a7a55a60 : SUCCESS in 1h 28m 07s","accounts_in_message":[],"_revision_number":1},{"id":"7a99a73b260e43acf5fa0bfb2e3ef116ffaa6d0f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 19:23:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3f6300721362651144060bea3a1dadd634f651e0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 19:28:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ffe0a23409a404b9520512b9acea693 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bc2e04f5040f52586f78098d03cd4dfd15986cdb","revisions":{"bc2e04f5040f52586f78098d03cd4dfd15986cdb":{"kind":"REWORK","_number":1,"created":"2022-01-07 14:43:18.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/08/823808/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/08/823808/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/08/823808/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/08/823808/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/08/823808/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/08/823808/1"}}},"commit":{"parents":[{"commit":"ceb2d650b53465c7ef6274f4ff20305a928e9639","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ceb2d650b53465c7ef6274f4ff20305a928e9639"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-01-07 14:43:17.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-01-07 14:43:17.000000000","tz":0},"subject":"update constraint for python-keystoneclient to new release 4.4.0","message":"update constraint for python-keystoneclient to new release 4.4.0\n\nmeta: version: 4.4.0\nmeta: diff-start: -\nmeta: series: yoga\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: Ic5c870f567a724f23d806e210dd557c66ba5d12e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I5e6ae53e993f8aa314514b9ca7a4cb89fec379bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bc2e04f5040f52586f78098d03cd4dfd15986cdb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bc2e04f5040f52586f78098d03cd4dfd15986cdb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
