)]}'
{"id":"openstack%2Frequirements~823809","triplet_id":"openstack%2Frequirements~master~I29356f6087d4635529270cd2d663d9310957bbbe","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I29356f6087d4635529270cd2d663d9310957bbbe","subject":"update constraint for python-tackerclient to new release 1.9.0","status":"MERGED","created":"2022-01-07 14:46:20.000000000","updated":"2022-01-07 23:49:40.000000000","submitted":"2022-01-07 23:48:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"823809-new-release","meta_rev_id":"36c40246b78d1db79fc4d016969a96904bdfc555","_number":823809,"virtual_id_number":823809,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-07 23:48:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-01-07 19:49:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-01-07 19:49:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-07 15:19:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-07 15:46:45.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-01-07 16:43:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-07 16:52:26.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"fa0c38c2e52663c1816a6b7f23419f12bb970db7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-01-07 14:46:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eaf22e8d243d352b161ec6b84bb9dbade7c22145","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 15:19:22.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/223fe9f467c344b48c401ab780bb1378 : FAILURE in 16m 33s","accounts_in_message":[],"_revision_number":1},{"id":"cfae73054a79fa8943d40d8ea9efb7f13328f352","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-01-07 15:46:45.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"06cfdfcd97626ae73ad950c0452a32e1c83ea927","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 16:43:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d72fae37d584adebd8ba9b578526d43 : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/948bdd7feeaf4630808564e3a465067b : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b2666f2e2da8472f837a2c2dc289d1f2 : SUCCESS in 4m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c181adedd12146a781a1631d2675ac68 : SUCCESS in 5m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/32e82b4cff7e4322aa7a0cbf35693711 : SUCCESS in 4m 35s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e12b0e4828cc4a2194a92041c028c373 : SUCCESS in 5m 03s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/28e7b28303ef4df79880c6d616cbe336 : SUCCESS in 10m 16s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/50e0d02dab2f41bea883230a19644183 : SUCCESS in 9m 50s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/48170543b8ea4819aaa12fc853cc9eeb : SUCCESS in 6m 39s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/8ef920c456a94fe9bd6db1996aabdc1d : SUCCESS in 12m 08s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/7ba045e258ba4a74a65fa09d2b2c5198 : SUCCESS in 6m 09s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/213c71eaf3b94d7d82cd1ad55f8aee0a : SUCCESS in 10m 05s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4c80aefe8dba4b739f0fb63b7f7ff139 : SUCCESS in 9m 48s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/49e50656cff746cf81969e9d47009d64 : SUCCESS in 11m 09s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/e02d7f44ebe242339edebe580ee0253a : SUCCESS in 19m 28s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/b211064ba388451bb695436cfaa7decd : SUCCESS in 4m 44s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/b96924ae2b8f4bf39569ff08585533b2 : SUCCESS in 16m 04s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/ba2dcb70b1f04484b896e24d617f8d51 : SUCCESS in 6m 17s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/73e81dc454684df0ae934ecffae3e64a : SUCCESS in 24m 23s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/3eb725768c264405833a499ed2e187f9 : SUCCESS in 13m 51s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/cd86bf0dc1074196bf7b734b9be56dcd : SUCCESS in 6m 12s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/43ca453cbdfc41ccb558df16b4d8591f : SUCCESS in 4m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a04c8794d5fa4ed7943464792dbf8589 : SUCCESS in 24m 01s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/3d5ba3887d924021a5ec83c736292c1f : SUCCESS in 8m 26s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/cb5db1a1e3f84c168a39879c6e7dba43 : SUCCESS in 5m 54s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/eaf8b6b7a9e94931a4a04744d75c9505 : SUCCESS in 7m 25s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/309e83cc4bc34f1f87bba16143062577 : SUCCESS in 15m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6d9eb429889e444788099508370b5bf9 : SUCCESS in 49m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7055880d7cac464bb81a9d5d9b61b9ff : SUCCESS in 1h 47m 06s","accounts_in_message":[],"_revision_number":1},{"id":"3345b5672769ee65f11079e3f430cfc81f47a06f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-01-07 16:52:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5c5875e2bd2cf4164ade950a339bfc657281b23d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 19:23:41.000000000","message":"Patch Set 1: Verified-2\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"b43342efb06ffee741a7bff4c60a126e473756c1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-01-07 19:49:25.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1ecadbe0b4b8551cab03d62f9472866c717e3a90","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-01-07 19:49:35.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d10bb2286dbec9a2c79dc15675be98d57df343af","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 20:21:11.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/86095c59405d403c907ddb990b0fd3e3 : FAILURE in 14m 41s","accounts_in_message":[],"_revision_number":2},{"id":"5a3fda195de2bd8081e2def0e3f92cbe3c960e2b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 21:26:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db95ad37b709419b85cdcd651718a8a0 : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3e51fcf11f0480691ec67c398b9d0cd : SUCCESS in 3m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d50d773de1304ad995b5d345407e06e2 : SUCCESS in 5m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/060fe23bceb14fe991c6368e433c804b : SUCCESS in 4m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a7e4814b885346d28f389875008c1e40 : SUCCESS in 3m 53s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a43af00a7a344747b43dbe5287ce6117 : SUCCESS in 3m 38s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b1b10ed9398b4a0db879cb482c587e34 : SUCCESS in 11m 10s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1786b84db6164a2594f88bce5dc3001b : SUCCESS in 12m 11s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/81ac3a0be5804c149b7b7d8cc4494e32 : SUCCESS in 5m 18s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/803d38399fce4bcf8e41ac083ee6d398 : SUCCESS in 10m 45s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/eebf939dd10948719ca4e435825854e2 : SUCCESS in 5m 36s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/ae1564f5594a452f8868d7074b860d24 : SUCCESS in 11m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cc8163b10bb5440aae714f95d4cd7ff4 : SUCCESS in 7m 39s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/153466c728c4442db6b365935464a3ae : SUCCESS in 8m 39s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/25255f539f6e4da391ac7aa487d771fd : SUCCESS in 14m 11s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/578dfe38bf1c4e47b0942a5c02000eaf : SUCCESS in 4m 07s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/178a586ac1d64d3491082c5af6b94e13 : SUCCESS in 15m 39s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/a8655692726743629af0ec2c9d486367 : SUCCESS in 5m 09s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/1f0576667a21451ab5adc1cd0518fac0 : SUCCESS in 24m 11s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/69e1404b141c4c158632c669e58adcde : SUCCESS in 12m 11s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/9344f68f6bb341aea63cbb497a7c0ca9 : SUCCESS in 5m 43s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/71f4a6a40e4048b88fc7852779774ee2 : SUCCESS in 3m 35s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a81458aa59484d559218c3b0eca662ba : SUCCESS in 21m 00s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/db3ba488c80a4168935d27a7730df2ed : SUCCESS in 6m 16s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/96aefd6a42494375b523b3c0dc56ac61 : SUCCESS in 5m 23s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2e54c0d989ee4fb486245dc48e66b3bc : SUCCESS in 6m 42s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/aa80f15bddee4a1bb4a8592a62d64348 : SUCCESS in 11m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ef743ffd51dc435e8fef6e1ac4d835b2 : SUCCESS in 51m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a8313541cbf342d9b9609a9f0efff9f8 : SUCCESS in 1h 31m 24s","accounts_in_message":[],"_revision_number":2},{"id":"b53fc25cb69e50d30a5cb2575d160359d5b8689e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 21:27:49.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c52683c0dc754d33e87841fff2f8d8b587021da0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 23:48:07.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7641208fcdd48408227dcbefb207ad5 : SUCCESS in 5m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/636f7aeb8b9c498eab11641e973463c4 : SUCCESS in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f4c10a9c531d4f79b49f5afe47d2de7d : SUCCESS in 5m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a5aae61b284448779590c9e95ebd9047 : SUCCESS in 4m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/54b349e6120a47e182847a5ab67853b3 : SUCCESS in 6m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2ed22d427fce46509bf77eb13a05a9bf : SUCCESS in 4m 04s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c163183d4940449ba72f7c70d737cb73 : SUCCESS in 10m 34s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ed13f1a32f9d43aaae16b1fd4c26debd : SUCCESS in 10m 38s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/66a3da56f8b64c748fe50ee48e718123 : SUCCESS in 13m 25s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/8dc39d89aff44225a4d613621aa1271e : SUCCESS in 6m 30s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/f15af76aef4546f996052c4925116b8b : SUCCESS in 14m 07s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/41238980c0054412abd12b340ee1cf15 : SUCCESS in 9m 27s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f5eda507daea445286189d5e0085af91 : SUCCESS in 9m 41s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/e3517dfa6ecc4ca4b2fad5059603fffc : SUCCESS in 14m 18s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/cc056f91e6bd4c1b9de755baa2ef7f43 : SUCCESS in 6m 37s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/c1f4128fc65744e8a176d087ff6b652a : SUCCESS in 17m 44s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/7b7955335aaa416a93e4b04bf151fc80 : SUCCESS in 5m 42s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/ae72fc25d7ff4e6baa3819d41c1afd85 : SUCCESS in 24m 55s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/ee200c6f7e004b8190dd4dc602d2b958 : SUCCESS in 12m 28s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/3dd1c6afc35743d38c5e3ff60f6731c0 : SUCCESS in 6m 05s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/93978c2d611a43ff85395f01e35178fd : SUCCESS in 4m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2c69b20ee4514a66868faecf453db4d5 : SUCCESS in 19m 58s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/4a5b453cfdb042c69422651d106636c8 : SUCCESS in 6m 34s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/e791d6eaf1b74545be2d66e59bd6df52 : SUCCESS in 5m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5b4959569a574f02810d07b573f5b967 : SUCCESS in 7m 26s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/1e3075080e9f4863a6d66e0724b821a2 : SUCCESS in 12m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7da2b7c38ffa4dfab6dd183f6fd8fa96 : SUCCESS in 49m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d40fde36f4d0425da371d2dcf3e6f60c : SUCCESS in 1h 25m 11s","accounts_in_message":[],"_revision_number":2},{"id":"87bb39b61eb10a35b3fc517cee46928e29facbdf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 23:48:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"36c40246b78d1db79fc4d016969a96904bdfc555","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 23:49:40.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f58c1f701b3740ef9c2a41657e612d1b : SUCCESS in 55s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b44c447fe6d0201a2b97c443ded289689c8ae84b","revisions":{"3c7d13bba7425cacb437c4eb2d1b81adc2be56d8":{"kind":"REWORK","_number":1,"created":"2022-01-07 14:46:20.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/09/823809/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/09/823809/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/823809/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/823809/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/823809/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/09/823809/1"}}},"commit":{"parents":[{"commit":"ceb2d650b53465c7ef6274f4ff20305a928e9639","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ceb2d650b53465c7ef6274f4ff20305a928e9639"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-01-07 14:46:18.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-01-07 14:46:18.000000000","tz":0},"subject":"update constraint for python-tackerclient to new release 1.9.0","message":"update constraint for python-tackerclient to new release 1.9.0\n\nmeta: version: 1.9.0\nmeta: diff-start: -\nmeta: series: yoga\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I39794072d0628cc4fa64272dc80860dbddd23846\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I29356f6087d4635529270cd2d663d9310957bbbe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3c7d13bba7425cacb437c4eb2d1b81adc2be56d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3c7d13bba7425cacb437c4eb2d1b81adc2be56d8"}]},"branch":"refs/heads/master"},"b44c447fe6d0201a2b97c443ded289689c8ae84b":{"kind":"REWORK","_number":2,"created":"2022-01-07 19:49:25.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/09/823809/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/09/823809/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/823809/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/823809/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/823809/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/09/823809/2"}}},"commit":{"parents":[{"commit":"c71f1c898aad22ca00306ec2c5b1998819e27ec3","subject":"Merge \"update constraint for python-ironicclient to new release 4.10.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c71f1c898aad22ca00306ec2c5b1998819e27ec3"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-01-07 14:46:18.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2022-01-07 19:49:19.000000000","tz":-360},"subject":"update constraint for python-tackerclient to new release 1.9.0","message":"update constraint for python-tackerclient to new release 1.9.0\n\nmeta: version: 1.9.0\nmeta: diff-start: -\nmeta: series: yoga\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I39794072d0628cc4fa64272dc80860dbddd23846\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I29356f6087d4635529270cd2d663d9310957bbbe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b44c447fe6d0201a2b97c443ded289689c8ae84b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b44c447fe6d0201a2b97c443ded289689c8ae84b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
