)]}'
{"id":"openstack%2Frequirements~823813","triplet_id":"openstack%2Frequirements~master~I04bfb842af5f70405b23039e7f36bae4ad002a23","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I04bfb842af5f70405b23039e7f36bae4ad002a23","subject":"update constraint for python-ironicclient to new release 4.10.0","status":"MERGED","created":"2022-01-07 14:51:55.000000000","updated":"2022-01-07 19:31:06.000000000","submitted":"2022-01-07 19:23:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"823813-new-release","meta_rev_id":"11d3229d4c18b0cc0357c085c8ceefe962f8e855","_number":823813,"virtual_id_number":823813,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-07 19:23:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-01-07 16:30:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-01-07 15:47:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-01-07 16:30:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-07 15:09:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-07 15:47:27.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-01-07 16:23:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-07 16:30:18.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"0ff76bccc83cd8786618d8ca1d21e20917f6aeb7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-01-07 14:51:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f8aed1e0e5f8446f09698cd96104569528e2aad5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 15:09:25.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/4d3467dc333c44489b382f001f7fda00 : FAILURE in 15m 19s","accounts_in_message":[],"_revision_number":1},{"id":"858c8305894d453bf536a7f854d2ba33865796c9","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-01-07 15:47:27.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"498da794483613412c9ed24ab9b87bee0208e137","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 16:23:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/115e9ac7d66048c8a611c03a433fef2d : SUCCESS in 5m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f839f681d6e447e8967a7cf68bb445df : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1bf8270bd22448e39b069695561f24d3 : SUCCESS in 4m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/21c4d94481724af0b9e5314ccebb6695 : SUCCESS in 5m 48s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5c6b3dc067074ac6b4ab41d530d9992e : SUCCESS in 5m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2912080de7204129850c54515654e44b : SUCCESS in 5m 26s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0716c646d9734ca38753d0153d338769 : SUCCESS in 10m 43s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d39f5fed1ddf4d9a91ae749f90a29b94 : SUCCESS in 11m 24s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/3e75dc7c45734ca1a4ae7a81c0b640a1 : SUCCESS in 6m 04s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/2fec8082b10b416a81303f78f5737dc3 : SUCCESS in 11m 20s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/d53181b0bd5a4135a62842b1643461d1 : SUCCESS in 6m 19s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/b510e46eeb4d483383a32766061fca12 : SUCCESS in 14m 32s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f09a7a88b11d4a05904bb2150f1c67e3 : SUCCESS in 10m 51s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/caf8297a6a034ff6822439d25b419013 : SUCCESS in 8m 21s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/2754b978e1164489a99e35fd8369bd71 : SUCCESS in 21m 41s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/7b8a122c10c84ecaa1d498fbe1bf8ed2 : SUCCESS in 5m 47s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/c928f8e2d65f4156b5ab88c43532d0d8 : SUCCESS in 20m 20s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/8b60a2b2928f4fca8ac01815cea25f41 : SUCCESS in 7m 00s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/83ade653d50945fcaf74b6c3064293a5 : SUCCESS in 21m 58s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/41e95868b36b4fab91b41a383611dfac : SUCCESS in 17m 55s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/97df40e15917495cb66892ad8da53ddf : SUCCESS in 5m 53s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0f7de3b1932348939a0908c5f37c0528 : SUCCESS in 4m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a428082a448e49a48de4bf1149debceb : SUCCESS in 20m 00s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/2f0f4e451aed4a2aaef87686c946b04f : SUCCESS in 10m 31s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/8ab661b4a95a4ee1bd6022e8ee34eb07 : SUCCESS in 6m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/89c1665fd73747618568944050b4bc65 : SUCCESS in 8m 39s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/e51218b0a4b1400fa0bd774068a8b261 : SUCCESS in 14m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a334985e0a054ba3a3d0143623942ad8 : SUCCESS in 47m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f1aa257d094c44cfb936cca67e43f8b1 : SUCCESS in 1h 26m 38s","accounts_in_message":[],"_revision_number":1},{"id":"9a15bf2dcde5d5c80fe5b12e6a79d687b5b8be72","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-01-07 16:30:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"57cf6cf59bfebc458543343ad2ac76e01deba1aa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 16:32:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2a23bf6b4c853727620dcb491794edfb121053a3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 19:23:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32c6f66d645844a8af9bb5e39e5e64ad : SUCCESS in 7m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e44876f02b844c37a4193bbaebcfda4d : SUCCESS in 5m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7f3c912b5fd7467298957bd034c2e16c : SUCCESS in 5m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/052abe3dac6b455f8d84ef87bb8f22d3 : SUCCESS in 6m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e6079d8a857049d893dd196047014305 : SUCCESS in 5m 22s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ec29a5b91cc5406ca0ffe37716c7af67 : SUCCESS in 5m 40s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4bfe3050f2db48ee87e4d233a67ebf69 : SUCCESS in 11m 34s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/4da8fa1a17d848ceaa4323d7b6d9cced : SUCCESS in 11m 50s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/6cca0735194d486d8eeedb48cb66acbe : SUCCESS in 14m 44s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/6c8bcf11709646afb0823fd8538ed3d5 : SUCCESS in 7m 34s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/0b06c928a3ba4dacac3a40e3d99cb020 : SUCCESS in 12m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/63b223a1299e4a4e8399df5d5fa090f6 : SUCCESS in 8m 51s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/dd24217b504f48e5830993d8a26e613e : SUCCESS in 10m 47s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/47fb68db1e8e4b36bf099fb94ee53494 : SUCCESS in 15m 34s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/70214f332af74f8b87d7ad3ff29c8868 : SUCCESS in 5m 38s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/29bc44b313eb48cb9365d435011e25b8 : SUCCESS in 18m 32s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/b7d52010ac2244109e76249dd82a405a : SUCCESS in 7m 26s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/babdcf2558024c4a8b7b36b48a8ee33f : SUCCESS in 24m 21s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/bdea78df8d8d4ae5ad1c78afeeda0597 : SUCCESS in 13m 02s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/92963cadd9cd4d51b90c9c02f7318cb9 : SUCCESS in 8m 10s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7482a2e6c57d4e33a0f5a9dbbef1dad2 : SUCCESS in 6m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/077317c2601d49ad9b8336320b272035 : SUCCESS in 21m 31s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/bcec3107ddf14b55a03b174c60f2c3a6 : SUCCESS in 7m 03s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/1893090eeddb413f9a9f2d6421010785 : SUCCESS in 7m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f662d05376064b42b1bc9ecfd7629df8 : SUCCESS in 7m 26s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/b836623c216b4061ad251feb57db7e1e : SUCCESS in 13m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6e060ed1cf93489d9cd6fd6ac115dd3d : SUCCESS in 45m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0d6914e7278847bda1ef900aa53a6d2b : SUCCESS in 1h 24m 06s","accounts_in_message":[],"_revision_number":1},{"id":"60a501bf400c0db8b23219d9cf1509db63da5e6e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 19:23:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"11d3229d4c18b0cc0357c085c8ceefe962f8e855","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 19:31:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d50f6520bb3472cb79a48f5f3a737b6 : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c3863d2c1c73b5253024b76f6f1306943b2a36ce","revisions":{"c3863d2c1c73b5253024b76f6f1306943b2a36ce":{"kind":"REWORK","_number":1,"created":"2022-01-07 14:51:55.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/13/823813/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/13/823813/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/13/823813/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/13/823813/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/13/823813/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/13/823813/1"}}},"commit":{"parents":[{"commit":"ceb2d650b53465c7ef6274f4ff20305a928e9639","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ceb2d650b53465c7ef6274f4ff20305a928e9639"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-01-07 14:51:54.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-01-07 14:51:54.000000000","tz":0},"subject":"update constraint for python-ironicclient to new release 4.10.0","message":"update constraint for python-ironicclient to new release 4.10.0\n\nmeta: version: 4.10.0\nmeta: diff-start: -\nmeta: series: yoga\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: Ia25e1971e884cc34738cb89ef4f4c232124382f2\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Dmitry Tantsur \u003cdtantsur@protonmail.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I04bfb842af5f70405b23039e7f36bae4ad002a23\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c3863d2c1c73b5253024b76f6f1306943b2a36ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c3863d2c1c73b5253024b76f6f1306943b2a36ce"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
