)]}'
{"id":"openstack%2Frequirements~824083","triplet_id":"openstack%2Frequirements~master~I6970bd998a515cd1116e7fe7265d75436215f6a9","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","attention_set":{},"removed_from_attention_set":{"14288":{"account":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"last_update":"2022-01-12 21:09:47.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I6970bd998a515cd1116e7fe7265d75436215f6a9","subject":"Updated from generate-constraints","status":"MERGED","created":"2022-01-11 02:52:04.000000000","updated":"2022-01-13 00:32:19.000000000","submitted":"2022-01-13 00:30:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"824083-openstack/requirements/constraints/noclob","meta_rev_id":"5273b5f87c92a148f5e3b6a75438846f1323d29d","_number":824083,"virtual_id_number":824083,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-13 00:30:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-01-12 21:10:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-01-12 21:10:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-11 03:15:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-11 04:38:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-12 04:31:40.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"ed03d367ea78b0dff8b5f31d016312668a4828bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-01-11 02:52:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7e3686b356e7b7efc48ab255a9fe93a8b65adc84","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-11 03:15:40.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/46ac84a1189e43758fe31aac568291e8 : FAILURE in 14m 12s","accounts_in_message":[],"_revision_number":1},{"id":"228dbc140f60859ab151e83796ff31dd588eb7ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-11 04:38:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25cb6cd9516444b8a20e619052d53625 : SUCCESS in 6m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/247bdffb465843beae1cab3d39fbcb9f : SUCCESS in 7m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2e11bc1166374e639a6740db90acd090 : SUCCESS in 7m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bb1bd414d00047f69c2efd0cd4ba141d : SUCCESS in 5m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f01db9d815ce4766abb151f0c71ddf94 : FAILURE in 6m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/dc8b2e75cf0d4f399786b75f99ee3057 : SUCCESS in 6m 51s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/38305e59c2444c7abcbb43ff72da4d66 : FAILURE in 7m 57s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/acf947f5812e4c26ae121529ab1e13fb : FAILURE in 7m 38s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/9056ad8f609d4a1a8af6b67e2801a595 : SUCCESS in 9m 46s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/c36d8aa0b3da46dbbdc4f59d15a7da36 : SUCCESS in 13m 38s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/a3a359d265684c5eb02a650cc0f19d0a : SUCCESS in 9m 28s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/05d87824eff34fb6adcd066cbef2efae : SUCCESS in 13m 07s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/406b2016950b4cc7bd9209b57481ba89 : FAILURE in 11m 19s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/411b66588e234b96809dcf6aed6846af : SUCCESS in 11m 01s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/184d594115a34a15aabd50ec084848d4 : SUCCESS in 15m 59s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/9459308413094365b15f9be4aca3b61d : SUCCESS in 5m 18s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/1e3c20f650344718b78f1264729461f4 : FAILURE in 28m 37s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/a0917fb147704727a81a9812d191c219 : SUCCESS in 7m 21s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/67fadf31381e4643b08338cfdf9fe463 : SUCCESS in 22m 54s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/82b3bdf5acb8488f8b1d99e9e8d3098f : FAILURE in 16m 03s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/8fd3b283fb6e452b95488774fff4f61d : SUCCESS in 8m 05s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/332360aefaaf40f78cdae85353570ae1 : SUCCESS in 6m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b5b4a36608a84cec80dc5e5b6b676b28 : FAILURE in 22m 18s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/a46bc41ac45948338278ed63a832ae9a : SUCCESS in 9m 12s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/6fbfccd8e7774c9794675f60259f1740 : SUCCESS in 7m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/356f20d43d4843caa28319bc7585712f : SUCCESS in 9m 04s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/5d25d9ea524d4d798e108f43f6ca62d2 : SUCCESS in 14m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/12c8090c8e3b4fec9d60c25096dd1233 : SUCCESS in 48m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3da8e8a461744e95b120009b17fe7e65 : FAILURE in 1h 36m 26s","accounts_in_message":[],"_revision_number":1},{"id":"94d20ca77b27e6ae05b117d8e8b468e5901ba34b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-01-12 02:55:34.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"30f7a06175d92df5dd8e494c5220b03b3cf132d2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 03:14:18.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/0d28e274cdbb46c79ca388b7fee00a0d : FAILURE in 12m 08s","accounts_in_message":[],"_revision_number":2},{"id":"6ca4cce19b7536d4464f82e609e380a8160ea467","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-01-12 04:31:39.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":2},{"id":"7636c80c110cf9e8d314301427002ca30713ecfe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-01-12 04:31:40.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"1af59aff903177694fd90b0482f030a06f04133f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-01-12 04:31:49.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"173077ecbb05f70bbbd02a8780c7b2856f477a18","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 04:44:45.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/89b28b5e4abc44969ae86215cb0bd43e : FAILURE in 8m 47s","accounts_in_message":[],"_revision_number":3},{"id":"6f87113d4ba1420a852f40a7fa3da2a5114c1248","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 05:57:18.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50a441a501f44aaebc9fd09e7e4753d7 : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91a62de7459c418d838c440713669756 : SUCCESS in 3m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/376ef36427ff451d9fe615158cb4c18f : SUCCESS in 4m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f26631bbf14e45548ea480fc582e4126 : SUCCESS in 3m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/75d2ca8e560b411f8b22232f4cf3d5e7 : FAILURE in 3m 22s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/53cb041b915a455785c4e142df61ff19 : SUCCESS in 3m 20s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/09e6761d31eb494fad5df3df5b83b595 : FAILURE in 3m 56s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6e7de49198ce44a79fd41507954a1ef0 : FAILURE in 3m 59s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/29358bf56b154abf8021f080e4ce2264 : SUCCESS in 5m 22s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/e4f30fc36258430788f2ad5d3a4dc194 : SUCCESS in 10m 00s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/0281406eb5b9416bb44dc51f019a0930 : SUCCESS in 5m 43s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/ac6a02a47fa94ef08730b4f0fbd0b4e7 : SUCCESS in 10m 45s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ba5794cd6a0d49929a064903a7963c66 : SUCCESS in 8m 01s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/2736474931ed459a8bce7bc1c82eb9c6 : SUCCESS in 8m 01s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/0bcd916656054b07bcdba55dab706018 : SUCCESS in 14m 22s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/c515cc91b4f1422c826e023aaad457a0 : SUCCESS in 3m 50s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/afddeac6bc9e4de989f018a7175c0e3d : SUCCESS in 23m 42s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/78a172d985494bac9160812fde57ffca : SUCCESS in 5m 04s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/e4ccddd23f3748a19b94123e50201a56 : SUCCESS in 24m 15s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/8b43c279848b419f8d73dd51e15269e3 : SUCCESS in 12m 03s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/cd43af56781f47d48843d1cda1dc7801 : SUCCESS in 5m 49s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/49c5a5ad02b0477087baa1a770e473b3 : SUCCESS in 3m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8dedd7a1c0bc4bd3b51ba9a7199ecc80 : SUCCESS in 19m 10s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/4df6a2901921419f829b25106ff7827d : SUCCESS in 5m 36s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/a2a1b693d9454c088e64e98ce72fa228 : SUCCESS in 4m 39s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e5d276c254754c85bf9d9397c99a5b3c : SUCCESS in 6m 44s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/08274b2ed7b04b178c10707a2b75a513 : SUCCESS in 13m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0ecdfc7681a44193956db78cfdf97bdc : SUCCESS in 37m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/57a7e649df9342ab9d8be57e45f425b9 : FAILURE in 1h 23m 31s","accounts_in_message":[],"_revision_number":3},{"id":"20f254ff2c29c32855842e31b1c604d47ed774ac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-01-12 06:04:12.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"481e924355a32b9b5ddb5b85645eb53ad68beb58","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-01-12 06:05:01.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"11aa85a24fd11cf5ddd7823fc6e32ddf64af0ff9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 06:23:12.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/0ac53b033d60417ab0c71421665951f3 : FAILURE in 12m 02s","accounts_in_message":[],"_revision_number":4},{"id":"369ae2ce6a2dc85cc555b5974813a8a3f4a48bed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 07:53:22.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3433b47d0c70479088dd16e9787259e1 : SUCCESS in 3m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/051b71ba560c4d44bfed9270a7eda8f6 : SUCCESS in 3m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b42aaf9140cd4f4394bbf04a3dae2492 : SUCCESS in 4m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b20d7213e74b4b1a9ea59d9bdb6c0c99 : SUCCESS in 4m 05s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2627b4f431c24c43a66361f1cbc04799 : SUCCESS in 3m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/52e0384520a54fdb83d7b2fe727d87dd : SUCCESS in 3m 33s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/206ecf4c228a4f70a11cc557bd611c7a : SUCCESS in 13m 38s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0ea7c32d60ba46f2b355d05abe792024 : SUCCESS in 10m 34s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/90e5c11c947c42c1a84dd8db37ef8a18 : SUCCESS in 5m 30s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/790fcee774644ebcbafc8bb2fa015191 : SUCCESS in 10m 36s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/c494ee75c42f46d981379dbfdb9ff0c0 : SUCCESS in 5m 47s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/1abdf07c7dfe4325be59bd5282d08f9e : SUCCESS in 12m 04s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/696e00f3bc724a52964c78b1a2631ae5 : SUCCESS in 9m 21s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f0ee4d7c7b7447a29a6e6b0a26868bac : SUCCESS in 8m 01s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/efeac42066ba4c31bd327b75d6d5a8c0 : SUCCESS in 19m 41s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/551d0b13d3f94b03886031e11686791d : SUCCESS in 4m 07s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/f4997b49bd26426cbd7b10a7cfe8e88e : SUCCESS in 15m 48s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/397bc1485a4f49c9b909815cc210073d : SUCCESS in 4m 59s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/6d043d5b37164589863dc1d695ca5e3f : SUCCESS in 29m 03s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/0e6080ca08b44518a3629d6402a94c8c : SUCCESS in 11m 35s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/9ef75217606d448cb2a8644482b03792 : SUCCESS in 5m 28s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/931388a66ec8445eadea11f826607642 : SUCCESS in 3m 43s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ac8ca1b0e5bc46bfb6333f857a60c5d2 : SUCCESS in 24m 07s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/b193b424818a43b2bc77ed14778525c8 : SUCCESS in 5m 39s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/7d316a939f0b4c08a5fa02a113386aaa : SUCCESS in 5m 23s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b60d243e870d430baf485135d30509e6 : SUCCESS in 7m 08s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/852d863f274a4391a6ba84557380ba89 : SUCCESS in 12m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/990d9f3db17d497cb65dcbadb81d3143 : SUCCESS in 50m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b692c05a9ca94609846d63c7225fb980 : FAILURE in 1h 39m 12s","accounts_in_message":[],"_revision_number":4},{"id":"dd94593f07443e113b85b107b79183e4bc588a62","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-01-12 08:06:43.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"3971738235f47191b9d5414e0c6f49cac94025fd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 08:25:03.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/cf3ae012cd474c6b9cf936daa94e0c14 : FAILURE in 16m 20s","accounts_in_message":[],"_revision_number":4},{"id":"df2bf4d2697aa50dc1010af4f1b2dbcc6d7e8b8d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 10:02:49.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d0e227ec69b4d64b6fe4c63bf23609e : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0068f7e265464e49bc15d5d3386cbacd : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/29d4ea890ac04aa58efb416cfb96b049 : SUCCESS in 4m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/60909ea01ca14bf5a1a9076d4e1673e5 : SUCCESS in 4m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7e55ab697058407b93ce86fc817d4b5d : SUCCESS in 4m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4fde64019aed4caab128b3db5df2174c : SUCCESS in 4m 35s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d80f506be7f143fe83bbc2c6a5f1ae39 : SUCCESS in 10m 29s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/7703f957acc645829344ec87efc991e1 : SUCCESS in 11m 12s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/127bd272aac94db2ad6e2d14d87e9cc1 : SUCCESS in 6m 32s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/de4e95d28ffc49fa80e49c5dafd681f4 : SUCCESS in 11m 14s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/89db1a42e38944ce9a385adfb9e6b9ce : SUCCESS in 6m 18s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/97d86541388d4398bbe3b3bd62d9d4d9 : SUCCESS in 11m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/91e87b39ad694e2493093ce55331e067 : SUCCESS in 10m 05s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/145d23545a604a97b5885ef2aa0cea79 : SUCCESS in 9m 47s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/1768c41bd83f42609c42a338cc3bf0c7 : SUCCESS in 13m 50s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/85e3e8b41eb54bf29493192bd3980976 : SUCCESS in 4m 56s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/53d524a496a74908b190f01b0f727c1b : SUCCESS in 16m 27s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/859143af946b44c598eee7bbd9c075b4 : SUCCESS in 5m 14s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/7739ba4e08a9453e9a275a8f35c4d8ca : SUCCESS in 24m 47s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/783ab0cb78ea48d98287bdc4b063c2e9 : SUCCESS in 13m 24s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/d1fcc5cb55d141e3983f1d0c425c1860 : SUCCESS in 5m 52s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ee36abba5b3f43189d57ea3f6563ca6a : SUCCESS in 4m 03s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c7af02fe2d554604be4f659649af6015 : SUCCESS in 19m 55s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/29743bb503064cd3b56f653d71359d80 : SUCCESS in 7m 02s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/3ccc851e12374fde87210fd674353a3b : SUCCESS in 5m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/71ce1204f8b34d92a12ce7bf54c45c2c : SUCCESS in 7m 24s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/9ad20744489d4e75a39cb08742f8453c : SUCCESS in 13m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/813cb3b77436439c8974e79f38edcb76 : SUCCESS in 48m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fdfd175171c2498685ead364cbda4518 : FAILURE in 1h 49m 35s","accounts_in_message":[],"_revision_number":4},{"id":"01d282e414fe778c84712dc1a58fb23f2dfad014","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-01-12 16:10:08.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"8bdddb5982f16d5cb6ef2ed6b7cef463635d9561","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 16:34:19.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/10d8380b7c20468193e04514f5961ba1 : FAILURE in 17m 21s","accounts_in_message":[],"_revision_number":4},{"id":"efad77174b43143af3628343e93013719065f671","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 20:58:39.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/993c0d9f52004a57ad30da7470012fcb : SUCCESS in 6m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22054a58c20547b4914264112d5cd73c : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ad5664d16fdf431da3660281fc5b0fea : SUCCESS in 5m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d97b27b041664f22b54e9ce84b0a8106 : SUCCESS in 6m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/75925bb26cf04eddae1e51463250213c : SUCCESS in 5m 45s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c59c6a3549df44bf879a1d5eaf7fef33 : SUCCESS in 8m 11s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/20e584fdcd654baa8b418718b5ee75a3 : SUCCESS in 12m 03s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/7a89a524707546baafd8a8c0f14a7c19 : SUCCESS in 11m 50s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/fe63d2a089e7474993b549de2ad71bfa : SUCCESS in 9m 34s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/cbb1e6d998af41989ec416e16c10b58c : SUCCESS in 12m 34s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/7ac09d0fa5bd4abf92d930d5136d9d16 : SUCCESS in 7m 34s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/f68f832f118148d5a6bc85861e1502ea : SUCCESS in 12m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cb8ad19dea344f428162347699ddae20 : SUCCESS in 10m 08s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/caf20ddb05214660beb782bebd61fbea : SUCCESS in 9m 36s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/10705ac52f69470a94fa86a12cc1603f : SUCCESS in 16m 35s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/d2a60440f0a8425b87a740a1b3189024 : SUCCESS in 5m 48s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/332ec1de4e2f485dab808ebb9a7910a1 : SUCCESS in 17m 46s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/41f3605f6bb74679b218a6e00a2785e6 : SUCCESS in 6m 37s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/fde0e06ed8994852b77f31e1e131c97b : SUCCESS in 29m 06s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/4abd0c45465a4f0e9ac79dcc508a2989 : SUCCESS in 13m 10s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/f8c102d4edd9468a971d064922b223f1 : SUCCESS in 7m 40s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/268394e2267345458a021e3b077f4fb6 : SUCCESS in 5m 03s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0076d9278fe5401aa8032e61d438589f : SUCCESS in 22m 48s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/dabf158512f74d38b21f154d0ad1a374 : SUCCESS in 7m 27s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/f523ca103d5f423491ed28cc89b80e1e : SUCCESS in 6m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/945501cb512b4ac3b0307c469672b996 : SUCCESS in 8m 31s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/35fac78445094bed8c4f715fe000c095 : SUCCESS in 12m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/07959aa4178d46f5b122893ef702552b : SUCCESS in 50m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/56d1adeded7947bc90dd84b39826dc27 : FAILURE in 1h 38m 22s","accounts_in_message":[],"_revision_number":4},{"id":"670c3a9211d12aa8764ab5037d1badbcf8a71142","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-01-12 21:09:40.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"869165bde0420bfce18b46a3b5d17d8b9f8f1752","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-01-12 21:09:47.000000000","message":"Patch Set 4:\n\nheld back paramiko","accounts_in_message":[],"_revision_number":4},{"id":"b7d9a5b6c3941106a9eb201b50f9841b6ff94b6d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-01-12 21:10:26.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"8b717f856127a89ed5da6755efb1ee439dfdf2ec","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 21:25:53.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/990456643e7f415d916c6835f132e10e : FAILURE in 12m 29s","accounts_in_message":[],"_revision_number":5},{"id":"38d251b291da7b53d3f8674ea3b258a8e4adc44f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 22:41:24.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe127b3025ec4f05b8667b96200e5ec3 : SUCCESS in 3m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3c8c11d81784288b98e077990484333 : SUCCESS in 3m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e383076ab7094edb81b67ea63cfd5cbf : SUCCESS in 4m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fb99bb6c667c4934801c8df5ed3c8edf : SUCCESS in 5m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8cd348e75ccc4ccb95150c98c26cd81b : SUCCESS in 4m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7e0201a5d5bf456da9160e948817cecb : SUCCESS in 5m 08s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5fca692739ed464189bbdc31f7c04dae : SUCCESS in 12m 57s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2afe4a992fb04a95a61f1eb3410f6b83 : SUCCESS in 12m 16s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/2602de39446e449b931c3f751c456829 : SUCCESS in 6m 22s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/c00d1e312a224f6b99b260b1aa1ed752 : SUCCESS in 12m 37s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/15c2033d04d142d4bfcd1fcd8a357a07 : SUCCESS in 6m 20s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/e26489f3e8604ef4b3a2fa61d1d3a04a : SUCCESS in 12m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/47f5919203504b70a0f9697b6aa9dd3c : SUCCESS in 10m 09s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/893952ef41b34eabad4b58ec35471956 : SUCCESS in 11m 13s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/ecd6f2de7700433abed0093e592a6451 : SUCCESS in 16m 08s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/c818dfcba96f41948d5c8d20081d0564 : SUCCESS in 4m 55s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/82c9b7c97fb54ffda318ba85f036d5ab : SUCCESS in 17m 22s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/901469d196c44777a616d62c5290f329 : SUCCESS in 5m 19s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/3f99f83654c442c581893020613adf6c : SUCCESS in 19m 51s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/027627d652fa4f63bcbe705e6ad57a9d : SUCCESS in 12m 06s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/2e8719d50cd0442f81ee2f473cabfbb2 : SUCCESS in 5m 57s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9ccc579ac88e45d3a420b9fe8ee7d334 : SUCCESS in 3m 58s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/59f932159e0f4b08bb999ac6ef0b3fab : SUCCESS in 19m 35s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/4bca995deecd46f79a6d6dfa9a10debc : SUCCESS in 6m 03s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/2d806b424b274096a6d40870223f0888 : SUCCESS in 5m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3552a3cd093b4aeeb40091ade4b5e4a7 : SUCCESS in 7m 49s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/7623079c541346bf843ac8560dfa272e : SUCCESS in 11m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/68aeebe2f78d4661a5596041cee664e9 : SUCCESS in 48m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9fdb3528593c4b9d84e4aef2e61f1934 : SUCCESS in 1h 28m 20s","accounts_in_message":[],"_revision_number":5},{"id":"7fb5e4a9512aaf81d23446f4b636f2116c5431d7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 22:42:17.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"882f5d12eb5b0b689e2538262467980c3f2365d9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-13 00:30:45.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e751cd35aa3442708b6c3a1cee524d42 : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f6373ff687a40b6af191cca2d0e6136 : SUCCESS in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/77133112fb164051a4a5cd2a4832c50b : SUCCESS in 4m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1d66b2e72751420bac9de5e5c6bd7caf : SUCCESS in 7m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9617540665a04503b24d453a8ba00190 : SUCCESS in 3m 35s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/52ca232ef5334512af5f7d59ba62a288 : SUCCESS in 5m 51s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/767d3c1182a34413b4274e3a61d02fdf : SUCCESS in 10m 14s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/567b42d0e9744df7862c35acd341cc46 : SUCCESS in 10m 46s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/6faa0f271e0d4453b715f57fbc1843ef : SUCCESS in 11m 38s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/c9c2264212924212973c6e59a7f8015e : SUCCESS in 6m 55s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/ac9db57d0ac148349f4fe7f37d392ef7 : SUCCESS in 10m 57s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b00c9d6577ac4e1fa5a3196708b982a0 : SUCCESS in 7m 51s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/2927d1a779214c63bfd3fc4e960edb18 : SUCCESS in 9m 00s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/bebc88c168cd4e8da17b398f9823b7dc : SUCCESS in 13m 36s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/f8f5521f7e2e4ff38d8f06f8162cf4f2 : SUCCESS in 4m 54s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/47e69e7bdc4f497b834b90f34bf22e30 : SUCCESS in 17m 25s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/9b9bd38ae5d844f1bba62c29988cfa15 : SUCCESS in 5m 16s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/ebb9aa58ac4b4455864835945003022f : SUCCESS in 27m 37s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/30eb28d652264f1783192e894de243f8 : SUCCESS in 11m 39s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/d3ff32622ed94232997bac21d409c3f1 : SUCCESS in 5m 10s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/56aba2ce90c748fe86be76f45e3c99fc : SUCCESS in 3m 37s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8f92b2039add4ee7a72fa0ecdc44a856 : SUCCESS in 20m 24s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/1691ba67fbd54a77966ccb48955f8562 : SUCCESS in 5m 30s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/cb290aa9d34d448888fafca4cb333225 : SUCCESS in 6m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/03b1cb00f0bd4ae9bc93d4b49dbcec73 : SUCCESS in 7m 34s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/b97cac9b6e4b49168d98e658fcd6ec6c : SUCCESS in 11m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/19e346f050164d77bc323d92de3b4fb5 : SUCCESS in 47m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f94c32d5d02049cf9071b2f2e3908138 : SUCCESS in 1h 37m 25s","accounts_in_message":[],"_revision_number":5},{"id":"bab96ec10f547584647c4046a21a970d380ca51c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-13 00:30:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"5273b5f87c92a148f5e3b6a75438846f1323d29d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-13 00:32:19.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03e621282402422282568884943a269c : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"9427e10ed9039d11c527bf933dfbd34d9e55bcd8","revisions":{"76382ddca4ab6ae4cbd6ede8b6549e0db1a21301":{"kind":"REWORK","_number":1,"created":"2022-01-11 02:52:04.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/83/824083/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/83/824083/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/824083/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/824083/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/824083/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/83/824083/1"}}},"commit":{"parents":[{"commit":"1fad8fafb5a46869c1f0df9dd458bea1a58251d1","subject":"Merge \"update pysaml2 to 7.1.x\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1fad8fafb5a46869c1f0df9dd458bea1a58251d1"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-01-11 02:52:03.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-01-11 02:52:03.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I6970bd998a515cd1116e7fe7265d75436215f6a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/76382ddca4ab6ae4cbd6ede8b6549e0db1a21301"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/76382ddca4ab6ae4cbd6ede8b6549e0db1a21301"}]},"branch":"refs/heads/master"},"2da2be70368fff63bea9bc71c90364319ed088ed":{"kind":"REWORK","_number":2,"created":"2022-01-12 02:55:34.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/83/824083/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/83/824083/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/824083/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/824083/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/824083/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/83/824083/2"}}},"commit":{"parents":[{"commit":"803231efa6cf05368eaecf61a902a9e2c44f03f1","subject":"Remove centos-8 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/803231efa6cf05368eaecf61a902a9e2c44f03f1"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-01-12 02:55:32.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-01-12 02:55:32.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I6970bd998a515cd1116e7fe7265d75436215f6a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2da2be70368fff63bea9bc71c90364319ed088ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2da2be70368fff63bea9bc71c90364319ed088ed"}]},"branch":"refs/heads/master"},"3736ce6f7dd31d934e783c2b7b18cefd8b31b5bd":{"kind":"REWORK","_number":3,"created":"2022-01-12 04:31:40.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/83/824083/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/83/824083/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/824083/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/824083/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/824083/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/83/824083/3"}}},"commit":{"parents":[{"commit":"803231efa6cf05368eaecf61a902a9e2c44f03f1","subject":"Remove centos-8 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/803231efa6cf05368eaecf61a902a9e2c44f03f1"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-01-12 02:55:32.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2022-01-12 04:31:17.000000000","tz":-360},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I6970bd998a515cd1116e7fe7265d75436215f6a9\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3736ce6f7dd31d934e783c2b7b18cefd8b31b5bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3736ce6f7dd31d934e783c2b7b18cefd8b31b5bd"}]},"branch":"refs/heads/master"},"448085eaf63d97a591122621d811565679f65972":{"kind":"REWORK","_number":4,"created":"2022-01-12 06:04:12.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/83/824083/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/83/824083/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/824083/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/824083/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/824083/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/83/824083/4"}}},"commit":{"parents":[{"commit":"803231efa6cf05368eaecf61a902a9e2c44f03f1","subject":"Remove centos-8 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/803231efa6cf05368eaecf61a902a9e2c44f03f1"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-01-12 02:55:32.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2022-01-12 06:04:02.000000000","tz":-360},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I6970bd998a515cd1116e7fe7265d75436215f6a9\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/448085eaf63d97a591122621d811565679f65972"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/448085eaf63d97a591122621d811565679f65972"}]},"branch":"refs/heads/master"},"9427e10ed9039d11c527bf933dfbd34d9e55bcd8":{"kind":"REWORK","_number":5,"created":"2022-01-12 21:09:40.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/83/824083/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/83/824083/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/824083/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/824083/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/824083/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/83/824083/5"}}},"commit":{"parents":[{"commit":"803231efa6cf05368eaecf61a902a9e2c44f03f1","subject":"Remove centos-8 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/803231efa6cf05368eaecf61a902a9e2c44f03f1"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-01-12 02:55:32.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2022-01-12 21:09:30.000000000","tz":-360},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I6970bd998a515cd1116e7fe7265d75436215f6a9\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9427e10ed9039d11c527bf933dfbd34d9e55bcd8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9427e10ed9039d11c527bf933dfbd34d9e55bcd8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
