)]}'
{"id":"openstack%2Frequirements~824552","triplet_id":"openstack%2Frequirements~stable%2Fvictoria~Ia046e4913eb266f0cc539fa380289bae93c62271","project":"openstack/requirements","branch":"stable/victoria","topic":"new-release","hashtags":[],"change_id":"Ia046e4913eb266f0cc539fa380289bae93c62271","subject":"update constraint for neutron to new release 17.3.0","status":"MERGED","created":"2022-01-13 12:04:43.000000000","updated":"2022-01-13 19:01:08.000000000","submitted":"2022-01-13 19:01:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"824552-new-release","meta_rev_id":"c78897231d39311e82db1086d5575758329bb30a","_number":824552,"virtual_id_number":824552,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-13 19:01:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-01-13 17:06:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-01-13 17:06:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-13 12:32:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-13 13:49:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-13 17:06:00.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"dfe4c55371394d8f26136df4955ab332076f20b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-01-13 12:04:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b5873d96b7790cabadf78b98652a657d51d7a5fc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-13 12:32:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/691f5177cb86463fb97f03dc27f37c50 : SUCCESS in 18m 31s","accounts_in_message":[],"_revision_number":1},{"id":"efb82b2e01567886dda723a197e000ecd0fc9cb3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-13 13:49:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c4eecf1b5604b9991d320532e9c34a5 : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/742d9d71ebb943cd85cf9cb55eecc18a : SUCCESS in 5m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b72175af52e7403e82ecd4cd78c0e2e4 : SUCCESS in 5m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3ab593e0d30640579d76cbdea8589578 : SUCCESS in 4m 43s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7c934c56b2d042988f9c53ea5b44cd03 : SUCCESS in 8m 39s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/27af1b993197498b9dba2cf5e4f35e1e : SUCCESS in 19m 01s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/95bdade50faa400dae49713ca63d7302 : SUCCESS in 9m 02s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/f2071315c55a433ca123c791ca8742a9 : SUCCESS in 9m 56s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/cca8848b832744bcb541ad8f48672c56 : SUCCESS in 6m 48s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/2e72da3042974256935d3eace0b992f5 : SUCCESS in 11m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/32dc61dbaa1d481c857d4769a5c791a9 : SUCCESS in 9m 59s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/e0df1a2f46d14beeab377ee40bd9be54 : SUCCESS in 7m 43s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/0f8253795b91442da32aca3a200cb5af : SUCCESS in 13m 58s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/08eeaf2d6dbc4705bf028afe91ccd48e : SUCCESS in 22m 54s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/9df0557946b24960aaf1224b02fdf050 : SUCCESS in 14m 03s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c3846470589c4dad9102d70f636bb9f5 : SUCCESS in 4m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f94214b5ec324e17ae1bd5846bc0e67e : SUCCESS in 19m 03s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/013db9cd4db54b57a0a79fcd0ee78413 : SUCCESS in 5m 33s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/527626eca48f4a229586ec59a370104a : SUCCESS in 8m 18s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/81dadbf97bab4193b3a4975de588a9fd : SUCCESS in 12m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a541308ff2954ffc9d017fc4710642c9 : SUCCESS in 59m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5e153569c90742e28fabc1a734a6f245 : SUCCESS in 1h 26m 55s","accounts_in_message":[],"_revision_number":1},{"id":"967b8dd784853711632f3b41334c5351d89b95be","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-01-13 17:06:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a7654ce4df6b5872bc7c6961692e35ccd6b5c833","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-13 17:07:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bd950e9a6b7367ef0b91df2bab8fa8dbdae1a098","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-13 19:01:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9bbd9182b92b4651ba87087ff02484c9 : SUCCESS in 6m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/62f3678b8d1944938a069401a3b2e930 : SUCCESS in 5m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/564f821c99af427f956935328292242c : SUCCESS in 5m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1a62bb0c9580426588b88d09e52c00e2 : SUCCESS in 5m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b245aaf3cf374c09bd63f665c0dee816 : SUCCESS in 7m 38s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3e2d79d204c547d3b7de484f82d3ba4d : SUCCESS in 12m 14s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/146a2e58b1ca415cbb572d8066619fe2 : SUCCESS in 11m 09s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/c9b69f1518924e7e9c31c6bef9277914 : SUCCESS in 12m 05s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/b91bd8d5488e4e51bc1e7fbabec86678 : SUCCESS in 9m 45s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/4cd9b8315aef45baabcd752a23c881ec : SUCCESS in 13m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4d52feb4e6694e96a976cfd361ecb2a6 : SUCCESS in 14m 19s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/2b5fca0e361c415188ee4a73340b524b : SUCCESS in 10m 04s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/065fcd2a80f64c23b245aa5781d14a00 : SUCCESS in 23m 50s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/39659945fbbf48b1ac5b3dbc541b52ee : SUCCESS in 25m 10s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/96521582cebf4c5cb31ccabdcd83bcf6 : SUCCESS in 16m 54s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a083ce162b5946a5b9f3e00ed26de0a5 : SUCCESS in 5m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e603d9575bce46e285add8e2732588d9 : SUCCESS in 17m 47s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/543b63ee2f7744e7bc571bd188d34b1d : SUCCESS in 7m 33s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/371ce5e418994a1db92b1135f1456eae : SUCCESS in 8m 24s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/20f56cc0d9094e9e9338277cafb09d94 : SUCCESS in 15m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/602d5021291145f9bfbecbc570d5a1b1 : SUCCESS in 58m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a7c0c910bd084478a0791992d454b2fe : SUCCESS in 1h 27m 38s","accounts_in_message":[],"_revision_number":1},{"id":"c78897231d39311e82db1086d5575758329bb30a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-13 19:01:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d44f078076454da6b7c340814879442cee086354","revisions":{"d44f078076454da6b7c340814879442cee086354":{"kind":"REWORK","_number":1,"created":"2022-01-13 12:04:43.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/52/824552/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/52/824552/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/824552/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/824552/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/824552/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/52/824552/1"}}},"commit":{"parents":[{"commit":"384cbaf81a811888a21bb2660a0580fa4197a185","subject":"Remove centos-8 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/384cbaf81a811888a21bb2660a0580fa4197a185"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-01-13 12:04:42.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-01-13 12:04:42.000000000","tz":0},"subject":"update constraint for neutron to new release 17.3.0","message":"update constraint for neutron to new release 17.3.0\n\nmeta: version: 17.3.0\nmeta: diff-start: -\nmeta: series: victoria\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: elajkat \u003clajos.katona@est.tech\u003e\nmeta: release:Commit: elajkat \u003clajos.katona@est.tech\u003e\nmeta: release:Change-Id: I5da83c078a107c0040ec28cbd303afb1941690ca\nmeta: release:Code-Review+1: Bernard Cafarelli \u003cbcafarel@redhat.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Ia046e4913eb266f0cc539fa380289bae93c62271\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d44f078076454da6b7c340814879442cee086354"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d44f078076454da6b7c340814879442cee086354"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
