)]}'
{"id":"openstack%2Frequirements~824699","triplet_id":"openstack%2Frequirements~master~Iad846270dabcdeefaf1e3973b8ca478d20bd12de","project":"openstack/requirements","branch":"master","topic":"remove_futures_py2","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2022-01-14 16:25:51.000000000","reason":"Rodolfo Alonso replied on the change"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2022-01-14 11:51:57.000000000","reason":"Slawek Kaplonski replied on the change"}},"hashtags":[],"change_id":"Iad846270dabcdeefaf1e3973b8ca478d20bd12de","subject":"Remove \"futures\" from constraints","status":"ABANDONED","created":"2022-01-14 11:45:53.000000000","updated":"2022-01-14 16:26:04.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"def0f21168ad54e5c9c4c8461353d6742aaf44fd","_number":824699,"virtual_id_number":824699,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-01-14 13:24:10.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":-2,"date":"2022-01-14 15:33:31.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2022-01-14 11:51:57.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-14 11:51:25.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2022-01-14 12:04:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-14 13:24:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-14 15:33:31.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"04651752cfc4980d7f9fa6ce9828b3cd7b3c32d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-01-14 11:45:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a77546527c521184b6f32ca6d870883c373223ae","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-01-14 11:51:57.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"70bda35ccb58a3a373bd86a082a480ad5824a58b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-14 12:04:03.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/797c5f86cf104dbfa4eabe598b08557e : FAILURE in 13m 14s","accounts_in_message":[],"_revision_number":1},{"id":"f5b05b76a12f4d0163b45309300ba902927e5ca2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-14 13:24:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6094832bf4bf476f8a7f4a05c7dfb380 : SUCCESS in 5m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f04dc9acafa042169e911c8f015280a1 : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/99104db110bf437aaa263549c83185e2 : SUCCESS in 4m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d3ef9c4248e043428096f251678f928f : SUCCESS in 5m 59s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b8dfeb87da5b4d1ca6dc45007bd20ec8 : SUCCESS in 4m 35s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f80d97dc2a7447c1ae94f751dfba199a : SUCCESS in 5m 09s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5791a44606254a5e9e7ffb60bd4c98a3 : SUCCESS in 12m 37s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/999c46647627428b9fc19599f4af1c56 : SUCCESS in 12m 04s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/68f53ba57d9b4b609ae3c5e68d01bb29 : SUCCESS in 8m 32s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/4190f60d82404ca8840d2c4f28cb2ad0 : SUCCESS in 11m 57s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/d5e04d3af9824cae8edb0d654a63a097 : SUCCESS in 6m 42s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/5034d8877d544d2f86179a4d625c0cb6 : SUCCESS in 12m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6744e1af8f7d4ef99fc9fb72c56e3ce1 : SUCCESS in 10m 56s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/7c50fada70ad46ad894d60b61fed962e : SUCCESS in 9m 31s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/4cfd243bc20b427db01717d301b4fa8c : SUCCESS in 16m 00s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/813ea910250c4d26b42be4a30262f63a : SUCCESS in 4m 55s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/c6c1aaca326c4d7dac43bd3a24b0bf8c : SUCCESS in 17m 18s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/d9f3eede7dea4087a2f07059cc196df3 : SUCCESS in 6m 38s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/a224eed2244842cdac58cc5ad3fbad78 : SUCCESS in 25m 55s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/4e49b5655e8043ceb081ab1dce760146 : SUCCESS in 13m 03s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/cf48ecd426b14057b91c365c0047e199 : SUCCESS in 7m 57s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a6a4e6f001424f9b9bd44dc08db473de : SUCCESS in 5m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e06a456043514d6899cf5df1053a1cf1 : SUCCESS in 21m 29s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/6cfaa28bbd194de9935a4492884685a8 : SUCCESS in 8m 04s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/9085692d16754e26a4c86bfaba71dc22 : SUCCESS in 7m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0757b072254c4441aacf467bc352b87c : SUCCESS in 7m 41s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/223a8d02edc44db9821fddcb7f23985f : SUCCESS in 13m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d4a648720b324c7faba496f5f1d3ec50 : SUCCESS in 49m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e2b0f4106c6e439883583bfb41181add : SUCCESS in 1h 24m 58s","accounts_in_message":[],"_revision_number":1},{"id":"970ebf76f43a4702a32eacd6a595ee2b2e4e9960","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-01-14 15:33:31.000000000","message":"Patch Set 1: Code-Review-2\n\nthis dep is brought in automatically (the constraints list is a computer generated list made from installing from global-requirements)","accounts_in_message":[],"_revision_number":1},{"id":"3f55106bc792f1a1a36b91c113e072d8ecf92e5d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-01-14 15:34:06.000000000","message":"Patch Set 1:\n\n(there is some dep that brings it in, not an openstack controlled one iirc)","accounts_in_message":[],"_revision_number":1},{"id":"d61e99d248d8977747b01956148f0a470ac949d5","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-01-14 16:25:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"def0f21168ad54e5c9c4c8461353d6742aaf44fd","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-01-14 16:26:04.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b9ff9c5a38f0daa2d04a34d19ae7eedb2005ff9a","revisions":{"b9ff9c5a38f0daa2d04a34d19ae7eedb2005ff9a":{"kind":"REWORK","_number":1,"created":"2022-01-14 11:45:53.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/99/824699/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/99/824699/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/824699/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/824699/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/824699/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/99/824699/1"}}},"commit":{"parents":[{"commit":"382c647def0307d5fad17185cf121dfbcc9ff9a7","subject":"update constraint for validations-libs to new release 1.5.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/382c647def0307d5fad17185cf121dfbcc9ff9a7"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-01-13 10:26:15.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-01-13 10:26:15.000000000","tz":0},"subject":"Remove \"futures\" from constraints","message":"Remove \"futures\" from constraints\n\n\"futures\" library is a backport of the Python3 \"concurrent.futures\"\nstandard library module to Python2. This limitation does not make\nsense in a Python3 only project.\n\n[1]https://pypi.org/project/futures/\n\nChange-Id: Iad846270dabcdeefaf1e3973b8ca478d20bd12de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b9ff9c5a38f0daa2d04a34d19ae7eedb2005ff9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b9ff9c5a38f0daa2d04a34d19ae7eedb2005ff9a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
