)]}'
{"id":"openstack%2Frequirements~824853","triplet_id":"openstack%2Frequirements~master~Ia4a1ca9b4138a111f152da6cb54eb9e168fcc557","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints","hashtags":[],"change_id":"Ia4a1ca9b4138a111f152da6cb54eb9e168fcc557","subject":"Updated from generate-constraints","status":"ABANDONED","created":"2022-01-17 02:54:33.000000000","updated":"2022-01-17 05:37:14.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"96839ba8767a696cdf1bdc12e1c7aee26d7080ee","_number":824853,"virtual_id_number":824853,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-01-17 04:40:26.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-17 03:09:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-17 04:40:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"718e77e4e6a52a2b1ba6d1cf1ab055ece24b8561","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-01-17 02:54:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cdf35e4f93b6d3d79b72d3fc3cb33862f7445406","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-17 03:09:36.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/b0fc314f6e394e199bc6d43e2cc5a169 : FAILURE in 9m 53s","accounts_in_message":[],"_revision_number":1},{"id":"f043a0e443ce68673c07d8fd1da573ea41d57118","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-17 04:40:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/713ed4a4cc784a86aecce411ccdc46a7 : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92b1ceb34df34b7f8c9b3020b43bd9b4 : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8f319911c3fc484389dd4b561daf94b0 : SUCCESS in 4m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a48a238aa56845ea8f5e9f29d2282668 : SUCCESS in 4m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e8df07611c2a4506acbd58ba37f60518 : FAILURE in 4m 55s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f3ff4d0c9d3446619957f8ff74908e7f : SUCCESS in 3m 53s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5dc265e6fff54556ac1d67409c85d54f : FAILURE in 6m 34s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d2ba9df194434eb1b768833d687b0fd2 : FAILURE in 4m 33s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/8dea848f829e48a3990bbc2d156b2a67 : SUCCESS in 6m 06s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/f7c47294900243a7976f900f10999604 : SUCCESS in 11m 19s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/a1a9f6a6b9414337b400b6d3569f5755 : SUCCESS in 5m 47s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/87daa2313eb74a25bcf7c19dddf28581 : SUCCESS in 11m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7933ebfd90a448b0a25c77ffece06482 : FAILURE in 6m 48s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/17e46a80b3694e04a2cafee0c49db928 : SUCCESS in 9m 46s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/2482b8689eb74029aaa707871a699447 : SUCCESS in 14m 29s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/0a65242b7a3c444f9d70ddcb69e65996 : SUCCESS in 4m 20s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/f45367f326a5467297caadc61750697b : SUCCESS in 16m 45s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/4da7355e329849c0987d6efa09912f51 : SUCCESS in 4m 58s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/7941c503f47e4d71a377f5a6cdac979f : SUCCESS in 24m 57s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/40d7ad1684a94ca385bb025b65e08114 : SUCCESS in 11m 39s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/84456119c9c443b3ba2845760fa31b3c : SUCCESS in 5m 14s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/cfdc45629754459ca83f0dad34fbc1bd : SUCCESS in 4m 03s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b6fb5e5e8b0d4f878c0ed75dfc927a4c : SUCCESS in 18m 39s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/5141d70391d54045b478fef722a8c3ce : SUCCESS in 5m 41s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/f2771fbc90ea4009a65caecde8a49d31 : SUCCESS in 5m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7ce3fbe60cce4e4e83971cf3d39d9415 : SUCCESS in 6m 12s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/4f88f54201c540f2ac8a5d5af45695c4 : SUCCESS in 11m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/299a6984666048a6a3d53740c6227f8c : SUCCESS in 48m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f7a0f31ff9e842f4ac2d87e51d1d14cd : FAILURE in 1h 40m 45s","accounts_in_message":[],"_revision_number":1},{"id":"96839ba8767a696cdf1bdc12e1c7aee26d7080ee","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-01-17 05:37:14.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d07465e77b65685c9dfe645551ca9af1ca9c3b79","revisions":{"d07465e77b65685c9dfe645551ca9af1ca9c3b79":{"kind":"REWORK","_number":1,"created":"2022-01-17 02:54:33.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/53/824853/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/53/824853/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/824853/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/824853/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/824853/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/53/824853/1"}}},"commit":{"parents":[{"commit":"adcb99ae08cafabe11a5d0be08621c56180c322c","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/adcb99ae08cafabe11a5d0be08621c56180c322c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-01-17 02:54:31.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-01-17 02:54:31.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ia4a1ca9b4138a111f152da6cb54eb9e168fcc557\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d07465e77b65685c9dfe645551ca9af1ca9c3b79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d07465e77b65685c9dfe645551ca9af1ca9c3b79"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
