)]}'
{"id":"openstack%2Frequirements~824934","triplet_id":"openstack%2Frequirements~master~I71b39c6019253df44793a87ccff078ff9f92b920","project":"openstack/requirements","branch":"master","topic":"validations-libs/1.5.1","attention_set":{},"removed_from_attention_set":{"32926":{"account":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"last_update":"2022-01-27 07:05:39.000000000","reason":"Removed by jpodivin by clicking the attention icon"}},"hashtags":[],"change_id":"I71b39c6019253df44793a87ccff078ff9f92b920","subject":"Bumping validations-libs to 1.5.1","status":"ABANDONED","created":"2022-01-17 12:48:51.000000000","updated":"2022-01-28 05:51:29.000000000","total_comment_count":1,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":false,"meta_rev_id":"0f345e915c1ff63858f4b44e98d65d9d6f9811bb","_number":824934,"virtual_id_number":824934,"owner":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-01-17 14:11:51.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"all":[{"value":-1,"date":"2022-01-17 20:26:34.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}]},"pending_reviewers":{"REVIEWER":[{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}]},"reviewer_updates":[{"updated":"2022-01-17 13:01:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-17 14:11:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-17 20:26:34.000000000","updated_by":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"reviewer":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"state":"REVIEWER"},{"updated":"2022-01-22 19:31:46.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"}],"messages":[{"id":"97d285ec610be924b819bfc04f6d13930039f82b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2022-01-17 12:48:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"415666b39c4ded459a8bff1a889554c30b981412","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-17 13:01:07.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/fadc0779a3ff4283907ee2b6dc5f5b71 : FAILURE in 10m 02s","accounts_in_message":[],"_revision_number":1},{"id":"df3b06a6b83e98f9b19307a547426442296bad7a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-17 14:11:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36a6899f2ce94e1592ad1fc6b683f7c0 : SUCCESS in 5m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/556cccb385304f9987211b02f4674c20 : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cd36cd68682b4cb89e2626858abc49f5 : SUCCESS in 5m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/86108e683b084308874eb6a65b15c321 : SUCCESS in 5m 38s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/292484d8b01f4223aa3e78b6b7d709ec : SUCCESS in 4m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1ec60a5fcc534675b29934788946780f : SUCCESS in 4m 39s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/852ed27f44f24d2aa7338dd56e6177c7 : FAILURE in 5m 00s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9b8851044b114d2692018388cca0d723 : FAILURE in 5m 26s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/03db517b690048c3a790d99ac76495c9 : SUCCESS in 6m 02s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/a92033be5e8a45d288db7192f8ba8b5d : SUCCESS in 11m 49s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/62e6ff81a7de4abbb6b019dac08cd32f : SUCCESS in 7m 27s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/164c808f512c4f019cda544d9b9352f1 : SUCCESS in 11m 32s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f192ea8d725e4a47930e64a33ed49a13 : SUCCESS in 7m 53s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/33aaadec15714f1bab089fdba7f5da67 : SUCCESS in 8m 36s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/25f35945fcb24e15b2728f5784cd8bd1 : SUCCESS in 14m 43s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/4813c68b1082400cb641239f975ed9df : SUCCESS in 5m 15s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/581e122f890a44b2bf3b6d8da2cccee4 : SUCCESS in 17m 18s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/6e5b0f1680b24cf8a1fa9e942dc63b6c : SUCCESS in 5m 55s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/04c32d6c61694885a4eb165455ec9a73 : SUCCESS in 22m 18s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/8dc91a7ec22b4911b13c9ff6e16ee721 : SUCCESS in 15m 52s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/175afdc7f4244198b236fe2f695e19ff : SUCCESS in 6m 55s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6dd6176a2cdb49c9ac0710ca602ac15e : SUCCESS in 4m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5d4ac4a9cb364f2090f7eb03f43c80d0 : SUCCESS in 18m 46s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/c88c825c294b47ef95facd7b50d7884b : SUCCESS in 6m 06s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/f23d066670ed4f948ee31926643227dc : SUCCESS in 5m 48s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/80b827bd73334c2db34d21ae7e245387 : SUCCESS in 7m 01s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/9c535cbdcd6a4faba7d23e7556855d92 : SUCCESS in 13m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b2382d7c2531459d8d8dc2120b74c151 : SUCCESS in 47m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/24ea27e723a3495cbafa3f886b2ecc50 : SUCCESS in 1h 19m 43s","accounts_in_message":[],"_revision_number":1},{"id":"03d32b0f48a0b3be4e6edd1a36806098b75046cc","author":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"date":"2022-01-17 20:26:34.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e61d2309e19ceac21b97874f5b5bc5ca22c85f38","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-01-22 19:31:46.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4b61952a36169b77140ceaf45c22692984f68c76","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-22 19:42:58.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/cf49ace1a80a48ee8b7376626b1620c8 : FAILURE in 7m 02s","accounts_in_message":[],"_revision_number":1},{"id":"5b6b95a3cefeaafc7a06a1a7fdf5c9ed04d74069","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-22 21:01:31.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91b886ca28384cecb1ad6087af5db75f : SUCCESS in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bdb7e1cf05fd465a8a2b492c3f4eb15f : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bde772e22eb04189ac59ac7f34c996c5 : SUCCESS in 4m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/774f82b29baf4a9495499737a71c8a6f : SUCCESS in 4m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/976e0e754be84ea0aeb8b3cab55102ac : SUCCESS in 4m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4b6c10af509649b2a52650d9ce3628b6 : SUCCESS in 3m 35s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a8d89fb144134b0aaa3bd01b3200e091 : FAILURE in 4m 29s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/208fab4d7d59426487e13838b2b912e5 : FAILURE in 3m 57s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/b71e719ac4a74b2ba76e74aaa24dbdd2 : SUCCESS in 5m 52s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/cd78c57f671a4eaf86e928999885026c : SUCCESS in 13m 50s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/3a839a6d85d14912bce5fdb5b6fb09b6 : SUCCESS in 6m 21s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/2ee6e6abc9c5499e8c925ff5046af866 : SUCCESS in 10m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/73c7bdb972bb42ebb448d02c2255bac4 : SUCCESS in 8m 36s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/a2a0a217da8e4ea79a9c12fa4cae045c : SUCCESS in 8m 18s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/08373aa0234f4621b84f12d41bec7793 : SUCCESS in 16m 13s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/d9bebb951ba04e8bb113472bb83f4530 : SUCCESS in 4m 22s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/2c2388feb0e84224b1b57334acbfaa2f : SUCCESS in 18m 08s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/7553b0d9754e4eb9948e9933b42b61e0 : SUCCESS in 5m 29s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/431c2bdaf54a40c69f2f7146cab70a10 : SUCCESS in 22m 20s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/700168406c8243a99b4c56ee1f1a92c1 : SUCCESS in 11m 59s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/57f35ed96d8f48d5b1ab48c9dc288e47 : SUCCESS in 5m 30s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/786314ef4dd84b58bc5a1ed88feef3d8 : SUCCESS in 3m 36s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f4acc05d5e0e4a259393a75470eb6400 : SUCCESS in 20m 20s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/ed7bff31c1954c49bc3af64810de121d : SUCCESS in 6m 09s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/a98e95f4f8f74cc79c6881867bc6764b : SUCCESS in 5m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e05a1e212cc14aea880e4f7ae8a683e5 : SUCCESS in 6m 35s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/753defc9ce57482abc6c53b592be30b6 : SUCCESS in 12m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6de5a639052349d8930a742463cf960a : SUCCESS in 46m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5609b1eda7e245f08bca4c2b5b946c7d : SUCCESS in 1h 23m 44s","accounts_in_message":[],"_revision_number":1},{"id":"0f345e915c1ff63858f4b44e98d65d9d6f9811bb","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-01-28 05:51:29.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0d00b3ed40b4943a98a402b9e9f4cfe07d66cf4e","revisions":{"0d00b3ed40b4943a98a402b9e9f4cfe07d66cf4e":{"kind":"REWORK","_number":1,"created":"2022-01-17 12:48:51.000000000","uploader":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"ref":"refs/changes/34/824934/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/34/824934/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/824934/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/824934/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/824934/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/34/824934/1"}}},"commit":{"parents":[{"commit":"adcb99ae08cafabe11a5d0be08621c56180c322c","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/adcb99ae08cafabe11a5d0be08621c56180c322c"}]}],"author":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2022-01-17 12:48:30.000000000","tz":60},"committer":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2022-01-17 12:48:30.000000000","tz":60},"subject":"Bumping validations-libs to 1.5.1","message":"Bumping validations-libs to 1.5.1\n\nSigned-off-by: Jiri Podivin \u003cjpodivin@redhat.com\u003e\nChange-Id: I71b39c6019253df44793a87ccff078ff9f92b920\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0d00b3ed40b4943a98a402b9e9f4cfe07d66cf4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0d00b3ed40b4943a98a402b9e9f4cfe07d66cf4e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
