)]}'
{"id":"openstack%2Frequirements~824946","triplet_id":"openstack%2Frequirements~master~Ica056e7c54026fd7ab389d52c46565223216dce3","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ica056e7c54026fd7ab389d52c46565223216dce3","subject":"update constraint for debtcollector to new release 2.4.0","status":"MERGED","created":"2022-01-17 14:36:01.000000000","updated":"2022-01-18 04:31:51.000000000","submitted":"2022-01-18 04:29:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"824946-new-release","meta_rev_id":"8e774fe38bd16ec8c603323bbc03ed10b1f94c18","_number":824946,"virtual_id_number":824946,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-18 04:29:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-01-18 02:35:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-01-17 15:06:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-01-18 02:35:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-17 14:57:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-17 15:06:08.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-01-17 16:31:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-18 02:35:13.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"295f523b904d713a5657d6c5fc5e54442dc94064","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-01-17 14:36:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"90ca99251fb6d649bba45bc680efc4e2acf65f6e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-17 14:57:05.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/412a76fe741549d08e237fd37fb15b4f : FAILURE in 17m 52s","accounts_in_message":[],"_revision_number":1},{"id":"3ffa72774bcf0cd7a49a2ecd28e128ee3f8b3dc7","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-01-17 15:06:08.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2a5c2360aab9fdab8375cd1435133e737996fa0d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-17 16:31:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d820512d678146d795d58bbce66251c9 : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a409173ad444e009c971832d47bb8b7 : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8c60e3134d03415ebb0d4b6001ad52d9 : SUCCESS in 4m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9a92f15fe439435bbbd5d5dc23b2af63 : SUCCESS in 4m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5f60590965dc43f28ce974bd315bbb67 : SUCCESS in 4m 26s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0bc3e172425c41079f66d4e969264f76 : SUCCESS in 5m 23s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/225e261641f740858a2e1957c2c6f8a2 : SUCCESS in 11m 02s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9707089c52f8462b85727a0d6f48019c : SUCCESS in 11m 19s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/268a7df14c2a405c9d257148fe44889c : SUCCESS in 6m 38s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/ef69d7125619483787c8778ace246b5f : SUCCESS in 12m 16s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/dc2b94a0a2704202b839f4f0adf8d743 : SUCCESS in 7m 07s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/a816e45a5fc9451bbbe27f66378dace0 : SUCCESS in 11m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8632a694617a45efbd902a2f3eed2a1c : SUCCESS in 8m 37s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/0a1021f037a5410d8b46a25640879ac7 : SUCCESS in 10m 26s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/55c101354354414aaec3365f9b9856ac : SUCCESS in 16m 21s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/bbfb422c58594d6fb257388ee90bb3b2 : SUCCESS in 5m 00s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/e05c89536c4243e49209847f981d61a8 : SUCCESS in 17m 52s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/f441cdc0aa904e40bf4bcee3818c5e40 : SUCCESS in 5m 46s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/b0eb2e3b007842d9a64036471172dae1 : SUCCESS in 24m 52s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/50a23cf014a64559beae4bad37b7927a : SUCCESS in 12m 47s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/196c3b1d62c346c8ab0f4a742d087707 : SUCCESS in 6m 35s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a4e611bb19ae466d900dcb13c945f959 : SUCCESS in 4m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/41c61691559647e38547f038de3975f2 : SUCCESS in 21m 31s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/a73dd05cdcc34b0393b756db7bb438fa : SUCCESS in 6m 04s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/00a735ca47274fe7b4fb4f143dabaf2a : SUCCESS in 5m 34s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c083e8c337c04eca9739c308cb1adc7b : SUCCESS in 7m 30s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/cac1ec88e4de45f5abbf31e40238eef7 : SUCCESS in 14m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/19b28281a25541b4a7a8619ee376fd4e : SUCCESS in 52m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e194a831f8374967a39157035784b3d3 : SUCCESS in 1h 31m 07s","accounts_in_message":[],"_revision_number":1},{"id":"e0a70fb29b90f6138074e929057504f2563a1ab4","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-01-18 02:35:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a9c75ee53111ad89d41d5fd7aae5c15fafe1097a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-18 02:36:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"516f963c38034c19bd8b3896cc73483c8eae2547","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-18 04:29:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df25effb6a1d40d7a2afc5e22e6040b9 : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ef14c85cfe64fec97f40aa0454d4d92 : SUCCESS in 6m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0929890e2c2a46d3a15372e55920aab4 : SUCCESS in 4m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/90540a4b09bd4bc28defcec0a8e14618 : SUCCESS in 8m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/739a6c7a84d14ccfb29b4cfebf39d0b0 : SUCCESS in 5m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/11c0ba3e717341878ccc8a65b9d2f4a0 : SUCCESS in 7m 21s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5bc63b5f1ca3414f8c5e5d8d53672ca3 : SUCCESS in 14m 09s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6c766e5414514f7c80fd61e34535daa9 : SUCCESS in 14m 12s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/c5ee41b128044dd39fbadeee604b50ff : SUCCESS in 13m 09s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/28d3b5ef74874d6e957aa62b2d5cc25d : SUCCESS in 7m 25s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/f9f0875564cf48b5ad7dde4e080ce95f : SUCCESS in 15m 23s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f751a6eb299e44c9805905c42cecb35b : SUCCESS in 12m 54s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/4298cfe25a4b42e48f1f1a4f96e84033 : SUCCESS in 13m 43s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/642e119514a94fbab0809fe79b3a1a37 : SUCCESS in 18m 41s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/fa672ebc6ef0477d8d3d865d13bafa09 : SUCCESS in 8m 16s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/1b4d57effaa94c9bb9bf4922cce5c486 : SUCCESS in 19m 35s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/a084bfbda3bc408aa65d5b68dd31d145 : SUCCESS in 8m 54s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/79c07b7cada14866bcc98ef4c31a3272 : SUCCESS in 23m 51s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/0bab5fd035444b5397805e31489ab5bf : SUCCESS in 13m 05s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/23a6a99b6979425a811832ba14ce2ba0 : SUCCESS in 8m 49s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0a6d0caf10174860af0b1776d13fc1c4 : SUCCESS in 5m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4690fb7e09a046ed8ed9ae01b2dc9218 : SUCCESS in 23m 36s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/c789092917cc4fd1ac01ec250cef2cf6 : SUCCESS in 15m 08s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/80e1b7be1230419cac50baed09feebb1 : SUCCESS in 7m 36s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/501b13c7e9574cd2b209901302d4fda8 : SUCCESS in 8m 02s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/6be5c8c9de684085981a5c840cd71bd1 : SUCCESS in 16m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/915007b716724fc59811cc8f370d03b8 : SUCCESS in 48m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8d299df1f1364bcfaf69935a0146eb02 : SUCCESS in 1h 26m 42s","accounts_in_message":[],"_revision_number":1},{"id":"8765aaf2f123012f92559e6b9132d84eeef143f8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-18 04:29:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"8e774fe38bd16ec8c603323bbc03ed10b1f94c18","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-18 04:31:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7eba6662774c4239ad4272bd0837f1fc : SUCCESS in 1m 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c381bd8257b68da8400f85865dee0827d7a9b8eb","revisions":{"c381bd8257b68da8400f85865dee0827d7a9b8eb":{"kind":"REWORK","_number":1,"created":"2022-01-17 14:36:01.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/46/824946/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/46/824946/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/824946/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/824946/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/824946/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/46/824946/1"}}},"commit":{"parents":[{"commit":"adcb99ae08cafabe11a5d0be08621c56180c322c","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/adcb99ae08cafabe11a5d0be08621c56180c322c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-01-17 14:36:01.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-01-17 14:36:01.000000000","tz":0},"subject":"update constraint for debtcollector to new release 2.4.0","message":"update constraint for debtcollector to new release 2.4.0\n\nmeta: version: 2.4.0\nmeta: diff-start: -\nmeta: series: independent\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Commit: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nmeta: release:Change-Id: I252cb38bf17920ba5d9c06d305754b4e6287c8c7\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Ica056e7c54026fd7ab389d52c46565223216dce3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c381bd8257b68da8400f85865dee0827d7a9b8eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c381bd8257b68da8400f85865dee0827d7a9b8eb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
