)]}'
{"id":"openstack%2Frequirements~824960","triplet_id":"openstack%2Frequirements~stable%2Ftrain~Ibbffdd0ce12b1b0a39717b5a9b6ded189795ae6c","project":"openstack/requirements","branch":"stable/train","topic":"vlibs/1.5.0-stable/victoria-stable/ussuri-stable/train","attention_set":{},"removed_from_attention_set":{"11904":{"account":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"last_update":"2022-04-22 10:28:56.000000000","reason":"Change was submitted"},"14288":{"account":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"last_update":"2022-04-21 15:37:09.000000000","reason":"removed on reply"},"16515":{"account":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"last_update":"2022-04-22 10:28:56.000000000","reason":"Change was submitted"},"32926":{"account":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"last_update":"2022-02-09 13:51:40.000000000","reason":"Removed by jpodivin by clicking the attention icon"}},"hashtags":[],"change_id":"Ibbffdd0ce12b1b0a39717b5a9b6ded189795ae6c","subject":"[Train-only] Move upper-constraints for validations-libs to 1.6.0","status":"MERGED","created":"2022-01-17 20:38:47.000000000","updated":"2022-04-22 10:28:56.000000000","submitted":"2022-04-22 10:28:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"824960-vlibs/1.5.0-stable/victoria-stable/ussuri-stable/train","cherry_pick_of_change":824959,"cherry_pick_of_patch_set":2,"meta_rev_id":"7d85140e9e56d0ca62f3138ca173de7dc2261b20","_number":824960,"virtual_id_number":824960,"owner":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":23266,"name":"lixuehai","email":"li.xuehai@99cloud.net","username":"lixuehai"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-22 10:28:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-04-21 15:37:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2022-04-22 08:57:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23266,"name":"lixuehai","email":"li.xuehai@99cloud.net","username":"lixuehai"},{"value":2,"date":"2022-04-21 11:53:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-04-21 06:38:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-04-21 15:37:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":23266,"name":"lixuehai","email":"li.xuehai@99cloud.net","username":"lixuehai"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23266,"name":"lixuehai","email":"li.xuehai@99cloud.net","username":"lixuehai"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"}],"CC":[{"_account_id":22954,"name":"Juan Badia Payno","email":"jbadiapa@redhat.com","username":"jbadiapa"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-17 22:19:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-18 02:29:16.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2022-02-04 16:14:39.000000000","updated_by":{"_account_id":22954,"name":"Juan Badia Payno","email":"jbadiapa@redhat.com","username":"jbadiapa"},"reviewer":{"_account_id":22954,"name":"Juan Badia Payno","email":"jbadiapa@redhat.com","username":"jbadiapa"},"state":"CC"},{"updated":"2022-04-11 08:29:32.000000000","updated_by":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"reviewer":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"state":"REVIEWER"},{"updated":"2022-04-21 11:46:46.000000000","updated_by":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2022-04-21 11:46:46.000000000","updated_by":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2022-04-21 11:53:06.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-04-22 08:57:25.000000000","updated_by":{"_account_id":23266,"name":"lixuehai","email":"li.xuehai@99cloud.net","username":"lixuehai"},"reviewer":{"_account_id":23266,"name":"lixuehai","email":"li.xuehai@99cloud.net","username":"lixuehai"},"state":"REVIEWER"}],"messages":[{"id":"b39bbba122a0137b4d620fe20a27046425bd56e8","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"date":"2022-01-17 20:38:47.000000000","message":"Patch Set 1: Cherry Picked from branch stable/ussuri.\n\nThe following files contain Git conflicts:\n* upper-constraints.txt","accounts_in_message":[],"_revision_number":1},{"id":"79a6fb8692f0f8681acd7221b0c15c41368c91a3","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"date":"2022-01-17 20:39:16.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"0a9be2706d2f26747bcf0893163df6036bf23f6b","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"date":"2022-01-17 20:39:20.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"d8ae4d761919923b11642b77aa43a2ff9a99be7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-17 22:19:49.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/492b3d1cba064f0391e98e166565c0ae : SUCCESS in 4m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/281aee75964d450692f1f50fb522152b : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a13f8bd1231e46aba840d1febbb9de98 : SUCCESS in 4m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f1b17baf5e7f407a8bf3a762b0e7b716 : SUCCESS in 4m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/af1c36ab19d7417d850e666e5219503b : SUCCESS in 4m 42s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a0773ac1ed284589a6068bca13082ed3 : SUCCESS in 4m 03s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/3063327d5a1b41ecaf7f5ea10bfe6112 : FAILURE in 7m 55s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9b9eb351913b4a138c3e44002714332e : FAILURE in 5m 14s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/c8c2bc8660c34e0bb9965ce3c6c6fd42 : SUCCESS in 9m 39s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d9446efb313b4be0ad2550cf4cdaaa2d : SUCCESS in 9m 19s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/b167253d8dd44c00b8beab45f84e7f02 : SUCCESS in 5m 14s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/68dd8b08b3904772aa10ebed3f52dbcd : SUCCESS in 4m 58s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/b5cf8aafb1b44a2cab7e2ce3187d949d : SUCCESS in 11m 49s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d81a336b230548c08ea9ee4f21792589 : SUCCESS in 9m 32s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/16c7b168a6854442a9427e3e04c56c26 : FAILURE in 11m 51s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/6f228c8087c745a3b78986bc73bb4a6a : SUCCESS in 13m 37s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/3856104f498f4719a383d5c1a24bf0a4 : SUCCESS in 13m 10s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/296278a42d5f4344883b1575abc81cab : SUCCESS in 23m 34s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/35b1f3997fe140a99179359b22028568 : SUCCESS in 18m 51s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/69a08c060f414e45bc94649a7f16b379 : SUCCESS in 13m 23s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/a55199a6eec44a65b904c490a2579cb9 : SUCCESS in 12m 48s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/9a060e3a1735462d810f8e9faf57d24a : SUCCESS in 13m 59s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/174d6439cf18442c84c2a9897d4776e9 : SUCCESS in 24m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c98ef1ba4bc34dc980b1f0a313ba9747 : SUCCESS in 5m 42s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/8a73fae02de2435996aa557c3e2715e0 : SUCCESS in 9m 47s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/6d38bbd2e16d40e2a0b50906c4a689e1 : SUCCESS in 10m 42s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/328dbc3b6dd54e72870e551aeeaee6d0 : SUCCESS in 1h 26m 55s\n- tempest-full https://zuul.opendev.org/t/openstack/build/2955001f1748474da65444c7e126a61d : SUCCESS in 1h 31m 51s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/54ed5636fe444a32815f853d023cf917 : SUCCESS in 1h 25m 33s","accounts_in_message":[],"_revision_number":2},{"id":"406b52890126c9eba5205494a9ca29a9ca89b25a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-01-18 02:29:16.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"e5c41b6a4eb25f51a243fd1887efb53e2edc1429","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-18 04:28:34.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13f59e825ed04736bfb1759bbc9a891f : SUCCESS in 4m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/148b854132e648708a42cc1fc0d307ea : SUCCESS in 6m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/667335db8b0c4e84904517a73fa0b4e1 : SUCCESS in 5m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8a5687c099154f79aa6b3b8cfcf80d26 : SUCCESS in 5m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c6b03fd371104e348df80d5ed3c95098 : SUCCESS in 5m 40s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6ac350ff665e4eb388f61d9fcf108488 : SUCCESS in 5m 36s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/c3ce70e64f0f4f48bd88c8e30b509df5 : FAILURE in 10m 14s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2d6bf1a768b74d12b30dc930f3c85e32 : FAILURE in 6m 29s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/42ad363d75fd417a86b0f771850f7dcf : SUCCESS in 11m 36s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/49a77ebd8d4a4ebea27028c9c799faad : SUCCESS in 11m 29s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/ac25af3d61f24f539010219b78f50d3d : SUCCESS in 6m 43s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/c2fb178754f64f5f8b153aad14bb84fc : SUCCESS in 6m 14s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/e0a0e6b599d04cf1b25d5b28d96fd9d4 : SUCCESS in 12m 13s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/18e13a10c0a5472883fd04ed00953494 : SUCCESS in 29m 04s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c24f0e40d59d46db9e6bbf7b9ee74d95 : SUCCESS in 14m 29s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/d8c888cfab7c4bc590a82657fab1880c : SUCCESS in 17m 45s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/faa23a9c4fc54026b4d449e89a74de64 : SUCCESS in 16m 36s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/96f05e3757e34d7581fc4ef44a79cc57 : SUCCESS in 21m 20s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/062c6cffc0e343be877a7f6d0d69756d : SUCCESS in 21m 20s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/526b773a1c6247e383d159e6ea02222b : SUCCESS in 15m 11s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/02f017cbef034e7597cf990602b12314 : SUCCESS in 14m 54s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/77c5f14528634fe3a2596c7f9d6d39f2 : SUCCESS in 16m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5f35dffcf7324e71be39db836c99ed23 : SUCCESS in 22m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0b1eb9e71161428a841a44a7d57e389f : SUCCESS in 7m 38s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/3757df6142a044d0a408f0e6f009ac26 : SUCCESS in 10m 34s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/03804b5d47a0469597d868757efbae12 : SUCCESS in 12m 21s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ff981c2db8ae41d5b63a2471b833c7a7 : SUCCESS in 1h 18m 51s\n- tempest-full https://zuul.opendev.org/t/openstack/build/e7bfd23d387a4bd3bc704732715e1b04 : SUCCESS in 1h 38m 57s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/e686e26b546b478c9ccf275ff9785ca6 : SUCCESS in 1h 26m 23s","accounts_in_message":[],"_revision_number":2},{"id":"70c75216280e52585738916bbdce559245564ccd","author":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"date":"2022-01-19 07:27:14.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"90becc9cc89e61e0a7c650512379f8b35d12035e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-19 09:24:38.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d607012c2364e5e81deb2c5064e9893 : SUCCESS in 3m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/356385cddcff4ea7b3ce01d2293d7008 : SUCCESS in 3m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/abb3f6b05ba842de99e1f0f64173c60b : SUCCESS in 5m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/93710ac0418a4a9cb4a0423e3252b6de : SUCCESS in 3m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0d6457c385b74816955001a0d4be922c : SUCCESS in 3m 43s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/24cb4cdf287e4205b429bf1729cff6ac : SUCCESS in 3m 43s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/b487c7bada924fa9be626dc5a8e64787 : FAILURE in 8m 38s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5210c53cd62d42378669cf1f417b7227 : FAILURE in 5m 06s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/7eff1b80d1ef444aa444d4bca7143279 : SUCCESS in 9m 26s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/54955aab83934bed979a4c0a507fb62f : SUCCESS in 9m 53s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/4745c95e55cb4638ad644e76a72c38f6 : SUCCESS in 4m 20s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/18061be72ff94c2691c45b85aa6d8771 : SUCCESS in 5m 09s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/b36d2a1236c941b8a5c4e8321952635f : SUCCESS in 9m 10s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/7624cc4497e9447682d7a3c01aecd333 : SUCCESS in 10m 21s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a0863e0e07de42c4bbac4948def45d5e : FAILURE in 12m 21s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/b36a060860e648c4afb8b965c8deba4f : SUCCESS in 13m 59s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/ffc557d02d6b41d6a3facc9ff30135a2 : SUCCESS in 42m 27s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/fefe0974389c4466844e511288748d5f : SUCCESS in 25m 13s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a55c5513587c43a0881ce233e633e50d : SUCCESS in 28m 06s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/8691237fb7ca49699091159aba5e0c4c : SUCCESS in 19m 56s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/10e93164832e42b0a3c4dc5e8c78162c : SUCCESS in 11m 01s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/0ad603793b2747f5a2e4cdaa5f0217a7 : SUCCESS in 14m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d97a7e4e831640e7a2b4710c1e0bfc7b : SUCCESS in 17m 52s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/cf7394105a82420886df8829f97c122d : SUCCESS in 5m 48s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/d45b3a81a9bf407e8506c71dab728e9e : SUCCESS in 10m 56s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/f2499c1b5e5043d69f564312c81fb1c1 : SUCCESS in 12m 04s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/23fc5d4f76b940d682890eafdc3cff09 : SUCCESS in 1h 03m 41s\n- tempest-full https://zuul.opendev.org/t/openstack/build/04ce18a6f3004c29a75345fe551bbc56 : SUCCESS in 1h 53m 53s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/85fad8e1b2fd4f09908378df7e9530af : SUCCESS in 1h 08m 00s","accounts_in_message":[],"_revision_number":2},{"id":"3d110e93b7942853c655dff7cc610b5b7c0c68d8","author":{"_account_id":22954,"name":"Juan Badia Payno","email":"jbadiapa@redhat.com","username":"jbadiapa"},"date":"2022-02-04 16:14:39.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f9ebd712745fc1dd4f85bead419699b1c1927107","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 17:49:08.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fdaac72a471b4b96b3a1ed21d7e554e3 : SUCCESS in 4m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/22f26400d99049409537871016c6651b : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c0fcd29097cf4700b08d51e8eabd2c62 : SUCCESS in 4m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f2cf1e5b627a4a0f947ab9d96d60b112 : SUCCESS in 5m 38s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e3a1b1948e5948b5a55486268d98ba1e : SUCCESS in 5m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/50cbec0bf27b407aa54a3c200a55b11c : SUCCESS in 4m 16s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/bdd0c2dd6e44469381dcdc3846c1ecbc : FAILURE in 10m 15s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/37249b648cdb47a5be0c1c8bd5f21c85 : FAILURE in 7m 41s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/1bf0d414ac02439189a4b5b8748807e0 : SUCCESS in 10m 18s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e677c758c966419c98630c3596382b9e : SUCCESS in 10m 50s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/985aa956f8344cfe8048f29bb190165e : SUCCESS in 6m 41s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/2d5b00cf02184f0f8d3ad00fd2e147be : SUCCESS in 6m 52s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/76d25f0269ef4fc68d942d0b03049359 : SUCCESS in 16m 05s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d7c892a888e3482bb6d004c6bc893c97 : SUCCESS in 10m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6b5496807d3e460fb5ee4c656f82aeeb : SUCCESS in 12m 31s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/e214fe31ddb348b0a1c5b8e95b9bf9d6 : SUCCESS in 36m 38s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/10a315c3a6f148d1a9dfd59c81722675 : SUCCESS in 34m 22s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/a850e8ca73f54358b1da6487e9180829 : SUCCESS in 25m 32s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d238e8886a164927b1f68774ccefb979 : SUCCESS in 25m 23s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/d356ee1187e64bfca0652bfcea9b2879 : SUCCESS in 19m 30s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/2746d2c5b45a4334acb4984bf22b1092 : SUCCESS in 14m 46s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/ac03888608564468987b79e5ed92952d : SUCCESS in 16m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6157bde8ef3a4075b64c9ddc892bf7af : SUCCESS in 18m 40s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/21ebe672af2340598b5ec36b049c2039 : SUCCESS in 5m 34s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/cf798e1a645946ce83d4e47655b9c973 : SUCCESS in 12m 13s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/9b88416141494a3ba1697f2d8cb14580 : SUCCESS in 12m 12s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f51ad5f4c09b4e538bfe5a16eebec7df : FAILURE in 21m 09s\n- tempest-full https://zuul.opendev.org/t/openstack/build/333512c28a1e44f7a5e47bed7b20e0c5 : SUCCESS in 1h 21m 43s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/d423cbded6b3427faacb5940b443884f : SUCCESS in 1h 01m 41s","accounts_in_message":[],"_revision_number":2},{"id":"3b26c3c52bbb2d28b738b0aba599b88098c25d3f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2022-02-09 08:18:15.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"976dfbb4f6fd96ac470dd1acba5112d7a3171205","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 09:38:57.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d49b621b66b94e1387c6410c7766f808 : SUCCESS in 4m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a5957f33bf2e41adb238fb9d943d1783 : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/18fafc703aec4a25b3caf2a11240262f : SUCCESS in 4m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/324026959ad44fe3b583167f81efcd2c : SUCCESS in 4m 50s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0393dce0fd474cd9a290bacd254d076e : SUCCESS in 4m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6489d8f443aa4bffbe993c21b47cdb07 : SUCCESS in 4m 35s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/995b6acc64fc483cba1217233c74d340 : FAILURE in 8m 19s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9a59a35d0759499f9cf8f65d8f195db7 : FAILURE in 5m 40s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/df044ff2ee9e44edab2339c897eb173a : SUCCESS in 11m 09s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/9b242af96bc7438292fefb7b81f5a3c4 : SUCCESS in 9m 56s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/a238d206d0bf4b13b67915534ec57e87 : SUCCESS in 5m 10s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/c5470f6bdb1a450fad1e9d6439b9f3b9 : SUCCESS in 6m 22s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/dcdfe5c1d979472bbbc042531beb69ac : SUCCESS in 10m 28s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/a6abe1e4cd7d4eeda5a59cf10c5b3c32 : SUCCESS in 9m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1599d84498664883ba8c88d928a7b71e : SUCCESS in 11m 05s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/88cd1257e9c9462eb876f55ee3b2517f : SUCCESS in 12m 57s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/574443c907b94bf8a17aff965035e054 : SUCCESS in 12m 12s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/9fdeadb654a1432c9ad7293d7b486987 : SUCCESS in 22m 06s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/6aa7aa5ddb8f4c7fab04a7cfdbe886e5 : SUCCESS in 20m 16s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/a6126c89731646fea691d68fdffa340f : SUCCESS in 14m 29s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/7e71a57eb6f9417fa43c72b1ab4f54e9 : SUCCESS in 15m 49s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/caaa55b065cd45528b14a002db1913f7 : SUCCESS in 16m 45s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/64fc50c730a344d08b54ddbcd8d8cf31 : SUCCESS in 22m 43s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/09df01063961481f865c36688ad87c39 : SUCCESS in 6m 18s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/42615d6d1337467791908566bd1d6db8 : SUCCESS in 10m 31s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/626b0a948edd4aafb7b0bc4cc0dc3323 : SUCCESS in 12m 33s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/edd7d125f48e408d841d3b03741702b6 : FAILURE in 27m 09s\n- tempest-full https://zuul.opendev.org/t/openstack/build/57cf07cd14714867ab48d4c7a8429d53 : SUCCESS in 1h 13m 55s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/b1902d0939c54d53a5f4215efdcebe3d : SUCCESS in 55m 31s","accounts_in_message":[],"_revision_number":3},{"id":"cb033da609603bee9103e0ce03b4736d6b1371bc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"date":"2022-04-07 13:50:57.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"7e02326452f91f82f0f94c801ca06d45c0f18fc0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"date":"2022-04-07 13:51:34.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"1d8294b545401db01734b4c2a05a27e6ea155e65","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-07 15:15:19.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5673655844924333a124fe148f20e5e5 : SUCCESS in 3m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5b4ec32eee0d4a1897dc57944b4f31f2 : SUCCESS in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e1bb30de473e4fd89f821518cd9b6c2a : SUCCESS in 4m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/79420dfeb62d41039278e76af17bcbb4 : SUCCESS in 4m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fdb47a1ee2ad44c19706acba67f77ea8 : SUCCESS in 3m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/15970eea058f476791ec7519515ccccc : SUCCESS in 3m 54s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/a968b4f65cf14da3a8ca425ac1adf73d : SUCCESS in 9m 16s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9b3c78e8c2de409193da3dfdfaf66b77 : SUCCESS in 10m 23s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/c139661f9a944e05b0b46dd4fdcf34b6 : SUCCESS in 10m 54s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/2ab27986437842ae81f983c5638ae6c2 : SUCCESS in 11m 01s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/36e1ed11870c471080a78fbb5ef95e3a : SUCCESS in 5m 20s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4e0105984d224711b4370be81cbaf4e1 : SUCCESS in 5m 47s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/ed5518a0c02d4b65b8b18e81046b60d9 : SUCCESS in 10m 43s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d50abe31148d4edb8b6f8278f15ed56b : SUCCESS in 8m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/40ea89be8b8e44929a8febfb39188a3b : SUCCESS in 9m 13s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/1981ad174d1649a8998b3c801589e385 : SUCCESS in 14m 45s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/f66b80d8572c44f68e537f7ae9addc40 : SUCCESS in 13m 11s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/0f94197d86934a08b1db408c122c276b : SUCCESS in 21m 03s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/8c04288199414ff09d4cc7b748bc5204 : SUCCESS in 17m 32s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/eafc6e1c234a433eb37981658c07bfb3 : SUCCESS in 15m 52s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/7f8c663420ea448db1950520c45e680a : SUCCESS in 12m 53s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/192c1cb42fb6482d89770bac7ff7e259 : SUCCESS in 12m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c146fafe7bf84c9981b8f6bdff089cf6 : SUCCESS in 19m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3e6c3d2c5c5f4a45921f950053123a09 : SUCCESS in 5m 27s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/f82126d8ec364cda9e9351c416d6909a : FAILURE in 10m 21s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/039e8f2885924e8cac5d5355660ff34d : FAILURE in 10m 08s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a05cdbd73a8049079d5de8238a714baf : SUCCESS in 47m 10s\n- tempest-full https://zuul.opendev.org/t/openstack/build/efe0ae68f111464eb84be73b2e8d0e2a : SUCCESS in 1h 21m 27s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/c52529db2f074351a1d8460ed65a7b1f : SUCCESS in 52m 18s","accounts_in_message":[],"_revision_number":5},{"id":"b4f5f84676658f5576d1f7a634565847fc0d1a14","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2022-04-11 08:29:32.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"f13d63756a1f560cf6b69325541049335e4a708b","author":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"date":"2022-04-11 08:40:05.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"a9ef8472eea2cde83834f0a964c96e2df8d0fc57","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 10:21:13.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30470d6b6bde499db766e19537909cea : SUCCESS in 4m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a5f122b4a27b4946bc8c4ed073d47b81 : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/070736d61f3d4583802d5994f31fc68c : SUCCESS in 4m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bf1ba5f675a2411abfee323b430ca624 : SUCCESS in 4m 45s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bdc5a7d6faa94d17a94eb79bae79c293 : SUCCESS in 4m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/99fb092526da494686120f8b5141c05e : SUCCESS in 3m 26s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/d16d4013c62d4b11b45d60c2d4eab77c : SUCCESS in 8m 32s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/718e43d84ba34faf91e64caeef38cc91 : SUCCESS in 9m 47s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/37b9056670fd487ea3a6aecb1c82fb92 : SUCCESS in 9m 46s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/5b0d64908ef1422ead0ce8459b59473f : SUCCESS in 10m 35s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/dd3bd7d81cbe473c99df8d9913764ffb : SUCCESS in 5m 19s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/c8728d16c0e94eeababd11a33c025ed1 : SUCCESS in 5m 05s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/8375d21022a04f94864575b13b9a19c6 : SUCCESS in 11m 28s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/2879a20eaad04d05be4d49c85dcae03a : SUCCESS in 10m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4b1588c0055e4d52a87ceacce0f828c9 : SUCCESS in 9m 07s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/5baacb9fca7648dea1ebc34060776380 : SUCCESS in 17m 13s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/d342a60f19fd4029a523389d206733dd : SUCCESS in 14m 23s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/9ea5dfe88d7f4e50a73fb01e2e89780e : SUCCESS in 26m 06s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/2348ae20f35348b7a1874947c19047a3 : SUCCESS in 19m 26s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/720b2f2fe62146cbbb7ba8e341199f9d : SUCCESS in 13m 55s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/54076f508ec1415c839568a936b129e7 : SUCCESS in 15m 02s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/017b2cf5133e40c3aa9442a250cf95cb : SUCCESS in 15m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e3a28dc9ae79418d8dee9373701f34ea : SUCCESS in 17m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b883f98644284789bef9f917d23c8037 : SUCCESS in 5m 04s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/bcd5874bcb1249eb979132399b515f9a : FAILURE in 11m 41s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/0066fc4684ff4235877b80a70f2e9f53 : FAILURE in 11m 40s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/fdb480cdb74442b7ad3e7ff4df0db9f2 : SUCCESS in 1h 07m 44s\n- tempest-full https://zuul.opendev.org/t/openstack/build/3e73eba45c9d47b2919db4c86bdcf279 : SUCCESS in 1h 35m 37s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/dbee63f01b2f4ea286736ed759e58007 : SUCCESS in 57m 20s","accounts_in_message":[],"_revision_number":5},{"id":"c3ab370695b29dc85746c81f0bb5387d91ce8345","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"date":"2022-04-11 13:31:28.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"8a1c7881ee718427218a07acc2672a6fccd90dd2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 15:04:02.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23a2af3677d640c5a699999488102ef9 : SUCCESS in 4m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5060cad9a90e4b3f83539f894f6fb1f0 : SUCCESS in 4m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5fed54653bfd4bbeba8cdb53c6a7e46b : SUCCESS in 4m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a60338bce31f4be2bf7f66b47ed4ab7d : SUCCESS in 5m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a9d8102f7ca348b3986396c8af9d6226 : SUCCESS in 4m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/129a46aadfa94654a95781f119d6708f : SUCCESS in 3m 51s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/702b896d1d054fcbb66e4855882bd6c4 : SUCCESS in 10m 19s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e6a5d8fc8f8945b2801770ff2b475e57 : SUCCESS in 10m 39s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/06a7751960804b8390c54acf1f6f00e1 : SUCCESS in 10m 30s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/6990eb57e4254214b487354d2a33d248 : SUCCESS in 10m 26s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/15199218907b495986b0824208b10c3d : POST_FAILURE in 5m 36s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/5c044958fc954a04ba40656e19659d3e : SUCCESS in 5m 15s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/ffc055d68b61459697776c96ef12cd45 : SUCCESS in 9m 50s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/5784c2da1d06413ea9492397414783be : SUCCESS in 8m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e48f66fb2f3f40f2a86a203aa0fe01ae : SUCCESS in 8m 50s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/7b9988c21a964f5b8c3e3f757c696a98 : SUCCESS in 14m 18s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/da1a8a6e11d0472b87ea07283a3fc3d5 : SUCCESS in 28m 15s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/aec2ec9f803c463a9cba03bab6fdd5ed : SUCCESS in 20m 06s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/059c9d625ebd4a0e9996eebd11b3a986 : SUCCESS in 18m 20s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/b4eb991442fd4f7da870ad5e17c84f27 : SUCCESS in 14m 15s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/e750a95cda2444cda379df6919105f05 : SUCCESS in 15m 32s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/d24d68a1b7834774b10881df492dcef8 : SUCCESS in 13m 44s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fc2283158e254a8e851f3beb3f1a1a92 : SUCCESS in 23m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/01467f4b00bf406b94ad7b9888054d63 : SUCCESS in 5m 58s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/dc73708d5d88450ea1c227dac21cf63b : FAILURE in 10m 47s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/b7053e6065534571bf53da8f1538e0a9 : FAILURE in 10m 38s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/24ced6fdfe034ce39c9412b2173ae6bd : SUCCESS in 1h 07m 24s\n- tempest-full https://zuul.opendev.org/t/openstack/build/394e06c6f33648348e012d050029a8e2 : SUCCESS in 1h 26m 31s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/623b398c4859456da547dac45fa62b60 : SUCCESS in 1h 11m 55s","accounts_in_message":[],"_revision_number":6},{"id":"9e4bd507df4c40bbcbe23347cb109cbc38907e07","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2022-04-21 06:38:32.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"917bb0517483c6496556ff50b1435172d9abc23a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-21 08:22:43.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02c14269c5c14f3eaa777cde84844232 : SUCCESS in 3m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d16df9542cba4effba7a014ecdb22c0d : SUCCESS in 3m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4e989353cfd948a8a82df48a4e1a9a03 : SUCCESS in 4m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8b6cdcf84ea845ad881a90cd748680fd : SUCCESS in 4m 34s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/db0d9380715c484aabe29e78d9c4208c : SUCCESS in 4m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/081dc6846e5243ecb6201a0069d0c778 : SUCCESS in 3m 15s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/c7c61a6f0b944eb6a55a64dd5c927799 : SUCCESS in 6m 58s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b90c5118906c4efcaeca611bf1fa74f7 : SUCCESS in 8m 16s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/0378614c11f243efbc3465848b6e10a9 : SUCCESS in 10m 21s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/6dfb4a430b97414aaf4fa5b314b879d4 : SUCCESS in 9m 03s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/2325d84a114d4f8f9f24251a52d46822 : SUCCESS in 4m 59s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/b26cdf1524184f08b8a8d764c34fc7ff : SUCCESS in 4m 49s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/7483734f063c47779fb9ab7d65cd138e : SUCCESS in 10m 10s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/6b0cb24a3761449ea2af5c3d741521ec : SUCCESS in 10m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c6837a073e5742f48f3512cf2007b799 : SUCCESS in 8m 41s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/825a891cbfe34902933bfd60eb15e73c : SUCCESS in 13m 25s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/61e83d3c62f24749b50737e3ae6ee299 : SUCCESS in 13m 58s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/84332f80bea943ff9a7f36c6bb7340b8 : SUCCESS in 23m 23s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/1f9a8f834a3e4695bf42de3dbd546bc4 : SUCCESS in 23m 54s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/9d3d04596e9f49e7a926997f341405da : SUCCESS in 13m 56s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ed94edb48dfc4859a092abaf3a30686b : SUCCESS in 14m 24s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/ec009595e3024c4ab4c17ed8958d7872 : SUCCESS in 13m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/688f1215d2854bff90e4ecd5bca46163 : SUCCESS in 19m 18s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7aa8eb792a814d73be55d949c6f07e30 : SUCCESS in 5m 10s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/78dc807feb6948b381c6ffbf7488d86f : SUCCESS in 9m 51s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/cf306b591ccf4dbfa42c5c9394ea38dc : SUCCESS in 9m 29s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0b6b9dd2602f402d8c9b3d7502b37f12 : SUCCESS in 1h 07m 02s\n- tempest-full https://zuul.opendev.org/t/openstack/build/c92aeea613304bcbbd7ca8bf03db9ed8 : SUCCESS in 1h 37m 07s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/5408927ade07460f9eaca2c4b36b40f3 : SUCCESS in 1h 08m 26s","accounts_in_message":[],"_revision_number":6},{"id":"97c623eab234faa0fd2689ead44faf595302745d","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-04-21 11:53:06.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"320f3aab4f4cb072a49176327af9260c9153cfbd","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-04-21 15:37:09.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"a026d212dfdf973d523176c74bc76c1423b35be2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-21 15:38:43.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"e5ada00dd08cb102cf5a8d0fa4c12831475869af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-21 17:17:22.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7565c26f05e44e59fe07f925babef35 : SUCCESS in 4m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c92fac08f23f43228b5068c5273cf065 : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4b4334231b4b48e9918f17e5d4f3dd5f : SUCCESS in 5m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7065078533e243e49f9f645769e5b8f7 : SUCCESS in 4m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1d1595e3b28642af816d7f15f73dc37e : SUCCESS in 4m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d532e5944e534f1596eed9042298e49c : SUCCESS in 4m 27s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/9e764efb452e41ab91eb70858feb4a34 : POST_FAILURE in 8m 36s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f82b619b938343c99dcd78aa59d48909 : SUCCESS in 8m 38s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/4eb2a413adbb4b92bce90e40bd43d625 : SUCCESS in 9m 54s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/0d48deb14e6640a7a2ae0b05c3157bbc : SUCCESS in 9m 55s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/9694a7de0c3d43369b26fed1e288dd8b : SUCCESS in 6m 20s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/5e59c4060240457c97c950b89f5a1eec : SUCCESS in 5m 11s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/a1145c18973b42af9f74f448b1fd1cd9 : SUCCESS in 11m 06s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/8b4c6d94954c41feaeb4e47e7b5a50c6 : SUCCESS in 9m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ec54f42c1fdb4763ae241c5a55a6c09e : SUCCESS in 9m 44s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/0e55e527608a425785fa047d691c8226 : SUCCESS in 39m 44s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/06becf25c3794bff90eb510d477af969 : SUCCESS in 39m 23s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/00f4169eb3694c2385f127a8b9255cbf : SUCCESS in 22m 21s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/278371a168114cdab6ed84b9c7e76da7 : SUCCESS in 27m 15s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/f27199bb522c4050a745ab1524da2745 : SUCCESS in 20m 41s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f4b05ab7392a482ba318879b14561c57 : SUCCESS in 19m 07s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/31e2c90b4a2b4a958b8bcf8db2414d9e : SUCCESS in 13m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/70f1464dd7b54b699586e2c6177eec7a : FAILURE in 22m 34s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/612ca574fe80403a889e18f46b158ffd : SUCCESS in 6m 52s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/6edc3995ebba4d89ab3836d31e0c05e4 : SUCCESS in 10m 13s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c17c039843b1402eb05a3f06908fd4f1 : SUCCESS in 10m 24s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b5eded66a46443038f2e58c88e28d788 : SUCCESS in 1h 00m 40s\n- tempest-full https://zuul.opendev.org/t/openstack/build/5da37a0eeab84a668f623c9a4d611ade : SUCCESS in 1h 25m 46s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/83d2d1d6860947f0a77a618fb4e2f3c6 : SUCCESS in 58m 35s","accounts_in_message":[],"_revision_number":6},{"id":"2ff754ceb2930c81214b93845c12cb4dbbda2fca","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2022-04-22 07:09:16.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"136b4dfcdcb7c1b50adfa71a7ce407abd7e04b14","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-22 08:40:52.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/175dea0c6fa94323864cb9be82046314 : SUCCESS in 3m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/63ba3966bb2f4ba9889063e9817532f3 : SUCCESS in 3m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0504b2ce269d45e68486ead8d0130c60 : SUCCESS in 3m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7aac2dc2878a4c1baab46194815acb50 : SUCCESS in 3m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/52664ade1832404295b5bca97788ae61 : SUCCESS in 3m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6f0b8ce42b824856930a53b96b9da50b : SUCCESS in 3m 04s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/0923d3035c3345b7a798cafeae892f8d : SUCCESS in 6m 58s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/82d8235da95842cebc13b105219afec7 : SUCCESS in 7m 32s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/0b304649162f44e696c5ae5e220e9aab : SUCCESS in 10m 07s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/64cd39c82e48464295755e9e3bde2ffe : SUCCESS in 10m 27s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/0c47776159e64c4cbb677716577c2261 : SUCCESS in 5m 03s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/8ec0d2226809411b83a218c31360c13e : SUCCESS in 5m 05s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/3976741b126244ae8f5b5c8f35d17b1a : SUCCESS in 10m 30s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/52cc64d862e4402e9dcda9056b0a89e2 : SUCCESS in 9m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6bf5fb0b54fa4b53b476496eb81d5b76 : SUCCESS in 8m 04s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/50a894aba02e4e8992831ec5bb65923f : SUCCESS in 12m 53s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a65d189d1c83433ca884511c52003dbd : SUCCESS in 14m 56s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/a4e347a104674a388030f99764f8957a : SUCCESS in 23m 05s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/4bca945352a14b77969472815b4d3901 : SUCCESS in 22m 46s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/914c06c8abac496c982b006764cdadc1 : SUCCESS in 12m 58s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/9f0190baa5824809891a2e14d5a8f13c : SUCCESS in 13m 17s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/3b38b8a5c68c4a7fb091adfc7a77637f : SUCCESS in 14m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/aff50d53241b4569a6c7e077292951e3 : SUCCESS in 21m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2a111fa423c24711bc44830b66e4311c : SUCCESS in 6m 12s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/1bfe899f40244d1baddebd3be16a1622 : SUCCESS in 10m 27s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/1c8ea9ba2d824ef28da6e3e2fc089704 : SUCCESS in 10m 39s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b021daa7f4bd460c9390f0ce375a0469 : SUCCESS in 1h 04m 31s\n- tempest-full https://zuul.opendev.org/t/openstack/build/10ff5b51ef4d411d837afbc036931f0f : SUCCESS in 1h 24m 40s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/c9f85559ea594896a3bce392b8ab2594 : SUCCESS in 1h 09m 27s","accounts_in_message":[],"_revision_number":6},{"id":"1a226b6aa8f094a516fcc155b7d276c6bea3bb26","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-22 08:42:21.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"86440eb8cd097fff8fdb08aab17b27e8579ac345","author":{"_account_id":23266,"name":"lixuehai","email":"li.xuehai@99cloud.net","username":"lixuehai"},"date":"2022-04-22 08:57:25.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"9db200e92fc272b91778631773f72568ce1d9ebe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-22 10:28:55.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc94aac1c55c4e158507685fc5a363e8 : SUCCESS in 4m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c49ebc5a5e414e26ab0eee569ed263c5 : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a15a76d7f93a4be9be78b494aedca19d : SUCCESS in 5m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a8abf238160848189f1aff1d6ad0168e : SUCCESS in 4m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/85151a5db6b54f6c8fa7ab46c0112316 : SUCCESS in 4m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/171f2806a5f94f7499569d04e2b6b2a2 : SUCCESS in 3m 57s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/a0421310fc784926bcd806c7c526f02d : SUCCESS in 9m 09s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/37aef5354b514b29acf60552a9eac82e : SUCCESS in 8m 53s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/121f88533b074546ae6dc167e8194d1f : SUCCESS in 10m 51s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/745c1d4b6f754237858a784733e2fb35 : SUCCESS in 11m 12s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/e8f999f7ddd8437290aee4b8c06b97cf : SUCCESS in 5m 12s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/71cd1e81c01946c89c465d1f51758d3b : SUCCESS in 5m 37s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/1df6d6ba5bb840628077938ba59d5ac7 : SUCCESS in 11m 40s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/4a2f8fd11cab43289563b2f86a4b10a0 : SUCCESS in 9m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e3488adb486c48deb1878b9013aa93e6 : SUCCESS in 8m 24s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/45ac419d99b545e6a738806d6a996624 : SUCCESS in 13m 43s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/7ceefab9656a4685aa54bbbcd03e6e9f : SUCCESS in 17m 11s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/e6238d9201fe44dda07d16fe0ec46644 : SUCCESS in 22m 00s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/6577da0d650c4df196eb80d0fb4c794d : SUCCESS in 21m 46s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/c650eb9e0b7540b7bb5d7b3c678b02ed : SUCCESS in 14m 19s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/d82a46bee62441609cdef394e4e3996b : SUCCESS in 12m 03s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/413497b9b1be40ef9ebd896579ed58fa : SUCCESS in 16m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1226dedd87434be7bb3fe8070ba44e3f : SUCCESS in 20m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/612b0b76c91b46afa872abb929c6ca04 : SUCCESS in 5m 51s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/36fe7b4de5574dba86dbb5abf5eb6170 : SUCCESS in 9m 58s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/13ef4bcb1c2d470e95f73c3ae2b6565f : SUCCESS in 10m 37s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1f4d5ace92074b26aa7c3415f42fead3 : SUCCESS in 1h 04m 24s\n- tempest-full https://zuul.opendev.org/t/openstack/build/30bfb355b4a04a2e943a9bd96b1ef943 : SUCCESS in 1h 40m 28s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/182e2f015e714081836524183aa45f7d : SUCCESS in 57m 51s","accounts_in_message":[],"_revision_number":6},{"id":"7d85140e9e56d0ca62f3138ca173de7dc2261b20","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-22 10:28:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"2165e48df6fed369b60dc70e5baeccddc32d3992","revisions":{"6647c66966412bd933b7092b2cea0870c85e40e7":{"kind":"REWORK","_number":1,"created":"2022-01-17 20:38:47.000000000","uploader":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"ref":"refs/changes/60/824960/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/60/824960/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/824960/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/824960/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/824960/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/60/824960/1"}}},"commit":{"parents":[{"commit":"e8b19cc35b8368d086a931202968748c23648a7f","subject":"Merge \"Cap setuptools to \u003c58.0.0\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e8b19cc35b8368d086a931202968748c23648a7f"}]}],"author":{"name":"matbu","email":"mat.bultel@gmail.com","date":"2022-01-17 20:34:11.000000000","tz":60},"committer":{"name":"mbu","email":"mat.bultel@gmail.com","date":"2022-01-17 20:38:47.000000000","tz":0},"subject":"Move upper-constraints for validations-libs to 1.5.0","message":"Move upper-constraints for validations-libs to 1.5.0\n\nChange-Id: Ibbffdd0ce12b1b0a39717b5a9b6ded189795ae6c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6647c66966412bd933b7092b2cea0870c85e40e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6647c66966412bd933b7092b2cea0870c85e40e7"}]},"branch":"refs/heads/stable/train"},"c7a133dd88dd55e58df13cf804e17f2b6afc7fbc":{"kind":"REWORK","_number":2,"created":"2022-01-17 20:39:16.000000000","uploader":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"ref":"refs/changes/60/824960/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/60/824960/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/824960/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/824960/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/824960/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/60/824960/2"}}},"commit":{"parents":[{"commit":"e8b19cc35b8368d086a931202968748c23648a7f","subject":"Merge \"Cap setuptools to \u003c58.0.0\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e8b19cc35b8368d086a931202968748c23648a7f"}]}],"author":{"name":"matbu","email":"mat.bultel@gmail.com","date":"2022-01-17 20:34:11.000000000","tz":60},"committer":{"name":"mbu","email":"mat.bultel@gmail.com","date":"2022-01-17 20:39:16.000000000","tz":0},"subject":"Move upper-constraints for validations-libs to 1.5.0","message":"Move upper-constraints for validations-libs to 1.5.0\n\nChange-Id: Ibbffdd0ce12b1b0a39717b5a9b6ded189795ae6c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c7a133dd88dd55e58df13cf804e17f2b6afc7fbc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c7a133dd88dd55e58df13cf804e17f2b6afc7fbc"}]},"branch":"refs/heads/stable/train"},"7b2c5d655baa96a74d456aba7b863710d8797776":{"kind":"REWORK","_number":3,"created":"2022-02-09 08:18:15.000000000","uploader":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"ref":"refs/changes/60/824960/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/60/824960/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/824960/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/824960/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/824960/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/60/824960/3"}}},"commit":{"parents":[{"commit":"e8b19cc35b8368d086a931202968748c23648a7f","subject":"Merge \"Cap setuptools to \u003c58.0.0\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e8b19cc35b8368d086a931202968748c23648a7f"}]}],"author":{"name":"matbu","email":"mat.bultel@gmail.com","date":"2022-01-17 20:34:11.000000000","tz":60},"committer":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2022-02-09 08:13:54.000000000","tz":60},"subject":"[Train-only] Move upper-constraints for validations-libs to 1.5.0","message":"[Train-only] Move upper-constraints for validations-libs to 1.5.0\n\nPatch also bumps ansible-runner version to 1.4.6 in line with stable/ussuri.\nAs the ansible-runner\u003d\u003d1.4.4 is dependency of validations-libs 1.5.0.\n\nChange-Id: Ibbffdd0ce12b1b0a39717b5a9b6ded189795ae6c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7b2c5d655baa96a74d456aba7b863710d8797776"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7b2c5d655baa96a74d456aba7b863710d8797776"}]},"branch":"refs/heads/stable/train"},"925165c61eba58fc5f76090684e9a62e04ffd089":{"kind":"REWORK","_number":4,"created":"2022-04-07 13:50:57.000000000","uploader":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"ref":"refs/changes/60/824960/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/60/824960/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/824960/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/824960/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/824960/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/60/824960/4"}}},"commit":{"parents":[{"commit":"e8b19cc35b8368d086a931202968748c23648a7f","subject":"Merge \"Cap setuptools to \u003c58.0.0\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e8b19cc35b8368d086a931202968748c23648a7f"}]}],"author":{"name":"matbu","email":"mat.bultel@gmail.com","date":"2022-01-17 20:34:11.000000000","tz":60},"committer":{"name":"mbu","email":"mat.bultel@gmail.com","date":"2022-04-07 13:50:57.000000000","tz":0},"subject":"[Train-only] Move upper-constraints for validations-libs to 1.5.0","message":"[Train-only] Move upper-constraints for validations-libs to 1.5.0\n\nPatch also bumps ansible-runner version to 1.4.6 in line with stable/ussuri.\nAs the ansible-runner\u003d\u003d1.4.4 is dependency of validations-libs 1.5.0.\n\nChange-Id: Ibbffdd0ce12b1b0a39717b5a9b6ded189795ae6c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/925165c61eba58fc5f76090684e9a62e04ffd089"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/925165c61eba58fc5f76090684e9a62e04ffd089"}]},"branch":"refs/heads/stable/train"},"74fe3dca06f45ef7882c6aacd8385d406942ecf7":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2022-04-07 13:51:34.000000000","uploader":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"ref":"refs/changes/60/824960/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/60/824960/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/824960/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/824960/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/824960/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/60/824960/5"}}},"commit":{"parents":[{"commit":"e8b19cc35b8368d086a931202968748c23648a7f","subject":"Merge \"Cap setuptools to \u003c58.0.0\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e8b19cc35b8368d086a931202968748c23648a7f"}]}],"author":{"name":"matbu","email":"mat.bultel@gmail.com","date":"2022-01-17 20:34:11.000000000","tz":60},"committer":{"name":"mbu","email":"mat.bultel@gmail.com","date":"2022-04-07 13:51:34.000000000","tz":0},"subject":"[Train-only] Move upper-constraints for validations-libs to 1.6.0","message":"[Train-only] Move upper-constraints for validations-libs to 1.6.0\n\nPatch also bumps ansible-runner version to 1.4.6 in line with stable/ussuri.\nAs the ansible-runner\u003d\u003d1.4.4 is dependency of validations-libs 1.6.0.\n\nChange-Id: Ibbffdd0ce12b1b0a39717b5a9b6ded189795ae6c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/74fe3dca06f45ef7882c6aacd8385d406942ecf7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/74fe3dca06f45ef7882c6aacd8385d406942ecf7"}]},"branch":"refs/heads/stable/train","description":"Edit commit message"},"2165e48df6fed369b60dc70e5baeccddc32d3992":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2022-04-11 13:31:28.000000000","uploader":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"ref":"refs/changes/60/824960/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/60/824960/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/824960/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/824960/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/824960/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/60/824960/6"}}},"commit":{"parents":[{"commit":"ef77a2ae95debb2be64598c1c624d5d4e08b4cdb","subject":"Set setuptools for different python version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ef77a2ae95debb2be64598c1c624d5d4e08b4cdb"}]}],"author":{"name":"matbu","email":"mat.bultel@gmail.com","date":"2022-01-17 20:34:11.000000000","tz":60},"committer":{"name":"mbu","email":"mat.bultel@gmail.com","date":"2022-04-11 13:31:28.000000000","tz":0},"subject":"[Train-only] Move upper-constraints for validations-libs to 1.6.0","message":"[Train-only] Move upper-constraints for validations-libs to 1.6.0\n\nPatch also bumps ansible-runner version to 1.4.6 in line with stable/ussuri.\nAs the ansible-runner\u003d\u003d1.4.4 is dependency of validations-libs 1.6.0.\n\nChange-Id: Ibbffdd0ce12b1b0a39717b5a9b6ded189795ae6c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2165e48df6fed369b60dc70e5baeccddc32d3992"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2165e48df6fed369b60dc70e5baeccddc32d3992"}]},"branch":"refs/heads/stable/train","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
