)]}'
{"id":"openstack%2Frequirements~826046","triplet_id":"openstack%2Frequirements~stable%2Fvictoria~Ie5f13c9dfdd3277273752304895ef24259b14ad3","project":"openstack/requirements","branch":"stable/victoria","topic":"new-release","hashtags":[],"change_id":"Ie5f13c9dfdd3277273752304895ef24259b14ad3","subject":"update constraint for ceilometer to new release 15.1.0","status":"MERGED","created":"2022-01-24 11:47:35.000000000","updated":"2022-02-03 18:27:38.000000000","submitted":"2022-02-03 18:27:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"826046-new-release","meta_rev_id":"b750b9002d93283f3f45db756026077e0042d622","_number":826046,"virtual_id_number":826046,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-03 18:27:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-02-03 15:19:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-01-24 15:01:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-02-03 15:19:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-24 12:37:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-24 13:24:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-24 14:13:00.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2022-01-24 15:01:02.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"575c0a44cedcfaf3e3f71b3aa2a8281beeb92e7c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-01-24 11:47:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5dcd0118c4d4ea37a89f1306c3330d70fe85e000","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 12:37:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/196821a7eecf4d45bb9a39ac695ad31f : SUCCESS in 26m 14s","accounts_in_message":[],"_revision_number":1},{"id":"35fb863696d0ccd0ae7f8397313ee8670610e47d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 13:24:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93a237883b2e4a0c9abb876d8ed0ca08 : SUCCESS in 3m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f38ffd983f1a463f949f34e30eeee80b : SUCCESS in 3m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3dd990b15aa44a6cbeb5aa73d7f9c246 : SUCCESS in 5m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/24c46b42e88c445491ba2ab25e8beb7a : SUCCESS in 4m 47s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/adb412d47b0a4420864eb93acbd7ce00 : SUCCESS in 3m 28s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3080db3a6c0746a5809944f68786aa50 : SUCCESS in 11m 22s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0391b5d18c4b41119504f27956c00ea7 : SUCCESS in 9m 35s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/3989683184204be8b1530951b7e3da90 : SUCCESS in 9m 29s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/3aaa51bdf1c448b7b6858f6061a22b4f : SUCCESS in 5m 33s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/19cc2e305d034284a034a8128b92e4af : SUCCESS in 11m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4a8e53ef47774f7ab52a96c31d84fb96 : SUCCESS in 8m 09s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/3b1ddcbad55b482485da71e3cfd50e5a : SUCCESS in 8m 23s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/f3e91ae232464ead82b5a39513004247 : SUCCESS in 11m 39s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/42dff0693213499387b90b614228bb3f : SUCCESS in 28m 15s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/a8108d8b6e6644e79b0e364e082d5881 : SUCCESS in 13m 44s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/70ea6f901cec48a0911e504fd920bde9 : SUCCESS in 4m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3333cd68620a4ba096f344586282b0d7 : SUCCESS in 18m 51s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/daad4d44bf5540dabda0fd1f8c3822b0 : SUCCESS in 5m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5a9d6b8b31dc422787e4646269dd3291 : SUCCESS in 6m 08s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/baccedbea1ec4f8c8cf2603c542e1606 : SUCCESS in 11m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/434acb54d30f4788b90e5b98a4ce31ae : SUCCESS in 48m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/86d19b3527d3448c82ace5580b389489 : SUCCESS in 1h 26m 46s","accounts_in_message":[],"_revision_number":1},{"id":"4bdb1eb3bb4b564571d45cf82ca9208bf8d78578","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-01-24 14:13:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7dd5fbab33db4e4233255a4d1b1c00417c97b94e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 14:14:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fdc0b09502bb09e6deaf5923ec1f897f4c837810","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-01-24 15:01:02.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"06bf6fbc54e20fae6164010ad9f0c1f8040ef37f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 16:50:42.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/794db362082942a5a3cbb79c41a98d84 : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4e9b1c175bad4e0bb3adf56a34172b98 : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/82aa5e9302a1421e916942d68115e45b : SUCCESS in 4m 17s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/38b4dc6ac6754d09a9e68d545e1e4fae : SUCCESS in 4m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bd6a177486ce4270971cb2092ccc11a6 : SUCCESS in 4m 18s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/359543ec05a84db4882525544c93b2b8 : SUCCESS in 8m 55s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/caa6522d3c3342e6a48e0b6911e8d14d : SUCCESS in 9m 08s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/9ba466c447a04add8924c6fb7e1997b7 : SUCCESS in 9m 35s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/46f3ab0fa9004e0daf9febcc28eebafc : SUCCESS in 6m 07s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/db74df30ced943e6aef7d648db33b935 : SUCCESS in 10m 52s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/bc7e03bf5ed0418a8ea56bfcccb0cab4 : SUCCESS in 9m 11s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/0cc3745ec4ab4c86a78abbafba29410d : SUCCESS in 8m 40s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/4ee4a30b15b644b29e053876dea67d7a : SUCCESS in 13m 13s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/3b051d51228d4e0faa1951ca7561a2a1 : SUCCESS in 25m 48s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/c03501c904564546abad2b25904168ba : SUCCESS in 12m 17s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7de0d058c5f64391b576faba540a5b38 : SUCCESS in 4m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/318f3d73ea20410ebc935d32a1402da7 : SUCCESS in 21m 31s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/ebc9b9d85d47487e987adf2739f8d978 : SUCCESS in 5m 59s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e955f69086a34ae0a9ee4a991e65db4a : SUCCESS in 6m 39s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/6a4e19411b7d410ea0fa4e728d251628 : SUCCESS in 11m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b8e35e78de1c4f8b9782257ff3e751cb : FAILURE in 52m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5274b5400bf844bea080a8a327bf5cde : SUCCESS in 1h 28m 29s","accounts_in_message":[],"_revision_number":1},{"id":"2142871f284a7f51376492143fc38b34833dbd54","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-01-24 18:40:49.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3663a5ed63308311a83c6d144937b961c0928097","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 19:09:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/0af90cd6a09e4ff08231dffeb852aa43 : SUCCESS in 24m 36s","accounts_in_message":[],"_revision_number":1},{"id":"bfe220693e4a931a74d211ce5ff85bb79f5032e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 20:19:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89a3a7274a264c31b341e4592ee912c4 : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5910e1a1620d44569594832c9f3483fe : SUCCESS in 5m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2bed3bfd3f4c42338386d3ab80788208 : SUCCESS in 4m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3ef016f76e9c4be28888dbd9b7c871f4 : SUCCESS in 4m 28s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/87af9f0f58a8460fb4b57e93cc020d60 : SUCCESS in 7m 25s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d2945d8199f94b1988fdcc5ea6c65f28 : SUCCESS in 9m 31s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/84ac3a6c57504877b7f1744287786d05 : SUCCESS in 10m 09s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/89ca56677bc8457eb3e7e4bfcddc3cc4 : SUCCESS in 9m 10s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/0bc388896b9b4be4b15eb05f0a8fc93f : SUCCESS in 5m 47s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/4ac9f22946c64719bb7953b6d220e708 : SUCCESS in 12m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b88cbd7313de49028b5ca7602f100f44 : SUCCESS in 11m 13s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/da5f1f1bbafd4919aae578f89723fef2 : SUCCESS in 7m 43s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/22b305bd3a794a6597e3c5d983a13e61 : SUCCESS in 13m 12s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a8841023d0f448c1bab15fd1f3ed2851 : SUCCESS in 24m 50s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/4df2b8d8ba2648cd9c998615361ba532 : SUCCESS in 11m 58s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0a2a1eef058f4986b97edd5403172f92 : SUCCESS in 4m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/776724eb1e874f0e9b53837e6990da00 : SUCCESS in 18m 27s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/c38f62c7f0b447d0a0e39b3b622b00d2 : SUCCESS in 7m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/092adb66708c4b6781425211eda89d05 : SUCCESS in 7m 34s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/28f87ac281bd4ab19ee8690855772b54 : SUCCESS in 13m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ac44bcaa28a348979c6255f6221ba129 : FAILURE in 48m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b22efd1920ce48628f1606981fefd1ca : SUCCESS in 1h 32m 38s","accounts_in_message":[],"_revision_number":1},{"id":"91bc1fed3004640c03e8e9294f342c8396dab823","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-01-27 04:16:02.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ead0ef2ce5117ba627ac60ffe2bd62cf269290b2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 04:35:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/1d8a4b9430844d65b93c690760b42e71 : SUCCESS in 17m 34s","accounts_in_message":[],"_revision_number":1},{"id":"39205e8d56c63797aaeff058f7dbdf18b28f3041","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 05:27:39.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93fe3e5a6b7a48b6b3dab8a27b95444a : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b18e76a638b749f6be87aef1f426c1ae : SUCCESS in 3m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/57404d71604d482dbdac41f139149d43 : SUCCESS in 4m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/00157b4fa7e44eaa82ea0f0244e756ec : SUCCESS in 3m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d48c8c1242824230a5ee9e0f823648e6 : SUCCESS in 3m 42s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/01fd645f49674d15b91b2e04d5e4fd04 : SUCCESS in 7m 42s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/62a86b02bca5494fbd8111c6467503c2 : SUCCESS in 8m 50s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/c1a1e695d95445e091c8dab867bf60a6 : SUCCESS in 8m 08s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/3d539ef8e03c4058bf743aca046def86 : SUCCESS in 4m 53s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/fe85530b2f8245cfb474b9bef49f7687 : SUCCESS in 10m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4d4f84446ae74e27919dd55f91b1b47e : SUCCESS in 9m 01s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/e06809b64f464974bc264889d0b0b2b5 : SUCCESS in 7m 44s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/508e666cd7a44355ad2bef2f78cfbcee : SUCCESS in 14m 28s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/e523066a085a47ba9a29dc2b555bbde6 : SUCCESS in 20m 41s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/6e6d6d1b55794f3a85ec8192c00c15ed : SUCCESS in 12m 13s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/45a2bffed8e343e2b48fa4b4f73dd64c : SUCCESS in 3m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8183d6fbca7949b4a26abbe64efa15bc : SUCCESS in 15m 55s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/834b82a01cf2476eac8d71326306aca1 : SUCCESS in 5m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8ba7189fea224b51ad353fb36c2cab4b : SUCCESS in 6m 31s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/242421ca15f14c37b168bda42173d705 : SUCCESS in 11m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/14864c7a068b43a9a4af6cf3f1c42c59 : FAILURE in 39m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fc59b1c1585c4de79bbde1cf44950ed0 : SUCCESS in 1h 05m 59s","accounts_in_message":[],"_revision_number":1},{"id":"532411aa696f7a92291229796b5be67545b6ac58","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-03 05:01:11.000000000","message":"Patch Set 1: -Code-Review -Workflow\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"aad4c0ed25883f1b0287da9d1e35b82a1f180623","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 05:34:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/b9ce018daed9425bbcd68221dbce5e8d : SUCCESS in 20m 32s","accounts_in_message":[],"_revision_number":1},{"id":"610276487898488f70ca0f0a395cc6600abf4b64","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 06:14:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b102b232ae0142c4962be8caf9ed8a12 : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ab7d1ca3ac63449eb9e87fce06b7f8dc : SUCCESS in 5m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/13556da385d5462cbd60c6dc263ed17e : SUCCESS in 5m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f980654406dc407788d6e54cd3cf555e : SUCCESS in 3m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/57659e1717924ac68f68fa49157bfa5a : SUCCESS in 4m 18s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e610ba5083cb495f99b32e37eb916579 : SUCCESS in 9m 05s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5861eb4174554edebbf903a51d643668 : SUCCESS in 9m 33s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/245a98fade3c42a7882b2e5dbded2eda : SUCCESS in 8m 05s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4a5ea8497697434faab0aa291df269be : SUCCESS in 6m 26s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/1853085107bc495bbe55c059aa27371e : SUCCESS in 10m 58s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/abc0548bd555404483b5276e26aae098 : SUCCESS in 9m 31s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/86f0d4f76574415099ba2b1a60c0ba79 : SUCCESS in 6m 43s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b5e675ee9ff44a5a9007651d1282ed60 : SUCCESS in 12m 57s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/5e4450b035f840da9c184f09fc1953f4 : SUCCESS in 22m 40s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/80a4c3eccc9840bf821a11e394937e61 : SUCCESS in 10m 17s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8a7efadd588f4c8e9e4a8e04b5a27496 : SUCCESS in 5m 31s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0203c91f9afd490f9aa61682f6d3bd0b : SUCCESS in 18m 24s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/3d0657955bb140ce8640177a599ceebb : SUCCESS in 6m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/19402733e88e4de0964e01bbcc692081 : SUCCESS in 6m 08s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e7fd7bdf89b34762b76a269fc7a8a4a2 : SUCCESS in 10m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/60dd67084efb47f29e551f395ad85bf6 : SUCCESS in 51m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a7c9db0377984d49ab291a740b5268da : SUCCESS in 1h 09m 52s","accounts_in_message":[],"_revision_number":1},{"id":"aee1eca01a7dcc50dd8a03b255f19831107f0d96","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-03 15:19:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4ed68c3ab946b8e8da672fbe0be6fdd5c557826e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 15:20:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4e31c1b50d739feb873173f318aee1f99dc0e799","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 18:27:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05c748545e924048b74072291f6e1f4b : SUCCESS in 5m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ddf5b3a5d8924eb1bed4e872840f1d4f : SUCCESS in 5m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c1e1c91c551a433a840f7be3ccb51656 : SUCCESS in 6m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e1a1ede912f943e7abfeb37c333f526d : SUCCESS in 5m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bf00320fd0d24f3b8943cb735b349364 : SUCCESS in 4m 06s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a870f94f8baa483484f18643b8ac71fe : SUCCESS in 10m 20s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e15d1a21c021420880bb348235de1f19 : SUCCESS in 9m 25s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/11640bdc01784f42a6529c390d443cf8 : SUCCESS in 9m 42s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/68c87e6b2dfd41a0985fa9ca3af26154 : SUCCESS in 7m 26s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/7fe5c00444e94affadfa3717c7d354b8 : SUCCESS in 11m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/52fd323f39ac4c59a357046219786902 : SUCCESS in 10m 33s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/2f1cbeb3febe4b2498cb56afdd9bc31b : SUCCESS in 10m 38s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a6d30e35841f48f98f90865850442878 : SUCCESS in 17m 47s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/fca4854bb0d444349609ad771c57abeb : SUCCESS in 27m 39s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/4f9f97a26e924fe8b86bd160eb7321d3 : SUCCESS in 13m 10s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f5e614087e3f49fa851e474356466da4 : SUCCESS in 5m 44s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ae6a79ece2d74ec0a047be508efe1e8e : SUCCESS in 17m 04s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/7f45bf186ff54e1c9c12da721742b282 : SUCCESS in 5m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/99bde487143d4768bbb3a14766fa0c80 : SUCCESS in 6m 32s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/3668b27b351846389f44eecab9e86059 : SUCCESS in 12m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/48e195f1094047e1b0c32f9c82be911b : SUCCESS in 55m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f6c895c74bc649d9b0e3fa7879b392f2 : SUCCESS in 1h 40m 21s","accounts_in_message":[],"_revision_number":1},{"id":"b750b9002d93283f3f45db756026077e0042d622","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 18:27:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f6ebd639e77180b21fe8a4c1285a1c39edd8e8e5","revisions":{"f6ebd639e77180b21fe8a4c1285a1c39edd8e8e5":{"kind":"REWORK","_number":1,"created":"2022-01-24 11:47:35.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/46/826046/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/46/826046/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/826046/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/826046/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/826046/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/46/826046/1"}}},"commit":{"parents":[{"commit":"f8214a80076fd47ba9d03477af12b9314d7734f1","subject":"Move upper-constraints for validations-libs to 1.5.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f8214a80076fd47ba9d03477af12b9314d7734f1"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-01-24 11:47:32.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-01-24 11:47:32.000000000","tz":0},"subject":"update constraint for ceilometer to new release 15.1.0","message":"update constraint for ceilometer to new release 15.1.0\n\nmeta: version: 15.1.0\nmeta: diff-start: -\nmeta: series: victoria\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Előd Illés \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I0dd2a898d7b47f118537c31e52ed1ac252cefdc8\nmeta: release:Code-Review+1: Matthias Runge \u003cmrunge@redhat.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Ie5f13c9dfdd3277273752304895ef24259b14ad3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f6ebd639e77180b21fe8a4c1285a1c39edd8e8e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f6ebd639e77180b21fe8a4c1285a1c39edd8e8e5"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
