)]}'
{"id":"openstack%2Frequirements~826047","triplet_id":"openstack%2Frequirements~stable%2Fvictoria~Ifecf3b41b5c323996ab214faf92579da8baeb0e2","project":"openstack/requirements","branch":"stable/victoria","topic":"new-release","hashtags":[],"change_id":"Ifecf3b41b5c323996ab214faf92579da8baeb0e2","subject":"update constraint for glance_store to new release 2.3.1","status":"MERGED","created":"2022-01-24 11:50:30.000000000","updated":"2022-02-03 18:27:41.000000000","submitted":"2022-02-03 18:27:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"826047-new-release","meta_rev_id":"5ffeeeb33e6eef810a5af5d5eba7444518ad7971","_number":826047,"virtual_id_number":826047,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-03 18:27:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2022-01-24 14:13:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2022-02-03 15:19:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-01-24 15:01:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2022-02-03 15:19:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-24 13:00:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-24 13:27:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-24 14:13:01.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2022-01-24 14:13:01.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2022-01-24 15:01:17.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"8c54471e25ce5f991fa65a0860097da0b9c3042f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-01-24 11:50:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dd784062c26af3ba02dbfee811af7ebab4ff45ac","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 13:00:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/71f7b828ae724d5a9615e996a2c4f92f : SUCCESS in 20m 24s","accounts_in_message":[],"_revision_number":1},{"id":"0c3ccdd49b8fc2a18eea74263c6063c35e0f650a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 13:27:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f9939599974445b8128d2ef2d5f6569 : SUCCESS in 3m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9789faabcacd48669c0af5884f27d72c : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/61084a44bf724599a0bd044663717486 : SUCCESS in 4m 14s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5a886329032e4adca3df16eed255052c : SUCCESS in 4m 50s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b78c8e4109534d6d9e694556c0d10738 : SUCCESS in 4m 00s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/22624b690ce943028ebf58c3b59273eb : SUCCESS in 10m 30s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/78b623cfe6fd47f981ac707719642bca : SUCCESS in 9m 52s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/da3518cc9bc84a22b51611d9d9494156 : SUCCESS in 9m 47s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ed004db61ba5417fa7cc2e36ae2a53ee : SUCCESS in 5m 40s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/85f0a1204edf41c7bae30ee57a4e3e39 : SUCCESS in 10m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c9ac20943b6b4bb0b87a012b08f58a65 : SUCCESS in 7m 50s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/d49cd541e79f4f0ea9add8dfea36ca65 : SUCCESS in 8m 06s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/d875eec1a8bd4a46a9f166cf507a9154 : SUCCESS in 13m 13s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d3823d7b519b41fe81086d43fd6f2275 : SUCCESS in 22m 00s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/6ab21d871f4c4efc99d656a9892e6da1 : SUCCESS in 12m 17s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a15b3a0f82864dc4b73773f911689d8f : SUCCESS in 4m 49s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7785d9d50b7f48a88fdb8e4908f05d5c : SUCCESS in 17m 41s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/e9eef2f3f3a347819ce8bba2afe39462 : SUCCESS in 5m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/65aa49c3bfa74794ae4e3291d5e7681a : SUCCESS in 6m 53s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/4c0fdea23c174d4893f3de0ebb48a729 : SUCCESS in 12m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9e340504e54b4010a73a807edfc1c7e0 : SUCCESS in 35m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a7b54aec0b37486f89d8b81d6664d152 : SUCCESS in 1h 28m 48s","accounts_in_message":[],"_revision_number":1},{"id":"9a0573b2b1f0eebe97a4e68e82da48df8fb0053a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-01-24 14:13:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"cdb993ec89bac42278168a75ad11f28c31021ccc","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2022-01-24 14:13:01.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"acafe83b4fa4d73a3f85945a8003c7ec9e2de213","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 14:14:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7d6d2495f9f72b1711a843e073dc91a335cbaeb7","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-01-24 15:01:17.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ca9fb89a888c5ae235b77226b2f9de3c565acdcc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 17:48:21.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/545e6527b88d4af196a8ba8d6878f403 : SUCCESS in 6m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5562564a76e24717b55cc4b9955c291d : SUCCESS in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/14c0f54a059a4215b58e0560b5366ac7 : SUCCESS in 5m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/01c94f11e535455c9cf2de7645af9fcb : SUCCESS in 4m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d4a745317842448a9dfcfd0baa0cfc3b : SUCCESS in 5m 32s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/7d148857d2a14bbeba93fe85b20440a0 : SUCCESS in 11m 56s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6d43e069d54746c3b2d05aefbb61bb72 : SUCCESS in 10m 06s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/a9aff01e130042efb9ff0b960412c2ee : SUCCESS in 9m 47s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/39eff11b16a446319ca4050f88d89d5f : SUCCESS in 6m 07s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/26fa22c679934d94a111ea1822eedd9e : SUCCESS in 12m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fc4a108bd4334fb799b16535dba54893 : SUCCESS in 11m 21s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/ec5a8819cfaf4d6c8193e065deb57977 : SUCCESS in 8m 31s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/af3e10d35b8d49feada7ee669051a090 : SUCCESS in 19m 44s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/4f40c6fd1dfd44ff878b183590e50b15 : SUCCESS in 23m 29s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/d16bdeef224542a5a59ed56f472bce30 : SUCCESS in 13m 05s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/198b23c81f174f3c94dffc41f893990d : SUCCESS in 4m 56s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b1ae86201c444c40aa6178185e5e7478 : SUCCESS in 18m 14s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/25d780eecc76484eb81f94674f56ae90 : SUCCESS in 5m 53s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/535118004ffd42c49dcffb92c0d3aedd : SUCCESS in 9m 38s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/b586a064aacb44fe98b6ee14ba90d722 : SUCCESS in 10m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/331abba1ca1a461fabdda5c0afc58b44 : FAILURE in 1h 06m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ce8bd432c7ac4c468c7d5cb44005d255 : SUCCESS in 1h 28m 48s","accounts_in_message":[],"_revision_number":1},{"id":"f839e90be39a2b3e3f8bcce72cbd6e2d4333655d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-01-24 18:40:50.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"08c4f7873b0923c7eeab75a78b45a850f5accfe4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 19:02:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/e3bfe7e187244490b826bc9d41423e90 : SUCCESS in 19m 07s","accounts_in_message":[],"_revision_number":1},{"id":"f7131b0d08995bb078d87d16493a6fefa35d5bdb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 20:10:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/944b6e1fa1c54b6fbc7ca8cb4dead33b : SUCCESS in 3m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7c11f12bf973430c863f1753489186f1 : SUCCESS in 3m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/55999a67a2ab4759b996e7c043e50648 : SUCCESS in 4m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bfe348ae1c464b57baf32c568e71a1f1 : SUCCESS in 4m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c693f5b96f7441cbb19b3194c6ed3de2 : SUCCESS in 3m 34s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6132fcfb02c9457d9168755c760db368 : SUCCESS in 9m 27s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8b171ae6d75e4fcc9931974d8522a883 : SUCCESS in 8m 58s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/c21a89c7b0a841bdbb3714f722fb7581 : SUCCESS in 8m 38s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/110bf92d3ad5452586edf260f28ee897 : SUCCESS in 5m 15s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/7f69180b2e0c42e29d10487a16469d4c : SUCCESS in 10m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/96f1d7bceeef43e4a5b6cc194a0ac164 : SUCCESS in 11m 59s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/45f98694bc08444a9b19f5dc2bc9daa1 : SUCCESS in 8m 47s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/cf14c21004ae4f9897972e9431f27a35 : SUCCESS in 12m 47s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/091d72a30d64478da958be9cf1ede661 : SUCCESS in 30m 13s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1094d2bfaf314eb48067c603965acca0 : SUCCESS in 14m 55s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/38428d9e6fda4a5085b2d4accb556fd7 : SUCCESS in 4m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8e175341cba0490d9c6603f3bc853762 : SUCCESS in 17m 20s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/851cdbdfb9f34ed2ac1918f11d598d09 : SUCCESS in 5m 55s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1b9af0ee9675459fa088755689588734 : SUCCESS in 8m 39s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/46b463ff18fb422682c5458978d89d3a : SUCCESS in 10m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a33769b3628347d795d91bccf2adf26e : FAILURE in 54m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d6bd0d08a6094772ba19e11bedb90443 : SUCCESS in 1h 26m 58s","accounts_in_message":[],"_revision_number":1},{"id":"de5c5baa56b32aa1dedfa6f8bbb81f7e3c520ba2","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-01-27 04:16:03.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8ac836e50779224f0699dca8e165a2ff67a780e6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 04:39:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/4c8d292a30984a209297342130ab65ca : SUCCESS in 17m 55s","accounts_in_message":[],"_revision_number":1},{"id":"f93c51d2f35ea2fcfaff2be18f059753c20aa1a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 05:31:46.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/903f73fee6294a2494b2ed3e24e99be1 : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8bf3640855f84a71acc558b4e1f35813 : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/077fb7f9551441909c88fb7f308bfebd : SUCCESS in 5m 03s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3f83043ea3e24ed1a4d579953fafa792 : SUCCESS in 3m 51s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fb89408c43a8487387eefca585fa363e : SUCCESS in 3m 32s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d5e0956c86544623bee3c7b146c09737 : SUCCESS in 8m 06s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/68538b777a0e46348c4cbd5b056b2629 : SUCCESS in 9m 10s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e0c2a7d09d484bc8bbe3edba13e46078 : SUCCESS in 8m 34s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/8d852b9f3e2648deb117e205ebd3aeb7 : SUCCESS in 6m 30s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/b6547500cdd04dc08fa39588bfcce793 : SUCCESS in 8m 45s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ba7fe8bef4e0414cba8310b2912b01bb : SUCCESS in 8m 09s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/3670670a92564e9da3c1d30cfa51373a : SUCCESS in 8m 24s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c106545aafb64a97ae20560a667e5577 : SUCCESS in 27m 18s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/dad36c0a9b314f098c6e2112be9a4cc1 : SUCCESS in 25m 02s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/a32731c49853464692c480b0fccf6f16 : SUCCESS in 11m 16s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4307792933b04b12af32ab854531d451 : SUCCESS in 3m 46s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/61e8d3dcbbae4cb5b162d2e2532b33c3 : SUCCESS in 15m 59s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/605e1a0f65e248749dfed4c81ee6a07e : SUCCESS in 4m 53s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/54fec1838b364a0e81f529be8d6011c4 : SUCCESS in 6m 10s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/2d7bf6bf09e048d2bbdad7e49ca2a231 : SUCCESS in 8m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1ff84f80a5cb4034b4ce90ff9dc5820f : FAILURE in 49m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fe60a369a88b4b05b9528900fb5452ce : SUCCESS in 1h 13m 06s","accounts_in_message":[],"_revision_number":1},{"id":"ca6f11e32295e9eabcd79134e0ac5c7f99265a12","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-03 05:01:12.000000000","message":"Patch Set 1: -Code-Review -Workflow\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d6e3a64b3cb5f8d933e02339e527352762f7e284","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 05:34:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/6478b67bcd3c4016b870c44bf62e2eb0 : SUCCESS in 20m 41s","accounts_in_message":[],"_revision_number":1},{"id":"b0cdfdfb8bc9058ac3dc275e77b33e71ddd5a8a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 06:20:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06fae1d0b9ba4da4830e40d947e3a448 : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8087152f9a37430995f753880e441b7e : SUCCESS in 3m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/137cd11965f142d7991582fbe3ddce90 : SUCCESS in 6m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cb0087608c464de2ac62318b18dfdce9 : SUCCESS in 4m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/255ef81df4ba4510881dc1d3deaa7874 : SUCCESS in 4m 51s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1b125590a5224012bceb20822b6a0f76 : SUCCESS in 9m 25s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/7b39f2a1e8c14ada8d8183a9c7935a47 : SUCCESS in 8m 43s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/6b438dc32c3b4717b84b592f61735351 : SUCCESS in 9m 41s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/69dc92cf58f9485eb8608589017612a9 : SUCCESS in 5m 00s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/6a662bb461db4d5a842dc7c14b39132d : SUCCESS in 7m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/99d59d7536c940ed9518974900f23dc0 : SUCCESS in 8m 57s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/983ed3b769484bd4bd017778d462bd08 : SUCCESS in 7m 48s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/51538f48a6934784a1a3d61d377bbe71 : SUCCESS in 10m 38s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/69b7ac77a6f34614ab19164c9d7935b2 : SUCCESS in 25m 06s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/4b721107d70747b5a0850c9655271913 : SUCCESS in 11m 20s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c2de70f4be72430787d29b2844b852f3 : SUCCESS in 4m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5f4ae9fc7df34bbab26c8a86f1090e1b : SUCCESS in 18m 47s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/e9a1db9dd88548dbba42cc27f4938261 : SUCCESS in 4m 43s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9e3b1f720b054053a95d33192f577c8f : SUCCESS in 6m 52s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/a495e68b0b9b4347a7fb5cacd22e0819 : SUCCESS in 11m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8490cddeae52482a9ddd9b4933e1f6f9 : SUCCESS in 52m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/62da8d9cd8fb49cd9d59cceb7cfdcdc6 : SUCCESS in 57m 18s","accounts_in_message":[],"_revision_number":1},{"id":"131fc3167ae59f8f2459663cfd9bf71689e54e86","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-03 15:19:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4c0827ec986a6d0c27331a5b1ed27d1060d433e6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 15:22:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cb066b4f7473d9e8aef4da5f3d2348a4bd9761ce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 18:27:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1087e8d6441c44bab2d4ebb8cdbea35d : SUCCESS in 3m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/070610b693ab473daa947d6910140dd2 : SUCCESS in 5m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7f84c21fa92c4303b535b96debbd1509 : SUCCESS in 5m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2309c2ac74d04117b033a8df0c1d9fc4 : SUCCESS in 5m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/87929286f08549478df315ced53155a6 : SUCCESS in 4m 45s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/91144816142e4a968fcc9ce4d3f8d3fa : SUCCESS in 8m 37s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/af273f88a0cd4561bf92cad820786db5 : SUCCESS in 10m 16s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/a8bd2010ca394c86a0220d5f881ccabe : SUCCESS in 10m 42s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f0fe88b53ea94377b105a16aebc8a106 : SUCCESS in 7m 05s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/5c8d98ac0fee4c52aa308a2232ea5b86 : SUCCESS in 10m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9b18e0728ba24ed7b8932845c46caa6a : SUCCESS in 10m 45s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/9796cefb46f54949a48b8a8c23cce3a6 : SUCCESS in 9m 08s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c2c19dfbdf22427eb0f9361255a93b3f : SUCCESS in 14m 16s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/e9169f42076a482593c4ab48ce89167e : SUCCESS in 26m 14s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1a698ba6662641908afd942613e1533d : SUCCESS in 18m 44s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9fb5b3fc018243e19386536c645fa4d6 : SUCCESS in 5m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f464af0c6bdd4e408a6d4047de20ef62 : SUCCESS in 19m 10s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/6e284378e3d847ffb3ad4398cadbebe8 : SUCCESS in 5m 52s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ff22c1849be04bc0bd7a1eee5470e9b1 : SUCCESS in 7m 36s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/36da892cf5bc436ab91a689081c5b687 : SUCCESS in 14m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5ff2cf369a834504bf5fc479ee8201f6 : SUCCESS in 55m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d559b0488668432088af33c3d487aaba : SUCCESS in 1h 23m 33s","accounts_in_message":[],"_revision_number":1},{"id":"5ffeeeb33e6eef810a5af5d5eba7444518ad7971","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 18:27:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ef8f6ef61087a122b42aec35faa5e6de3f29f6c4","revisions":{"ef8f6ef61087a122b42aec35faa5e6de3f29f6c4":{"kind":"REWORK","_number":1,"created":"2022-01-24 11:50:30.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/47/826047/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/47/826047/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/47/826047/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/47/826047/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/47/826047/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/47/826047/1"}}},"commit":{"parents":[{"commit":"f8214a80076fd47ba9d03477af12b9314d7734f1","subject":"Move upper-constraints for validations-libs to 1.5.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f8214a80076fd47ba9d03477af12b9314d7734f1"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-01-24 11:50:23.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-01-24 11:50:23.000000000","tz":0},"subject":"update constraint for glance_store to new release 2.3.1","message":"update constraint for glance_store to new release 2.3.1\n\nmeta: version: 2.3.1\nmeta: diff-start: -\nmeta: series: victoria\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Előd Illés \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Előd Illés \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I2c354abb8b13d31cccbc0129291c5a25df2909dc\nmeta: release:Code-Review+1: Abhishek Kekane \u003cakekane@redhat.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Ifecf3b41b5c323996ab214faf92579da8baeb0e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ef8f6ef61087a122b42aec35faa5e6de3f29f6c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ef8f6ef61087a122b42aec35faa5e6de3f29f6c4"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
