)]}'
{"id":"openstack%2Frequirements~826049","triplet_id":"openstack%2Frequirements~stable%2Fvictoria~Ie891008b10f90f83332ae65db5c5c95c6a48e663","project":"openstack/requirements","branch":"stable/victoria","topic":"new-release","hashtags":[],"change_id":"Ie891008b10f90f83332ae65db5c5c95c6a48e663","subject":"update constraint for sushy to new release 3.4.5","status":"MERGED","created":"2022-01-24 11:52:54.000000000","updated":"2022-02-03 18:27:44.000000000","submitted":"2022-02-03 18:27:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"826049-new-release","meta_rev_id":"7456cb65c7e004300393bf81013fb99d49c6812d","_number":826049,"virtual_id_number":826049,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-03 18:27:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-02-03 15:19:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-01-24 15:01:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-02-03 15:19:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-24 13:12:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-24 14:13:01.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2022-01-24 15:01:31.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"be8cafdf88580d0389e63e234014f050913e9529","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-01-24 11:52:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b2e1ab6cedb51aa2ab164975f654489eadd5d1bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 13:12:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc398ab8992f417fb7ab05663d2c566e : SUCCESS in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5304cf10da464290908f46375e76ffb8 : SUCCESS in 4m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4f80aa0625d9431fa1046ffe46c7fb66 : SUCCESS in 6m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b762a01c9b1b496e977a01a005755b5f : SUCCESS in 4m 32s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fb6b2c3deac34057a94eeae0633b6283 : SUCCESS in 3m 59s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4290e85502f64d07b472486602ce4e0a : SUCCESS in 9m 00s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/521ae51181fb46f7a8053389b87e9d3c : SUCCESS in 9m 47s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/40f265567c6e48d68fa0edee188b3c9f : SUCCESS in 7m 38s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/70cbc9df66f4433782e5b256b54ca31c : SUCCESS in 6m 04s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/dde6f7b540d34c11afed993ead839945 : SUCCESS in 9m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c84e797b02914de58f74a110d2ddc16e : SUCCESS in 10m 13s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/3edb5ddb4cdd429cb72bf126e4f78e98 : SUCCESS in 7m 04s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/615322682773439498ff46646362deeb : SUCCESS in 11m 31s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/dbc4e04c3e7e4c2bb6123804831d1860 : SUCCESS in 21m 09s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/6a3082668eb041ceaec745ac8b40d527 : SUCCESS in 13m 10s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b0e02ddce78146d9b5bd6c078a43acb6 : SUCCESS in 4m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ddf5b77a60614cd3a395eaaf068946f7 : SUCCESS in 16m 59s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/815c3fe78e5742efa9a478a746356dd0 : SUCCESS in 5m 25s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c1295be87712404b8c8bfcd0505511f0 : SUCCESS in 6m 08s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/71fc849774f74cdd9d58d5b1b5ad8d0a : SUCCESS in 12m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/47dd0f4f709d488dad0e6a1cd8027127 : SUCCESS in 1h 03m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8326eece91644fe0bcbe73adc3373fd1 : SUCCESS in 1h 08m 53s","accounts_in_message":[],"_revision_number":1},{"id":"9d4efa70103ab4d50030cf3eb70a1b88ae2ebbd2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 13:25:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/b1edf6fcf4b242db8dacf94b52173ad2 : SUCCESS in 19m 52s","accounts_in_message":[],"_revision_number":1},{"id":"e4b01fc51eea5c9f1e32ae8bb048510c721596be","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-01-24 14:13:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"db2939a257a9345d3769a2bca6b58588b299c548","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 14:14:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dcb4fa3db2820d37ee067c092d34ccab7f73a2e7","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-01-24 15:01:31.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"38a7f803802597f396034d4f991d27b538a54f2c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 18:56:11.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3105cc9ac9be4b7aa1dcbb44238d8384 : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e4f810b1ef5f44fd87bd52650545246a : SUCCESS in 6m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b44d2bff858e43f583d757389813ceaa : SUCCESS in 6m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/27b341f9b1144d6da709f4e0c1e6b276 : SUCCESS in 5m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e20097e512cf4022988b1f43d2892413 : SUCCESS in 5m 41s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8842a23b3ec242098ef6e9c173b2870c : SUCCESS in 8m 34s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/eb6ea88c39774b619bc6b339117c3264 : SUCCESS in 9m 03s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/65ac74a639a442b3a082ce0f7f3447bb : SUCCESS in 10m 43s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/fdfde235b370445b9beb5c442411abeb : SUCCESS in 6m 48s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c0e59d3fd6f0431697cd5ed735636842 : SUCCESS in 10m 21s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e9a665c96c494be98b31108e4f050b50 : SUCCESS in 10m 26s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/8578063cf43240ad95ad60751d270845 : SUCCESS in 9m 26s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/8e95df87588f47b19601a6188bcf1480 : SUCCESS in 25m 11s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/fd72b7f64afc489486df589f5581adcd : SUCCESS in 24m 14s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/d704d869d482425fb1e984f4d19ef730 : SUCCESS in 16m 04s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4aed8123a2144924bdfb08af21a8c84b : SUCCESS in 5m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c79d41d9cfb64d3d9b0219de8175d9c0 : SUCCESS in 17m 29s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/433d531f21a249be968ad313dbbde40e : SUCCESS in 6m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e8969f8b37214ca98cc1691a25ff0898 : SUCCESS in 8m 08s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/1110063b597e484ea5cb08b25e2dfd00 : SUCCESS in 15m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bd67e0de02ed45ae80fba94c3b91f662 : FAILURE in 52m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b5fb9b50df274733b4838f7f3af61c5b : SUCCESS in 1h 26m 40s","accounts_in_message":[],"_revision_number":1},{"id":"696d151df9baea5689c24bcc9becec57dc2cb448","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-01-24 18:58:46.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3c437aef42efbb173dfdbc477aa8e77a634c6f26","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 19:19:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/37f72cbfc7b64a00a9c73e8aef3937ea : SUCCESS in 15m 13s","accounts_in_message":[],"_revision_number":1},{"id":"adaf91248f85177f9bc485ae334802d50897a621","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 20:02:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91870a17bcfb4dcebbc57be133b0da69 : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bf2503e48b304b3daf03bbb919176b85 : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d196abf612904370ba25f94e38d2c09c : SUCCESS in 5m 34s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8f522eeefd0347d1a32d32b12c7e7892 : SUCCESS in 4m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7bc16bb86ead4ab4a7849ce2437e5789 : SUCCESS in 4m 19s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c63059419621496a9c3417bbe6a5eb1b : SUCCESS in 8m 20s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/7a4a39633eb5413491b2573632c0d308 : SUCCESS in 10m 13s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b98d0bd4de9f43ac9dba96b45d269b4e : SUCCESS in 8m 29s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4aa26586be964931b059aa4ed5cb5852 : SUCCESS in 6m 25s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/4a7d265d670f49238121d76aaae87662 : SUCCESS in 10m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9402dc78511e4b9da23455db1b6aac19 : SUCCESS in 9m 54s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/e8306dde18244ba2ac41bc35c4c640db : SUCCESS in 7m 05s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/cd41ea46537d4ff5842ec05ef83fa2b4 : SUCCESS in 14m 37s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/3e362d3cb247491e886eaad92d1c6366 : SUCCESS in 24m 33s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/20e63a3f40dc4198bae33abceb7a48ae : SUCCESS in 13m 54s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2b68689be86a477d89f49a943e276dcb : SUCCESS in 5m 29s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/66335792fd8e4f3ea2c6370523d21fb5 : SUCCESS in 17m 58s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/3d29e9c47eda4036a5eae04b1f7a4377 : SUCCESS in 5m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1e0814754778466081b834e6dbe7eec0 : SUCCESS in 5m 57s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/db1242b0e89848b986e0ece43e827a72 : SUCCESS in 9m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9554c0afaff749ed8fac04bdf91beb84 : FAILURE in 45m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6c903c30cdc24d1180289f40eb20fde6 : SUCCESS in 1h 00m 26s","accounts_in_message":[],"_revision_number":1},{"id":"fe3ad485c84d68458754ba21cdaf3e7dd62ab9a0","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-01-27 04:16:04.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"19e55d86e031a4503db4e7c31bdaf875740656aa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 04:39:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/60ba09baf32a4da990c5560a601312e4 : SUCCESS in 17m 51s","accounts_in_message":[],"_revision_number":1},{"id":"18f2af1fe58bc616aaba45c67ce32d91ef2252e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 06:06:24.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42c35e79c745416fa3e71649046cf5cd : SUCCESS in 3m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/41f644471d204010b5c265932221aeec : SUCCESS in 3m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b382e79eef834dd2a1f4f6701c99f257 : SUCCESS in 4m 02s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/001b4bf291484568b1fec67fb2f8f643 : SUCCESS in 4m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/760a372a9a104ac1b0c337149e2cce84 : SUCCESS in 6m 03s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e67cee87b3bd490fb33ad3bf3bf7cd1b : SUCCESS in 6m 13s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0ba07d1776fd45daaf75fb5ed9ae8902 : SUCCESS in 16m 52s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8747ce07cce144c6acd0f954013ee9ac : SUCCESS in 8m 08s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/bc95835320c54a389fd182f420d6a97f : SUCCESS in 4m 28s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/9922d182218b45f39df9d0417e3d8c56 : SUCCESS in 7m 39s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/46e410878e8a4469831451450d968c66 : SUCCESS in 7m 30s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/307667f344014d51a366423102c5cce8 : SUCCESS in 5m 42s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/741777d070d341c2aaeec3bf9a954667 : SUCCESS in 11m 22s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/058367e221694412806e6cb5d928e543 : SUCCESS in 20m 16s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/af82f071c3214aa89d0de0ede1f51df7 : SUCCESS in 11m 07s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/88d263cd99294777b1fcf51a801295c2 : SUCCESS in 3m 43s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c3566554c339475cabb9234389320d5b : SUCCESS in 13m 42s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/a0fee4d926874d8f8e9ee0022c02d2ec : SUCCESS in 4m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/571e4d46532e4883a8c7c664dd02980b : SUCCESS in 5m 25s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/d841827eb3374049a9ac9c3b4de77ebc : SUCCESS in 9m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7538fef4835c47afb269405b623f87d3 : FAILURE in 38m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2b01cd98860d427694363cb91e3e0641 : SUCCESS in 1h 45m 11s","accounts_in_message":[],"_revision_number":1},{"id":"4cefdac237bcb382a9af7d8a0641740ae69fb95f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-03 05:01:13.000000000","message":"Patch Set 1: -Code-Review -Workflow\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"fce3a1f35dcdb7f8ce67c88a6b9e5ec4b8db092a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 05:31:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/63cc1b3a5d0b4ad7bbf289f1cf761271 : SUCCESS in 19m 53s","accounts_in_message":[],"_revision_number":1},{"id":"eda363b54d7f19270fc71ba6da84d3c8e0f82cf1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 06:20:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae97045ab4094a0ebf447ce47bedee85 : SUCCESS in 3m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7489e4b224194805afa5347a2a11d137 : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e273f8edd1b74225bee6fa96b7a5ff0b : SUCCESS in 4m 09s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fa2d472d77b14f1e8ec733c309fb8497 : SUCCESS in 5m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0999c2cdfa5d4881b9e0e6eea2959269 : SUCCESS in 3m 40s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/7a2894623b4e45369ff8e182fcf03941 : SUCCESS in 7m 05s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8bc9c1a397ef4657bc7a03bc461be84f : SUCCESS in 9m 25s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/35278c74d1994a12bb1e464f4e755e62 : SUCCESS in 7m 37s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/08a9461dc36b46b2b0672b516f075ed0 : SUCCESS in 6m 55s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/a950a2151d014fdaa657831752d5c8ba : SUCCESS in 8m 58s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e74b7202fa754ac4b81f46ed7f55e487 : SUCCESS in 9m 42s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/8c21f23333ab4d66a7c61162e49b278c : SUCCESS in 9m 21s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/74f99e2fd1754f67a4e8e4f10e16ed28 : SUCCESS in 10m 49s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/7c34c2eacad64f16833ccb33db8f6bc0 : SUCCESS in 21m 46s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/2e70c43d6f7c4bb4aca6abaf10c95d2d : SUCCESS in 15m 02s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d3c40e08fe474f5e8eb756088f8b37c4 : SUCCESS in 4m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9ecad406f0404ca5a08d625c695b135e : SUCCESS in 14m 07s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/0252750195494d7c8c7764287cc7ed77 : SUCCESS in 5m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c018d9f3541940729954c6481ec46b99 : SUCCESS in 5m 44s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/cd9486bbef1a4bdb9e022eb52a9fc8f9 : SUCCESS in 12m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9ef6a2a00563454bb72ad3ee3dd064ec : SUCCESS in 39m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/83201c345651403bb4e54718f4332b55 : SUCCESS in 1h 15m 50s","accounts_in_message":[],"_revision_number":1},{"id":"5d4b39b642f8ddaa64c64701742942f76f613a6b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-03 15:19:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0db0b7f06138573035677c94e89c78daf9399d48","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 15:22:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a980009ac66843db1bf58ece2032bb6d38f16fcc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 18:27:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56bd2ddd86904f42842f48b69a6e875d : SUCCESS in 6m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/148e1dea683c4d9baedfd50b3fa75df1 : SUCCESS in 5m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/047fa6bddab347c8b0e40989bfbf480c : SUCCESS in 7m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c78cd230b3fe430db7f36e158c4faa3e : SUCCESS in 4m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/70e78c9571b24d718a863d5a6c288ff1 : SUCCESS in 4m 05s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9361d89e4acd4904aec302f4b76d6b6c : SUCCESS in 10m 33s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5dbba2eabd714a9d81728f8d16263562 : SUCCESS in 12m 04s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/63e3e02d2ba04e74bf3f8fd8f48428ea : SUCCESS in 11m 39s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/98b9b2f996da424f80e11bda7042023b : SUCCESS in 6m 45s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/2f29821971c64bfa81a916d13b3c1912 : SUCCESS in 13m 19s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8f7d893fe587417da8ff926e9cffd13f : SUCCESS in 10m 20s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/7b6a0007a09b430c8c4f955fa7bdc0b9 : SUCCESS in 12m 08s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/95568af989954033a6c85c6361c673a8 : SUCCESS in 15m 09s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/e96c930f091a4203a39906dbf9d55079 : SUCCESS in 27m 03s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/e983d72e40654030869436a8862c22d5 : SUCCESS in 12m 55s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4f6c3e683a3f4d95a7926c48c3e3974c : SUCCESS in 4m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/24ccaf6e18ff4c50a30236a1f4e72198 : SUCCESS in 21m 13s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/8afaf541f808496e88a2e2fe2064df9b : SUCCESS in 6m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1cb42727b0274db3905f165884a1f0f6 : SUCCESS in 8m 25s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ea94218646ed46cc957b6cbd9c6a5c42 : SUCCESS in 12m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cdf5f92ed3974cdfbdfd0f40f31989a8 : SUCCESS in 56m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0129dbe206c94682a712cc129e937c6b : SUCCESS in 1h 26m 51s","accounts_in_message":[],"_revision_number":1},{"id":"7456cb65c7e004300393bf81013fb99d49c6812d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 18:27:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"160eb878e1d937e131a9757d6e1523a7cd124c53","revisions":{"160eb878e1d937e131a9757d6e1523a7cd124c53":{"kind":"REWORK","_number":1,"created":"2022-01-24 11:52:54.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/49/826049/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/49/826049/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/49/826049/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/49/826049/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/49/826049/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/49/826049/1"}}},"commit":{"parents":[{"commit":"f8214a80076fd47ba9d03477af12b9314d7734f1","subject":"Move upper-constraints for validations-libs to 1.5.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f8214a80076fd47ba9d03477af12b9314d7734f1"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-01-24 11:52:53.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-01-24 11:52:53.000000000","tz":0},"subject":"update constraint for sushy to new release 3.4.5","message":"update constraint for sushy to new release 3.4.5\n\nmeta: version: 3.4.5\nmeta: diff-start: -\nmeta: series: victoria\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Ib27d51485700555369a9bf51bafbe3f1da9a5b72\nmeta: release:Code-Review+1: Dmitry Tantsur \u003cdtantsur@protonmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Ie891008b10f90f83332ae65db5c5c95c6a48e663\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/160eb878e1d937e131a9757d6e1523a7cd124c53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/160eb878e1d937e131a9757d6e1523a7cd124c53"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
