)]}'
{"id":"openstack%2Frequirements~826070","triplet_id":"openstack%2Frequirements~master~I8345bdd1271cd61cd03e45f2e9e9b8a6f09877e0","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I8345bdd1271cd61cd03e45f2e9e9b8a6f09877e0","subject":"update constraint for python-manilaclient to new release 3.2.0","status":"MERGED","created":"2022-01-24 11:54:08.000000000","updated":"2022-01-24 19:46:27.000000000","submitted":"2022-01-24 19:44:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"826070-new-release","meta_rev_id":"f839c7a1ae3cf4d2b1146b05727b868665d193e1","_number":826070,"virtual_id_number":826070,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-24 19:44:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-01-24 14:13:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-01-24 15:01:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-01-24 14:13:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-24 13:28:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-24 14:13:02.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2022-01-24 15:01:47.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"f1f3a93e1168ae98638d2922cb67826ab441fef0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-01-24 11:54:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7dbfe4a8e3ad0adcc623bc5f1c87f83ac1d46f47","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 13:28:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ffdac6eedb8446baec613c4e5df3ea0 : SUCCESS in 4m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e8c6f2d1a9b433ca8857ee481e7ad2d : SUCCESS in 3m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9b401440f1664020b03b03b138cdd7c0 : SUCCESS in 4m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/36c43e8d6caa4b0ba0ddff63ec2d86ab : SUCCESS in 4m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4bbec18098d84f96bf4ba7b6fbf7e004 : SUCCESS in 3m 52s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/59b63c9a3b054a3d9f4da06e559b4b6c : SUCCESS in 4m 05s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1fd5f0006d504ab9ae7fb7aad469a285 : SUCCESS in 10m 16s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/97f619fd27194a508e94b07de3ed02a0 : SUCCESS in 10m 55s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/b016f4aed7fb4b8e9c07b53f7a118e88 : SUCCESS in 5m 43s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/f421b0d8482b4a0a9b2d9f66acd61d97 : SUCCESS in 10m 22s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/95fdd6d6f9514d149e240cd5240e39c8 : SUCCESS in 6m 06s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/4456c0ec589d434799e2b81f03b0f024 : SUCCESS in 10m 26s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/72b32dc94c4c4d42b4cc627ac79eab47 : SUCCESS in 8m 41s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/47b706f0a25e4ce6bb34dd962bfe0d3c : SUCCESS in 8m 24s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/de703056518c4b3ca02f862fa1633ac3 : SUCCESS in 19m 02s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/068a9c37c60f47f18c6e21059202c9c9 : SUCCESS in 4m 12s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/cce027b46c5d49f9aba57b0f899bf371 : SUCCESS in 15m 43s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/b585d905a33f4d0799be4f30321feb83 : SUCCESS in 5m 41s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/abcee20e8f6d4c74800240bca0d8a7db : SUCCESS in 25m 15s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/4844479970af458188afee35cbbd30bb : SUCCESS in 11m 51s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/031f51a3feea42d7b11733b36c31f6cb : SUCCESS in 5m 17s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/bd562161619746ba9073c90c05a09e8c : SUCCESS in 3m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ae8fbf38d6ea454dbcf55f36ee172b9b : SUCCESS in 22m 53s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/a477651c256f4133a5088351a91f02b1 : SUCCESS in 5m 55s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/c66713ca2f04402db84962f3df1159bf : SUCCESS in 5m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/678ff5bfd11447e485168c6987565961 : SUCCESS in 6m 44s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/cea7a87baa074b69b0f0ca2ebd2de682 : SUCCESS in 11m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d45dee46ad334fcf9cdcaaa66ce1e4cd : SUCCESS in 47m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a8f234b6e78a4d48984ba425eb5b5ec8 : SUCCESS in 1h 24m 29s","accounts_in_message":[],"_revision_number":1},{"id":"346d3ee9541b106ac8ce4c76c16b50d49e3b08d8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 13:35:23.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/7966fe7640c44ac0b10f8886836cb648 : FAILURE in 18m 25s","accounts_in_message":[],"_revision_number":1},{"id":"927a96d4bb6272b559f52036dd28215c6ae71995","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-01-24 14:13:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6479bc0040661fd8578c2fe69baa837643388f4f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 14:15:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8f55a4553e7ae951eae7ff4d72eac55876684571","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-01-24 15:01:47.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"458afb61e4b6584beae638425e21457531f51905","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 19:44:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a025dc81774d4186bb6e883770e859ae : SUCCESS in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e126ecf1bd141fc80cc79f4a04b1dbf : SUCCESS in 5m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f2c6dd5e60874710bf4a48935e0cb9d6 : SUCCESS in 5m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/07f30729d73b4078ae5ffc742bddc964 : SUCCESS in 5m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fd2f69fdcaee4da3bfd4fb0e970faf9c : SUCCESS in 5m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3dbd402baac94151a24b4d6ef17b5278 : SUCCESS in 5m 28s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/39cb1bf4faf944feaa1f00b84649ca76 : SUCCESS in 13m 20s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/06eb011f169445e98adfe96fbdac58ec : SUCCESS in 10m 45s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/0c3b24bd253e42518c5da9ae734b3ee2 : SUCCESS in 11m 20s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/bc40dfb3a44a4ed49f4cc304f9d1892f : SUCCESS in 6m 25s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/16445f793baf4a1297dff9bab3859bf8 : SUCCESS in 11m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0c3b0baeda2b4273a6a5024f84796f88 : SUCCESS in 9m 39s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/df5d1e5bae2749c78ad7008de2b4f6b9 : SUCCESS in 8m 59s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/a384c84a6c2e46a6977ae96c5816d9b6 : SUCCESS in 13m 14s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/339ed78a221046f6a9afd1fa65db84c3 : SUCCESS in 4m 39s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/16c83168db224b118f27080ed277e8f7 : SUCCESS in 16m 51s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/f1e065e091674f75b2adc2da068b6d2c : SUCCESS in 5m 55s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/6ae2792094874194abf7cff62902e075 : SUCCESS in 26m 25s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/0ef6e6e2fc174c49ba810d940abd6c35 : SUCCESS in 12m 56s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/b2269467346b4e6996a741bad8283408 : SUCCESS in 6m 54s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8025b5376a4e4ab4aa8a9354f521b2ef : SUCCESS in 4m 46s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d46f624a8af44225a8b7f66887704411 : SUCCESS in 18m 26s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/75cf83c7a17c47438fc358ac144f6dae : SUCCESS in 6m 53s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/fef2ef5e1fe848d4900c25bcb64c93e9 : SUCCESS in 7m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b5452409dfc1447a86352edc49673720 : SUCCESS in 6m 59s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/3a4bf6f9ab3a4217bab3ccdc54416032 : SUCCESS in 9m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bb16b6cd326e4cff8c7bb293ab82dcc0 : SUCCESS in 44m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8fba95a0cb894699a173b8fc4ad10668 : SUCCESS in 1h 19m 21s","accounts_in_message":[],"_revision_number":1},{"id":"631e85e98c940ce04f28146f17d122099669b542","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 19:44:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f839c7a1ae3cf4d2b1146b05727b868665d193e1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 19:46:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81b2d223c4624566bcc8fd321c5cce48 : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"51744d605a243a710fce3a4f4f78a86554709550","revisions":{"51744d605a243a710fce3a4f4f78a86554709550":{"kind":"REWORK","_number":1,"created":"2022-01-24 11:54:08.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/70/826070/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/70/826070/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/826070/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/826070/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/826070/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/70/826070/1"}}},"commit":{"parents":[{"commit":"2075d89ac59f3a9362912fb62109c6e01bdfe12b","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2075d89ac59f3a9362912fb62109c6e01bdfe12b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-01-24 11:54:05.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-01-24 11:54:05.000000000","tz":0},"subject":"update constraint for python-manilaclient to new release 3.2.0","message":"update constraint for python-manilaclient to new release 3.2.0\n\nmeta: version: 3.2.0\nmeta: diff-start: -\nmeta: series: yoga\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\nmeta: release:Commit: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\nmeta: release:Change-Id: I449f5a921bd15b15fb4f323383662eb694492871\nmeta: release:Code-Review+1: Tom Barron \u003ctpb@dyncloud.net\u003e\nmeta: release:Code-Review+1: likui \u003clikui@yovole.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I8345bdd1271cd61cd03e45f2e9e9b8a6f09877e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/51744d605a243a710fce3a4f4f78a86554709550"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/51744d605a243a710fce3a4f4f78a86554709550"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
