)]}'
{"id":"openstack%2Frequirements~826076","triplet_id":"openstack%2Frequirements~master~Ie77f94b42c0c9d54795e649ca92e80eb4a6ffad5","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ie77f94b42c0c9d54795e649ca92e80eb4a6ffad5","subject":"update constraint for neutron-lib to new release 2.19.0","status":"MERGED","created":"2022-01-24 12:12:38.000000000","updated":"2022-01-28 15:19:26.000000000","submitted":"2022-01-28 15:15:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"826076-new-release","meta_rev_id":"bacbde16f54f8bd7a8cdaaea4a368602ee2d4130","_number":826076,"virtual_id_number":826076,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-28 15:15:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-01-28 13:42:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-01-28 13:42:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}],"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-24 13:36:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-24 14:12:29.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"CC"},{"updated":"2022-01-28 05:54:22.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2022-01-28 13:42:54.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"b152df3efd2128238755d86da59e657b7548b3d9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-01-24 12:12:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3ca7ab1c419f2d6d04d7a99f3e5b9baae1d9e383","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 13:36:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04f51b7e31584dfaaf930b2a7f393414 : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bdad232ad131480dbb11efabd8a9d325 : SUCCESS in 3m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/da14eabbc223467c9ad56dc13f4e411a : SUCCESS in 4m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3470310de3fd4dd6b8bb2d7fcbba059d : SUCCESS in 4m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f6f6c3d7efda45f5a3d4e2f98505db7c : SUCCESS in 3m 35s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/49efd1e838b84011b64c0ef2703ed069 : SUCCESS in 3m 33s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/385ab81fef384fd183d0cc51c88b7980 : SUCCESS in 8m 49s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2ca77b01681b42b885b270fd816e80e6 : SUCCESS in 8m 11s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/a7a00dcb7c274a6fad85bf424a18a147 : SUCCESS in 4m 59s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/72db39a518c74cf6b877a85e52ccb26e : SUCCESS in 8m 38s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/459bdd76d8694c4ea2c62205e14c0f60 : SUCCESS in 4m 59s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/dc5c1283eb5b4bb6b3da09820d2fbbc9 : SUCCESS in 10m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0bcca2ba3b024c058e56a761ea32e148 : SUCCESS in 8m 53s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/8504ec46a8314e83aa73602a8403ac5a : SUCCESS in 8m 54s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/691edc93dfb14572970941d0b7dbd500 : SUCCESS in 13m 50s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/981ce2357da044ceba552bb6018aae8b : FAILURE in 2m 40s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/6474c4b1651741aea2fe8db54e44f13a : SUCCESS in 17m 58s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/087a13e8e65f40f1adc892720bcb82d0 : SUCCESS in 6m 05s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/9552fe73c0954fea991577ce5c0a05c3 : SUCCESS in 27m 23s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/179f7f620a374a95b9421cb466edb1cf : SUCCESS in 13m 03s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/e9c2e9a631244ece917335f3cd9942ea : SUCCESS in 5m 30s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/29399857c02f4334984ff9ec849c27ef : SUCCESS in 4m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/217954515b5c402c81554cf0cf1dca40 : SUCCESS in 24m 24s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/e35a6135e7e34140b92ba826c33ea83e : SUCCESS in 7m 01s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/d40d750251e4400e9b89a506806782ac : SUCCESS in 6m 15s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/027cd33784f84642bc48fcb448572ead : SUCCESS in 6m 14s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/7236fee0fe044a3fad6c5a351c78e652 : SUCCESS in 14m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f2d640a16b5c4689b0757a81d0c3d2a2 : SUCCESS in 52m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4bbb4e7321fa4c899233200a8acad581 : SUCCESS in 1h 21m 07s","accounts_in_message":[],"_revision_number":1},{"id":"4b29d56dae22f94b8fb69c6fbe47fbc61672b59a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 14:05:28.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/8d033e6b56a041ddaaff654bdeed93bb : FAILURE in 10m 51s","accounts_in_message":[],"_revision_number":1},{"id":"f13dbea8529bbddd7e88b9e0fb17935a07c08720","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2022-01-24 14:12:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7050cbc9aaa7d17648965ccf217a5049cc5dc11f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-01-28 05:54:22.000000000","message":"Patch Set 1:\n\nthat doesn\u0027t make sense to me, we are installing a version \u003e\u003d1.13.0","accounts_in_message":[],"_revision_number":1},{"id":"4300aab0086453c2725b80576bfa2ea7cf0cc50e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-01-28 05:54:27.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6fcc7fe13b5333344ed210cd2ceec754262036d3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 06:10:56.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/0988c97164a04bfaab5370f607fecedf : FAILURE in 10m 24s","accounts_in_message":[],"_revision_number":1},{"id":"d8a701a05546968787246cff39f5dbd9410d5daf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 07:15:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3fc2ccfbd764609b5f2cf45deecb8e5 : SUCCESS in 5m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e3b348ab85b4b818caccbf463da7a05 : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/61b785bc76df41f0b0f1b2b488e0146b : SUCCESS in 7m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4c11687b916c4e1ab675a59fd2d4b3a4 : SUCCESS in 5m 14s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/413c17a7a5b64579946e8c9664a7be84 : SUCCESS in 5m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/95b88d2206074a6ab9f53936180fc28d : SUCCESS in 5m 51s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/282f570601da4f0885a0fd76458d384f : SUCCESS in 10m 18s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/38f52728f8734e55b5f5e74b0fde4039 : SUCCESS in 10m 11s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/ecdbdc5ff1d3423a857f0ad9367b7b89 : SUCCESS in 5m 24s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/508ef32ce42b4cd3b881765f950338fb : SUCCESS in 11m 44s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/42fa03e1e28944f180837bc65b52bf85 : SUCCESS in 5m 17s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/7ce7b5dca3c74203b715d5f5c8d68a2a : SUCCESS in 5m 49s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/d75a2e33379e45a3ae18a47fb2cddd6e : SUCCESS in 10m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/eaccde7caffe4bb4a6557f96af101a0e : SUCCESS in 9m 44s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/a5a5be60f9364686bd1c794bef35545a : SUCCESS in 9m 10s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/5a9e0f36d1354e92a7338e9bce08ce04 : SUCCESS in 13m 00s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/62e193ad536b486db5728c8d28d7a1c3 : SUCCESS in 4m 15s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/f11df98b49f24715ac96d5ac7ca0ceb9 : SUCCESS in 17m 01s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/89d10247ff084e2baa778fd7e8c32a1c : SUCCESS in 5m 14s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/7eae1843bbbb4da3aa34c14697d0465c : SUCCESS in 22m 37s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/b2b11fe4b1b342ab8917a170af8ef6bc : SUCCESS in 15m 20s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/2dc790654d0b485eb6902b3993423016 : SUCCESS in 5m 14s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/31a9eaa42e1f4a4cacc279bb95afbc1d : SUCCESS in 4m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fbf23b8f283448a7a2d38a0a0269c37d : SUCCESS in 27m 05s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/de9c34cd96cd40419f81c9a67493109e : SUCCESS in 6m 15s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/0dc8f824bf7e4056a5be026909aaa724 : SUCCESS in 6m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/68cf7d076f6545208a061e9124a5edd7 : SUCCESS in 5m 50s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/f5759d2c136949e99edc41fbb43e15c0 : SUCCESS in 14m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3011b63bfee94f6da1e2971e1eabd0ee : SUCCESS in 41m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/343f1b73fc344db3ad84eba12a8bf748 : SUCCESS in 1h 17m 27s","accounts_in_message":[],"_revision_number":1},{"id":"eba2e750830d25d9961f47f4efbe5357cfddabf6","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-01-28 13:42:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e4fd12e177156c0260ff0024ef193b7179a75680","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 13:44:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2b88dac01d2fb12938cd07045cb42db5cc23fc51","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 15:15:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10e1ed4347654230805fc4abe04de729 : SUCCESS in 5m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bff549c5c1ff42048fbb6fd419cd3fde : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/95d61de9ab224cd18627b09151d46f45 : SUCCESS in 5m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e007930ac1f345e8b1d79aa28de88481 : SUCCESS in 4m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/01923dcea43744a9b83454dac957abaa : SUCCESS in 5m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a7ee2c8e99d64db0ac39be4a6a3fd6f9 : SUCCESS in 4m 50s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2f69eca82fee43e38a806b1705095d1f : SUCCESS in 9m 37s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b86501f208384f16bfc6dcbf685c38cd : SUCCESS in 9m 25s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/3215316a1bed430ab6bbd87a3ee373fe : SUCCESS in 9m 27s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/3d402e2154f848bc90d29254913c3e37 : SUCCESS in 6m 04s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/ae62df56fbb043d9810ad5a9a2e946be : SUCCESS in 5m 54s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/6a481fe421d24afaaefcc389d1fa3a4f : SUCCESS in 11m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ce3d9e250ca14ace95473c88dcb25496 : SUCCESS in 10m 35s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/5d898b909a8e4a3081550332c315ab57 : SUCCESS in 9m 35s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/415d74fdd52441569b632fd42a11b55e : SUCCESS in 17m 09s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/261be22f4d9e4f0bbdcb8b53fed4baac : SUCCESS in 5m 34s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/132b00595c394bc9a47a235e9dc7176e : SUCCESS in 19m 51s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/b535c28ef9ba45dab969734ce40ca69b : SUCCESS in 6m 57s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/eddf12d363d94a5ca6e7738c02cacf7a : SUCCESS in 21m 14s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/3e011bb564334f759f4c0008197493fa : SUCCESS in 13m 09s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/cddc4f2b25914fe2a899c3314205d8d6 : SUCCESS in 7m 05s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3bdc6cb52fda4fb6ba091e952888586e : SUCCESS in 5m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a83fa03c78a04584a239c0be88665f63 : SUCCESS in 19m 19s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/fd32cf481d6640cda0f74440f7dde6bd : SUCCESS in 7m 47s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/1818cff7051f474fbc93d5217d9e1308 : SUCCESS in 6m 07s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bc6989b85e1445e0a62508846ee42ca2 : SUCCESS in 8m 41s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/7b269def0e8d4786b5e364855a6cb688 : SUCCESS in 14m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e5764fc3f13b44d58ea2be3f6ac09f00 : SUCCESS in 43m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fc6aa49ed23d4ef1aef89ab3f9433635 : SUCCESS in 1h 23m 14s","accounts_in_message":[],"_revision_number":1},{"id":"3dae3ffb28ce72e73ccead90ede771e49e2180b4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 15:15:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"bacbde16f54f8bd7a8cdaaea4a368602ee2d4130","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 15:19:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f57db9575084d018677201e35fd43a6 : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"386caa4b7dc8336e209c5dd70022d1aa1458f3ac","revisions":{"386caa4b7dc8336e209c5dd70022d1aa1458f3ac":{"kind":"REWORK","_number":1,"created":"2022-01-24 12:12:38.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/76/826076/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/76/826076/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/826076/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/826076/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/826076/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/76/826076/1"}}},"commit":{"parents":[{"commit":"2075d89ac59f3a9362912fb62109c6e01bdfe12b","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2075d89ac59f3a9362912fb62109c6e01bdfe12b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-01-24 12:12:37.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-01-24 12:12:37.000000000","tz":0},"subject":"update constraint for neutron-lib to new release 2.19.0","message":"update constraint for neutron-lib to new release 2.19.0\n\nmeta: version: 2.19.0\nmeta: diff-start: -\nmeta: series: yoga\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nmeta: release:Commit: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nmeta: release:Change-Id: Ib3b01b2d34f023774863c0577fc226b2fbf4a764\nmeta: release:Code-Review+1: Rodolfo Alonso \u003cralonsoh@redhat.com\u003e\nmeta: release:Code-Review+1: Lajos Katona \u003ckatonalala@gmail.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Ie77f94b42c0c9d54795e649ca92e80eb4a6ffad5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/386caa4b7dc8336e209c5dd70022d1aa1458f3ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/386caa4b7dc8336e209c5dd70022d1aa1458f3ac"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
