)]}'
{"id":"openstack%2Frequirements~826098","triplet_id":"openstack%2Frequirements~stable%2Fwallaby~Idc012bb1e79c3a9e3859b0cc30b46fe873f26fb9","project":"openstack/requirements","branch":"stable/wallaby","topic":"new-release","hashtags":[],"change_id":"Idc012bb1e79c3a9e3859b0cc30b46fe873f26fb9","subject":"update constraint for os-ken to new release 1.4.1","status":"MERGED","created":"2022-01-24 15:03:20.000000000","updated":"2022-02-04 00:13:20.000000000","submitted":"2022-02-04 00:13:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"826098-new-release","meta_rev_id":"9e1b65a7c317123b3f0104b761b018b5b89a304f","_number":826098,"virtual_id_number":826098,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-04 00:13:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-02-03 15:18:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-01-24 15:03:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-02-03 15:18:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-24 15:03:43.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-01-24 15:17:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-24 16:35:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-28 05:56:56.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2022-02-03 15:18:58.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"db1e19214b89bd2c0c08651d3456550a9859e007","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-01-24 15:03:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0e6369ae1b9aad305afef1143fc16d1fd298c87c","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-01-24 15:03:43.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"22e4110e50ebe6855aecb6b866af3a8b9ae8c8fa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 15:17:27.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/88313d14508147b880caff9e14c8b915 : FAILURE in 9m 49s","accounts_in_message":[],"_revision_number":1},{"id":"fa16e58c03e2f0e413d6b2cd8f1965f80071595a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 16:35:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e32584fd357142df88abbd5afc84e864 : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a57cb9c6f58c49759fac9760be282139 : SUCCESS in 4m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/68146554266a4a5aa7e7de776a9ea41b : SUCCESS in 4m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a2e5844dee5e49e89551a39eeb0f47e9 : SUCCESS in 4m 25s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0b2e28508b7446179b32e1e649787562 : SUCCESS in 3m 52s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/213c2b580ab14eccad5f2869cc948571 : SUCCESS in 4m 16s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b3912b780aa24d7a94ab948a0384496b : SUCCESS in 8m 19s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/816c48774509461aa9e4c70f6cf1e054 : SUCCESS in 7m 09s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/5d8f24077c9f47c48dfe6b453c44f1a9 : SUCCESS in 10m 35s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/85366c50789e402da1f68496674d4e93 : SUCCESS in 5m 31s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/bdd11f687b9447bdbd71750553879a93 : SUCCESS in 12m 01s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9da52317316e48cfb55146cd0dd75161 : SUCCESS in 7m 47s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/3d84a5fc16614e0dac4147f58c5a2b2e : SUCCESS in 10m 55s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/24d93cc99e234235b80e5a9745bb42ae : SUCCESS in 12m 57s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/0a6c4bf31c9942219279356f993affe0 : SUCCESS in 5m 22s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/31e3878327fd4cd89d0f09076bdfc9c5 : SUCCESS in 33m 08s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/bb987dd430f34529a0aae088cd123a49 : SUCCESS in 11m 04s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e6bbcef17e504ff783ecef40d24c3561 : SUCCESS in 3m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6b327149b10e4963aa9ceb8e53036a5c : SUCCESS in 19m 48s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/20f0cd622c714addb9015a03ec79b656 : SUCCESS in 5m 07s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/63c3a755184b4bf2af2f3edeb0a46d2f : SUCCESS in 7m 05s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/948a15334d5b4f7d97b34ec8f77b3260 : SUCCESS in 11m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0ab5e1d072254b0e82c984dc11171655 : FAILURE in 41m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/91e183e9d990448dba033a7201e202c0 : SUCCESS in 1h 29m 57s","accounts_in_message":[],"_revision_number":1},{"id":"743f4878889737ae5faa0e15bb77afbea9a4cfd7","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-01-28 05:56:56.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c9fa47e9fe201f849cc51c0e43734cf273dae464","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 06:15:52.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/475546a97be446a2a7ce3a89783499da : FAILURE in 15m 56s","accounts_in_message":[],"_revision_number":1},{"id":"791f9834e85f52eea7e9ee0978f5caa94b96fa50","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 07:26:33.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c29422ab48040d2b13b23c62e6b9c3d : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9ae36b3240544521ac4790c591301d02 : SUCCESS in 3m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/32084da7cd8641bba24a905583834b63 : SUCCESS in 5m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/06d9d147a7964952ab99ddd1724c5e6c : SUCCESS in 3m 56s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/de57582ce22a4f21a711ea3d97ccf2d1 : SUCCESS in 3m 52s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/df12cdc209d24d9982248cdd6b37353c : SUCCESS in 3m 36s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/352f878d97f34a9cbd7141dbe37d5ece : SUCCESS in 9m 02s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/504a9b6c3cde4b4b88b05ff05c1f2cdf : SUCCESS in 9m 35s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/f6927ae0ae0e45f4832ab1a2ebe5a418 : SUCCESS in 11m 46s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/274a3000e03c498d87a1944eb2590d32 : SUCCESS in 6m 45s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/8522b3dfd1204b60b8a751e8b101d28c : SUCCESS in 10m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/95f4b296593b49759cc7de810376b05e : SUCCESS in 9m 41s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/801e6d6bfa834506bf850f5222f0fcc9 : SUCCESS in 8m 23s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/659fb7fdc5ed4093844f35d27be9084a : SUCCESS in 13m 51s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/75c5656e1686430f9e6c51c8bc859b6d : SUCCESS in 4m 45s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/4af8fcaf2db04ccd89029d694f5febf9 : SUCCESS in 22m 45s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/58132f80701d400eb68090b1e20f039f : SUCCESS in 14m 28s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/35d9a185647645b68341efbec2499c2f : SUCCESS in 3m 43s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f0b67304245347a4af009026cef161f4 : SUCCESS in 19m 05s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/7e2c85093b2e45c2830177747a02cdf3 : SUCCESS in 6m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/50af79c473c944ca84ec2d9ca44481ec : SUCCESS in 7m 03s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/13a337176e3f43ad93acda0654f1d596 : SUCCESS in 9m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9f0bea226fd3458ab9032a1fb9e02d24 : FAILURE in 47m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/071369b9536a49ebad40b7b13e974625 : SUCCESS in 1h 23m 48s","accounts_in_message":[],"_revision_number":1},{"id":"6517f69e2665f4170082941927041b8b8e7e4ef1","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-03 05:01:09.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b531f34b5275923b9881e9569b54d6ac37bff642","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 05:19:55.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/b96a8242a4d74c0aa0b9d42639dd5753 : FAILURE in 15m 51s","accounts_in_message":[],"_revision_number":1},{"id":"198c98a9d9527acfa2f269d85426fe16d74e4331","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 06:22:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b5728d8850a44c7833a6b737fcf35bd : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/32ccf2c496f247cf967f9a8320c61f91 : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a0321230b8a74ad38a18de45d7e021f1 : SUCCESS in 5m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e3225911ef10493083e3928a4703c887 : SUCCESS in 5m 44s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2b21f3971e914d31ada28444c3141fb4 : SUCCESS in 4m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/32b43df246144778bb2af0fe06b378f2 : SUCCESS in 5m 14s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/64d703e8aec9416f8702dc49c26b878e : SUCCESS in 10m 27s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e7984383eba949209127c64c3007e469 : SUCCESS in 8m 59s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/5d63241f1d974523a26efcb796a5ca0d : SUCCESS in 10m 37s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/1bd60c3761a74ba2a2c328374e6a400d : SUCCESS in 6m 28s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ef82dbac9014492caf8d08ddc7c79274 : SUCCESS in 10m 12s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8f02e067987a494f8814da118c00d475 : SUCCESS in 9m 12s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/3ca66f2c2f1745739f9303c11b5a2c2d : SUCCESS in 10m 03s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/97048a901b944ca1b27ea0f11b5b6759 : SUCCESS in 18m 25s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/ae3b084a1745459eacd1836603eaf90a : SUCCESS in 5m 53s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/85f9ad4bdcce4282ab86315852bfe1f9 : SUCCESS in 26m 25s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/a6e27659e1d640b3bf1aee445b353fcd : SUCCESS in 14m 24s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/16992ccc03e9433dbdd9f192c359f9bd : SUCCESS in 4m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6156134d03c24ebe9d36164c7939b470 : SUCCESS in 21m 04s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/5d34560c437b4a5fa10454bbfe9b9635 : SUCCESS in 6m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ae85e2c55b0746799d3afce2a8f5a14b : SUCCESS in 7m 51s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/87e37be0f1494e2d858aaf4b4b0a4ad3 : SUCCESS in 15m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b7be02838b794be2a375b2c57258922e : SUCCESS in 50m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0881363be05e40cfa48aa3ba30a3a368 : SUCCESS in 1h 18m 38s","accounts_in_message":[],"_revision_number":1},{"id":"3296037f2e166832d2deefd44fadac08fca7f91a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-03 15:18:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"119f3a6897a4d2c89d8226b8902f3b5fd3a53555","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 15:20:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b7046201ac1ebacc955f42e639974b47b8f66694","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 16:37:19.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbba18d9d82d4617923e2876ca92c279 : SUCCESS in 6m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f6813cb619464618a5462a3136a6f245 : SUCCESS in 5m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/10730c56f6de45bcbbbefa83c5bf1ed3 : SUCCESS in 6m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c03cda193704497a98f66d07cea4739d : SUCCESS in 6m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a9bf9eca4dbf43dc9f803d00f2183d1a : SUCCESS in 5m 25s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/67eae962475144a5ada014fc3ed7a40b : SUCCESS in 10m 35s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e33bb87c734744a09bfb686ea61f2fcd : SUCCESS in 12m 29s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/fa9ba59517ec46709b0f51dbc78c40b2 : SUCCESS in 11m 31s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/26c59d9821aa467aa075f6c7cb5a3aef : SUCCESS in 6m 53s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/2e14f06d9b274bb3aa58be4dfa11755e : SUCCESS in 12m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/161f312ba82f4d4db01a08fbe3ad0291 : SUCCESS in 9m 17s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f87a7a0763d84ae795be2a189db0eb3e : SUCCESS in 11m 10s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/8e05ecff7a304595b786ff663bbcf21b : SUCCESS in 20m 16s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/7d988ca8d72340c9ac7711cba427a003 : SUCCESS in 6m 59s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/fb9ddb41b2a34c2fb6e332851b2ed37e : SUCCESS in 26m 31s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/7bdf6fcebf82466a8a58504ca2ef7141 : SUCCESS in 14m 37s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c45873848f754e6990da8f2bb7347a9e : SUCCESS in 4m 56s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2c55fb3834f64820bb98928dbbc02da2 : SUCCESS in 25m 51s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/f9c3cbcb4de3401d8ee2cade3d787737 : SUCCESS in 6m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6f2cbef3ccea4937b7e497b1249e32ce : SUCCESS in 9m 45s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/368e1aec7ea042b7b86d8f8d1eaa5722 : SUCCESS in 19m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/68a642ed5bda4352a314da8dd0daf217 : SUCCESS in 56m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/941c4983382f4ab1b4ab49c8f9e4cd92 : FAILURE in 1h 14m 28s","accounts_in_message":[],"_revision_number":1},{"id":"f6e67cbc39edc1ff4aed9cc658b5a505265075ff","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-03 17:28:53.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9fcb6e9da4041815799b7b1bbb04076ee073f5ee","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 18:58:37.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/97ea807ce6ed4a6b8c08187bfd33f1f4 : FAILURE in 20m 42s","accounts_in_message":[],"_revision_number":1},{"id":"e9602f2b6220cadb9c8d188a02a2956044c67877","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 19:52:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92ce44fdb3fd4de5b64f2c7449a597cf : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/11361755c12c4548b7d5ded94a65bcb1 : SUCCESS in 5m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e5220bc374ca4db38cf596738df6b00b : SUCCESS in 5m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/58f29596cc9545639cea72dcfb06d868 : SUCCESS in 6m 53s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3ef06eda561e4a12b5190a2d8ec89446 : SUCCESS in 5m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3d2f5b35a4394f7cb8866a610dcb595a : SUCCESS in 6m 19s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c7b86244421c43678e4a76d4f4cb6556 : SUCCESS in 8m 33s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/4c569f7924bf4d7886a2df5ddf6fa793 : SUCCESS in 11m 11s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8a5c4451b7524aff9db54b88fd9baaaa : SUCCESS in 10m 32s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/9fb48706573c4cab8a7ee415a38f8e96 : SUCCESS in 6m 13s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3f5d69d0ffc84a8db9dd9cd43801fbef : SUCCESS in 11m 57s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/13402391b8ce4257bd68ff066427bac6 : SUCCESS in 9m 00s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/68efcf000a6c4207a2bcd2a01295aaed : SUCCESS in 9m 03s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/1c0503f9eab44b3b97f97dab94c6d8b3 : SUCCESS in 14m 39s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/a59ccc708b184332bebbbb762f996411 : SUCCESS in 7m 42s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/09e1607dcdcb48f5b501839d9bb7f7fe : SUCCESS in 35m 12s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/00b9b84f02b14d2a9fc91e8ebd0b2278 : SUCCESS in 13m 14s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/69c364f223604c26bee587eba898a7a3 : SUCCESS in 6m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7410b8ee1ec04a12b57d6cb8a915665b : SUCCESS in 22m 21s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/e40ee35e625642ed9714d0629e53fb25 : SUCCESS in 7m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/87b04a6076984c868782345d2fbc8d08 : SUCCESS in 10m 21s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e7103c7a3eca4d90a6c70f30a2499b60 : SUCCESS in 13m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dfb7c63b025b4921808c53b6f33c28ae : SUCCESS in 53m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/eec7dfaec3d145c4aa8f12b55bf2a73d : SUCCESS in 1h 27m 03s","accounts_in_message":[],"_revision_number":1},{"id":"5442ed0a2274c8b5ee960934ade62925a9a769a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 19:53:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"00e66ff1ab036793e00f931aacfee2463ae2a2fd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 21:37:25.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/658808bd23bd486ebe0028863b9a4b6e : SUCCESS in 6m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/727916a720d848298306c2b6c60b76ac : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/69a3aeb25da841509b307ae8e559aa03 : SUCCESS in 4m 17s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6fe025b161464c03a250012c0f023f9b : SUCCESS in 4m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6fce16aa0ad94c9a83b882ab1faf8ec6 : SUCCESS in 3m 50s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1dd6ab159ecc49e894ded158719509af : SUCCESS in 8m 22s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/50cb76e854ef4e6b940de3a74829098b : SUCCESS in 8m 36s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b31583e41456459abc9e255c113766ac : SUCCESS in 8m 14s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/5cb9f305168348b1969324e5c9418346 : SUCCESS in 5m 27s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/a606158058dd4717992800f970c4db02 : SUCCESS in 10m 01s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ba8710a1f1214c6b894100c4b6c6b1db : SUCCESS in 7m 25s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/6e3099b66f7840f38592bca7d1321034 : SUCCESS in 9m 42s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/4c5316b1742a419b9d141842846d3523 : SUCCESS in 13m 07s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/b2b67122b0c9426e9fc74a4459389754 : SUCCESS in 5m 20s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/dd9c85349ed443da8f5e477b1abf51c4 : SUCCESS in 23m 54s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/27f916e1218e4fcabe1b8c68896dc5ee : SUCCESS in 11m 56s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/20bc1bbb0f624dfeaa9d98d179bd6fac : SUCCESS in 4m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4f4aeba0dee444ab9ee9740c43f7fe50 : SUCCESS in 18m 42s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/5d5130d29e134c9992420d4c07fa07c9 : SUCCESS in 5m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a2242576b3484bc187873136ca27d2a7 : SUCCESS in 7m 18s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/75f508c03f624862bd433b86329f6994 : POST_FAILURE in 17m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9b6e5d8398904b39a663842332a989f1 : SUCCESS in 41m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d0e4ea644bce432c9c08c68f7df94cfb : SUCCESS in 1h 38m 51s","accounts_in_message":[],"_revision_number":1},{"id":"e210dcabf4c778db2424621ee92e6940b213aec4","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-03 21:39:59.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"550748430ae8e4a8d4a1d1888ce2155873192c63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 22:58:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab2dd180315148748821981b76955574 : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/78a7f3af46454d0792cc8a9de1ac823a : SUCCESS in 4m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/77a6d4e6d94c4ec1b3867f54f6378e51 : SUCCESS in 4m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bb506515cc1a45049ddb1ecbce5f6c78 : SUCCESS in 5m 06s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b710a039569e4476b46d893d99c2c0c9 : SUCCESS in 4m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/223ccca4fd4d4a4784696416cb3f4a83 : SUCCESS in 4m 16s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/60755e976d2845a88ccd03728514a136 : SUCCESS in 9m 51s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a0631dcb4fd2441784c611ed048b5dd7 : SUCCESS in 10m 12s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e2fff780ea43482481836165fa94f49d : SUCCESS in 9m 58s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f8a3bb025f6347e28c812fbe6695c18a : SUCCESS in 5m 57s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d7753b477dc042ecbf6775fedcccaeb2 : SUCCESS in 10m 42s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b3ff1188944845fe921ad32df679b5e0 : SUCCESS in 10m 21s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/fde3bfda54ee4608923952d18ef611a9 : SUCCESS in 7m 27s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/290ad3fbe8994037914fa6d84b5ebbc9 : SUCCESS in 14m 12s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/1cd83e00f82a4bd1855b975406aef2d6 : SUCCESS in 5m 46s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/6a1e6d4bee78427da9bb7a528bbb0eb2 : SUCCESS in 29m 54s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/453d57d62c5d4492a229ff541572ae89 : SUCCESS in 12m 13s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a9e730c7b96d4a30a53bdbd932c7d44f : SUCCESS in 4m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9c27429a3a304640bea09a5ea6d9038a : SUCCESS in 19m 19s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/173e3bfb0e0245e09b25f25dd180c7f8 : SUCCESS in 5m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/18b54af697ad48b4b2d7ebc4aa0c65b0 : SUCCESS in 6m 45s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/f5f9a60e7414449b9ba2d3a54fa9477a : SUCCESS in 12m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/00589d78458645d5bacd9023e7ac3533 : SUCCESS in 55m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5726446b2a214ea1a0fd41312d0fe4e0 : SUCCESS in 1h 16m 38s","accounts_in_message":[],"_revision_number":1},{"id":"2422b776c8abec632b09d01aff5ce39d9f283bb0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 22:59:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c4c56e84fb70cf97e442986c87bf947ad64f6f8a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 23:04:02.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/74d1f7e9c10b47859ea771355cb2bb5a : FAILURE in 15m 54s","accounts_in_message":[],"_revision_number":1},{"id":"e79d46053a778be95cd6402847a93f34d075d46b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 00:13:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ab0ee13a35a4419a7704c03feef486f : SUCCESS in 3m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1a4f888933674570b47c8eb5a13a2e60 : SUCCESS in 3m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/465b51fd58744b78b9e5f369b22b5888 : SUCCESS in 5m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0deecb9888ff48e494dd263f4ba2be89 : SUCCESS in 4m 35s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/32d31750714549c58bed977d0da3f773 : SUCCESS in 4m 54s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/57475fc33c2740bf8b0e5e1a470c5996 : SUCCESS in 9m 10s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/062ff561168f41cdad0c6903d2a195ff : SUCCESS in 10m 05s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/971f7ca79916453fb1f2bca91eb27526 : SUCCESS in 9m 26s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/772e47c1b7d74b82b52439d540d57e4d : SUCCESS in 6m 21s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d4e1cca7878d42799d0dea91fedf1a23 : SUCCESS in 9m 17s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/65c82cdd2ca34cdf8fc9c7c5f57fbe27 : SUCCESS in 9m 24s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/65bc8ef9e2ce4777baa95d5c40e45254 : SUCCESS in 8m 56s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e707536e8aae4c0dbac74d200ab30762 : SUCCESS in 14m 08s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/2be19571ebdb4d67addcdfed488b9178 : SUCCESS in 5m 44s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/9dd388fb4a984b5d9b7a4c0959929e95 : SUCCESS in 23m 15s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/fecda360b37b4789992a9a7dd72e0289 : SUCCESS in 12m 43s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ee46708779ac4b53a0bc79ebb0f247cd : SUCCESS in 4m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a9c2b6629b8d4a84be09bb835a734c2b : SUCCESS in 16m 56s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/75ac7e38bf634cfbadc85a2ab36b594c : SUCCESS in 5m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2e1681be03f14294b0fce51f62bc846d : SUCCESS in 7m 02s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/fd1ab1d7fcd143fabc5431512bd5b25a : SUCCESS in 12m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/912b0c3d31c84d968908fce9db7929c3 : SUCCESS in 48m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7bb0acdb4ebd4edba49222ba9b93cb14 : SUCCESS in 1h 12m 35s","accounts_in_message":[],"_revision_number":1},{"id":"9e1b65a7c317123b3f0104b761b018b5b89a304f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 00:13:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9246471317e2f29cb7e216c0e0a4c0ffff68d83a","revisions":{"9246471317e2f29cb7e216c0e0a4c0ffff68d83a":{"kind":"REWORK","_number":1,"created":"2022-01-24 15:03:20.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/98/826098/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/98/826098/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/98/826098/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/98/826098/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/98/826098/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/98/826098/1"}}},"commit":{"parents":[{"commit":"0e67f1d4bee22d1d2d56ec24fe3282c9a5a7ffd2","subject":"Merge \"update constraint for oslo.policy to new release 3.7.1\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0e67f1d4bee22d1d2d56ec24fe3282c9a5a7ffd2"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-01-24 15:03:19.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-01-24 15:03:19.000000000","tz":0},"subject":"update constraint for os-ken to new release 1.4.1","message":"update constraint for os-ken to new release 1.4.1\n\nmeta: version: 1.4.1\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: elajkat \u003clajos.katona@est.tech\u003e\nmeta: release:Commit: elajkat \u003clajos.katona@est.tech\u003e\nmeta: release:Change-Id: I8ef64f405d0e9588d49c99e6a955ddd88f39d521\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: Idc012bb1e79c3a9e3859b0cc30b46fe873f26fb9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9246471317e2f29cb7e216c0e0a4c0ffff68d83a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9246471317e2f29cb7e216c0e0a4c0ffff68d83a"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
