)]}'
{"id":"openstack%2Frequirements~826318","triplet_id":"openstack%2Frequirements~master~I3b0bb9da60b3d1e4317e804251fc2540b3120cd2","project":"openstack/requirements","branch":"master","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2022-01-25 20:11:08.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3b0bb9da60b3d1e4317e804251fc2540b3120cd2","subject":"update fasteners to 0.17.3","status":"MERGED","created":"2022-01-25 16:13:07.000000000","updated":"2022-01-25 20:16:43.000000000","submitted":"2022-01-25 20:11:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"826318","meta_rev_id":"c2c4a237c664a174969e6607e44544c5329b89d1","_number":826318,"virtual_id_number":826318,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-25 20:11:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-01-25 18:04:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-01-25 17:33:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":1,"date":"2022-01-25 17:49:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-01-25 18:04:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-25 16:48:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-25 17:33:33.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-01-25 17:49:07.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2022-01-25 18:00:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-25 18:04:16.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"e5ec66564b8ed67f99fc28522cb3172dd0cdfb4d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-01-25 16:13:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bd28a9c3608848c82e23288612d4a6acddca319a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 16:48:50.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/033f398a532942b7ab5474dba03ef569 : FAILURE in 24m 35s","accounts_in_message":[],"_revision_number":1},{"id":"7a3269e7c5ed81f36da1681d60968d4e5f0be55d","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-01-25 17:33:33.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"68f9307fc8a89dc488e6629f0506ea47e5d1275b","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-01-25 17:49:07.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"812d891feeabc43896a441ec55f105084867b825","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 18:00:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/513df3538d5b47d9b055838191588d19 : SUCCESS in 6m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44e179d63a79421d92ddfad5e04245b1 : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/960c9ae90d75430e8890d41f3e0e8ce5 : SUCCESS in 4m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6ba197321edc492d945349f9845cf949 : SUCCESS in 5m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1456d2e486e34129bccfa73a0c863d94 : SUCCESS in 4m 47s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/77ff57d9c0e24a9c8df7de6deacde844 : SUCCESS in 5m 00s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/43cdeaa4275644eca96a3460a6aa6b7f : SUCCESS in 11m 24s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9a4997a0270a44d0a35639599418733a : SUCCESS in 12m 37s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/6fd70064d096478b9df8b60e7814bc8d : SUCCESS in 6m 33s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/0907fe1d974343c48632cb3284e12cc7 : SUCCESS in 13m 01s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/34492f9ca1a1448691dff27c3b0d0739 : SUCCESS in 5m 34s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/f27a72ecef7b47a2acd1b0930b9e7c77 : SUCCESS in 11m 26s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f3f64d2e6be1411abe62ac2b4f3718bf : SUCCESS in 11m 35s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/14ca384bb4fe4cfd996da78c71ac2b43 : SUCCESS in 8m 00s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/96862fc6db1241778071dee43a184c03 : SUCCESS in 12m 46s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/dce7bf4c575847d19ba3cb774b2f561d : SUCCESS in 4m 52s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/ec86a5d0327d4bddb0d80c779cfd766e : SUCCESS in 16m 36s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/2ed99bcb5fdf48fb8099df200c3e4037 : SUCCESS in 6m 52s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/f937238aa7b64e8d9338f622841d835f : SUCCESS in 23m 46s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/ff628403abde47548cc6657f63b62ff1 : SUCCESS in 14m 13s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/30e9335cbd644e968f83a734123fbb12 : SUCCESS in 8m 01s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c869dac6c83d4e828822b86bc956cf06 : SUCCESS in 4m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/67b8bf077bdb4afe8daa0ea45cedeb05 : SUCCESS in 19m 30s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/109d18b554654bbcb5961ae72c379cbe : SUCCESS in 7m 57s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/f29c950e127d4b41802fe830661d5b54 : SUCCESS in 6m 16s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7cfc32f4bfb543ea8dc048be5cf44c33 : SUCCESS in 8m 50s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/b6ee5d22c8d243b0b811e16eed50fc22 : SUCCESS in 13m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6698d1925f0a4c68a5a9e365bae14a33 : SUCCESS in 49m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/eca73eedecdb49d9baa201ca1c3990a1 : SUCCESS in 1h 36m 57s","accounts_in_message":[],"_revision_number":1},{"id":"3e0b5330006aac9545cb1a3e5535c3b0062e9636","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-01-25 18:04:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b47b33dea849f9ad9ddf36b5af1e858b2f4deb7e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 18:05:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d24f45219535666711f74376e01b478447568dde","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 20:11:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0c2f45860104be08fe9fd6eed2bcd05 : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f0137b13bfd4047b39e1d7a48a5bbd8 : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/726c5423ee8c49b28cfca1a897f11d40 : SUCCESS in 4m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/21015c9cfde047f39cf7cb09edf8248e : SUCCESS in 4m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fa5f7fa2a65842ccbdf18fa4b14640a0 : SUCCESS in 4m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/aa084a783f8148299d1c6621824f66e3 : SUCCESS in 3m 55s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/7532aecc561141caa2344d32954b6889 : SUCCESS in 11m 32s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c4e0525ca1e4465d8f3d44bebb7531b8 : SUCCESS in 11m 41s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/2a1e707bd4794e03bb888aa75280db5f : SUCCESS in 10m 33s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/2ede8e52cc034367b858fabfa55d9c35 : SUCCESS in 6m 03s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/7bf627a8c3b247149addadc6f4d0f539 : SUCCESS in 10m 58s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c53650ed9eec4fff8d07d7506fac2c85 : SUCCESS in 8m 45s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/5db731a1cbd24bb8ac97801abc287e37 : SUCCESS in 8m 29s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/746bbaff5dc54a8d8575ed941bb4ac49 : SUCCESS in 13m 30s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/787ff1a7b54f4edcb5f442f8830413e8 : SUCCESS in 4m 47s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/460344bfaac54abdb4f478fbfceb71fb : SUCCESS in 18m 14s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/8db89f8772df4614a17b7f64854193ce : SUCCESS in 5m 58s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/b0f8239920c64d1daf559d0cf632d92c : SUCCESS in 28m 47s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/b2210378344947c1bb62967e278b5858 : SUCCESS in 12m 56s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/91caa00a1f6b48da8312118b289d26cf : SUCCESS in 6m 30s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/24181520be834039b66a9c68a3c502a6 : SUCCESS in 6m 03s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/15de7cf85c8047c09ec76160a20d94ae : SUCCESS in 28m 54s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/64de7e19edb6496782b5e4ebdfe8bf7e : SUCCESS in 7m 18s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/8f9c32967c9548af977dad54b5dcd4e6 : SUCCESS in 6m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bcd78e430b1e4961afd551dbfd42f4a4 : SUCCESS in 8m 32s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/5dd51b0bb4934ae08a59e9e23e2af36d : SUCCESS in 13m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/214ffa37d5ba419eaddade6b25e8ac2a : SUCCESS in 37m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/802e31e86c3047b1b6c67a9d9cbd854c : SUCCESS in 1h 30m 53s","accounts_in_message":[],"_revision_number":1},{"id":"ee2d9d6b40c31e99cd3e5140d506476975f9d0aa","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 20:11:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c2c4a237c664a174969e6607e44544c5329b89d1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 20:16:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25b74cd774824a5dac1de3200c9a7838 : SUCCESS in 1m 29s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c086fff066e40f3a87ff81fe72c8c16cdefe7c64","revisions":{"c086fff066e40f3a87ff81fe72c8c16cdefe7c64":{"kind":"REWORK","_number":1,"created":"2022-01-25 16:13:07.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/18/826318/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/18/826318/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/18/826318/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/18/826318/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/18/826318/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/18/826318/1"}}},"commit":{"parents":[{"commit":"c41643e539273121e71e34e1ec825cc18b713396","subject":"Merge \"Block dnspython 2.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c41643e539273121e71e34e1ec825cc18b713396"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-01-25 16:12:02.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-01-25 16:12:02.000000000","tz":60},"subject":"update fasteners to 0.17.3","message":"update fasteners to 0.17.3\n\n0.17.3 fixes a deadlock with the ReaderWriterLocks\n\nChange-Id: I3b0bb9da60b3d1e4317e804251fc2540b3120cd2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c086fff066e40f3a87ff81fe72c8c16cdefe7c64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c086fff066e40f3a87ff81fe72c8c16cdefe7c64"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
