)]}'
{"id":"openstack%2Frequirements~826366","triplet_id":"openstack%2Frequirements~master~Ib035e71db755d45b29d2a79d9169fdec9fe3fef5","project":"openstack/requirements","branch":"master","topic":"oslo.limit-1.5.0","attention_set":{},"removed_from_attention_set":{"4690":{"account":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"last_update":"2022-01-26 17:06:56.000000000","reason":"melwitt replied on the change"}},"hashtags":[],"change_id":"Ib035e71db755d45b29d2a79d9169fdec9fe3fef5","subject":"Bump oslo.limit to 1.5.0","status":"ABANDONED","created":"2022-01-25 21:01:43.000000000","updated":"2022-01-26 17:10:17.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"691daa6015dcb018b04e9e888b228e22c032e2ad","_number":826366,"virtual_id_number":826366,"owner":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-01-25 21:27:39.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-25 21:15:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-25 21:27:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ffb9bdf0b3caaec58d48a10756d8b2fc456eae2d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2022-01-25 21:01:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"352e80e968281315c3b63326e228611e49ae4e51","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 21:15:08.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/41841e55e93e41b7a4d656025327454f : FAILURE in 8m 22s","accounts_in_message":[],"_revision_number":1},{"id":"19249b76efe31ca1d46067ea1f4e25a117f9e94a","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2022-01-25 21:21:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"97e64a3fd9631894be1262554708364dbe38b63e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 21:27:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec0cb802d1584eb49b8717f0ed8b2b33 : SUCCESS in 5m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c05569363c354edda5b83b9baf06b24c : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/85d49d5cf52c41f3b6c6d6d085f1b13e : SUCCESS in 5m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e2a7d4a2b3964c878e957bdac3133386 : SUCCESS in 5m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e351f840cb1245ff8df06f4b292af34f : SUCCESS in 3m 51s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8d1e617318a7407b9dd730049c568d94 : SUCCESS in 4m 32s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0839d291a2a14a0ea9784bc1e05e1fdf : FAILURE in 5m 10s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/34486d2b1eae4cd1b41987978fe12e78 : FAILURE in 4m 06s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/5abab37e7e534b27818ef7c8c144c52c : SUCCESS in 6m 01s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/ac3049f49b85490dbe98d461f841f778 : SUCCESS in 10m 29s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/99fcaf6369124532bd204d6c44b03f1d : SUCCESS in 5m 49s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/c86f402312e34cc9adedf112d2176002 : FAILURE in 4m 42s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/fa08093f3b234abc96a3c0b08f8dacc0 : SUCCESS in 10m 07s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/773fc27d77e341c18465a9b0bf6e5332 : SUCCESS in 8m 15s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/5880e55301aa44ef8de37a29b5609f24 : SUCCESS in 9m 42s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/1d3c8876ab304225a021842fa105709b : SUCCESS in 13m 56s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/628363ac78c94124b9a3946637bd86c3 : SUCCESS in 4m 55s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/6107ca2a284f415789748573c4d0389f : SUCCESS in 17m 00s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/ae84eb862bdc4f19a62eabaef7024313 : SUCCESS in 7m 13s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/7984675562f344b1a268e7ff6cb709e9 : SUCCESS in 21m 04s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/3b6564143ac845c4a7ea7d58fd05d262 : SUCCESS in 14m 45s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/cc292386284841eea4fd26e3528842c8 : SUCCESS in 5m 54s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1d87e0ff835e4e4b9d36c83e44be3923 : SUCCESS in 4m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a13c451f75ba4da88cd01012c33c351b : SUCCESS in 21m 33s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/dcb81646d9f24cf285d4849bfa4571c2 : SUCCESS in 7m 14s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/bdbba6728c2a43a384e4f3e3436e7a76 : SUCCESS in 5m 34s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9d5ea685b05547119f4b1fd686166f3a : SUCCESS in 8m 48s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/cde44fd4259e40539427a422555e25b4 : SUCCESS in 12m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d99f3695c9c44897a1211dda92e695f7 : POST_FAILURE in 18m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d249715bf1b1448f930f0c1086fff3e5 : FAILURE in 13m 37s","accounts_in_message":[],"_revision_number":1},{"id":"dd2262ef22376b569938ece8a37dd341be99fa10","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2022-01-26 17:06:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"691daa6015dcb018b04e9e888b228e22c032e2ad","tag":"autogenerated:gerrit:abandon","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2022-01-26 17:10:17.000000000","message":"Abandoned\n\nLooks like the bot is taking care of this since the release builds were reenqueued:\n\nhttps://review.opendev.org/c/openstack/requirements/+/826497","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ddcdb04e0158a3989559a8eb00d427812ca89fa0","revisions":{"ddcdb04e0158a3989559a8eb00d427812ca89fa0":{"kind":"REWORK","_number":1,"created":"2022-01-25 21:01:43.000000000","uploader":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"ref":"refs/changes/66/826366/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/66/826366/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/66/826366/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/66/826366/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/66/826366/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/66/826366/1"}}},"commit":{"parents":[{"commit":"88801393f63419128653fcab9b04b33c3160654d","subject":"Merge \"Add cross project job for Designate\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/88801393f63419128653fcab9b04b33c3160654d"}]}],"author":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2022-01-25 20:59:35.000000000","tz":0},"committer":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2022-01-25 21:00:42.000000000","tz":0},"subject":"Bump oslo.limit to 1.5.0","message":"Bump oslo.limit to 1.5.0\n\nNew public interfaces were recently added for projects working on\nunified limits support to consume [1].\n\n[1] https://review.opendev.org/c/openstack/releases/+/826204\n\nChange-Id: Ib035e71db755d45b29d2a79d9169fdec9fe3fef5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ddcdb04e0158a3989559a8eb00d427812ca89fa0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ddcdb04e0158a3989559a8eb00d427812ca89fa0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
