)]}'
{"id":"openstack%2Frequirements~826457","triplet_id":"openstack%2Frequirements~stable%2Fwallaby~I8e338cefb460ce1618e938e768e829e2375c6841","project":"openstack/requirements","branch":"stable/wallaby","topic":"new-release","hashtags":[],"change_id":"I8e338cefb460ce1618e938e768e829e2375c6841","subject":"update constraint for python-manilaclient to new release 2.6.2","status":"MERGED","created":"2022-01-26 12:32:16.000000000","updated":"2022-02-03 18:17:48.000000000","submitted":"2022-02-03 18:17:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"826457-new-release","meta_rev_id":"df60b71832e09d29c24ff1c0a5eadd493abc6564","_number":826457,"virtual_id_number":826457,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-03 18:17:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-02-03 15:19:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-02-03 15:19:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-26 12:47:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-26 14:07:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-28 05:56:57.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2022-02-03 15:19:00.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"ca0449429a1bfc08bb733349bb22c383c22bf678","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-01-26 12:32:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0ab0c8c12dba88ebf4102209ff689a08db428736","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-26 12:47:30.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/31cd0a727b2f4ecf872ab3035c31ee09 : FAILURE in 11m 31s","accounts_in_message":[],"_revision_number":1},{"id":"301bfcc3df69a318f5c7d8bf19dd5b92140f9bd7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-26 14:07:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f704e7343ebf4ad9aba04e6dc1a3ef32 : SUCCESS in 4m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/374e143c688248a4952a1fa5750c8582 : SUCCESS in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e45451de84a0432c9be5a54870ce5c40 : SUCCESS in 4m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cad7c6df17a34d3ca6ba89b07401089e : SUCCESS in 4m 20s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c897977139ae4ff894bdb4a990047f1c : SUCCESS in 4m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7ec6d2d8d11640d6ba1db408a3cb305a : SUCCESS in 3m 42s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/cceba44be8614239be21056a3962357f : SUCCESS in 7m 28s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2e1a50fc88db49c18886a02fbb6e0aba : SUCCESS in 7m 51s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/bd595fc6ad4b4051b6753c02a91484af : SUCCESS in 9m 49s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/5fab164812784e819a9db5cd8cc97c78 : SUCCESS in 5m 30s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/f7d442a9bc5644869b829b8e73948319 : SUCCESS in 8m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5f07852b33ad48afb30febac8e22f647 : SUCCESS in 10m 08s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/d0cb485bdca94c63ba14952f5b18b22b : SUCCESS in 7m 50s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/95067d53f06c42be94303b002ba84ba2 : SUCCESS in 14m 34s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/36e3014bbcdf4e9694bac418b2219ab9 : SUCCESS in 4m 35s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a4456726c64e4c8d9e4b16807bbc7bd0 : SUCCESS in 28m 05s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f74de447be6e4c7cbcb50e6c602c8444 : SUCCESS in 12m 14s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b156eb7247bc49afa956976d7b922f2f : SUCCESS in 4m 31s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1d0c2cb7b63b446b9b5926957cc0df56 : SUCCESS in 19m 44s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/a5cec1b2a4584c30b21c05e1bd6c3e83 : SUCCESS in 4m 59s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/513d7375a0984fea89352f85a0543d32 : SUCCESS in 6m 56s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c274e471393343969221b0c4afaf54be : SUCCESS in 11m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d42d921a15684a2a92bcb03266a2ae3c : FAILURE in 54m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/73993ac6d05e46c19daf63a9b20ef3c2 : SUCCESS in 1h 29m 13s","accounts_in_message":[],"_revision_number":1},{"id":"75b5dc506d46421d8a27e651759f29f9715dabe1","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-01-28 05:56:57.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"abdd1f5bbce433109c725af4746de203c6f135a9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 06:16:05.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/a86f0aede72843188067e5a2b028ca9e : FAILURE in 16m 04s","accounts_in_message":[],"_revision_number":1},{"id":"0728c1ba9ecf43d01c45e22b3f5d877e26aaa830","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 07:29:23.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ef47db757184b5caf7cc23f65f1e39c : SUCCESS in 3m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f4ec4c02c293440ebeae7cc539f005ad : SUCCESS in 5m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aa15cd15dc0442ce981a6c0a1c24ef05 : SUCCESS in 5m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/910802a565694595a4dc695ffafe2fdd : SUCCESS in 3m 49s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a895613629af4372a2131a43437ad2c1 : SUCCESS in 3m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3e0e35bd45cd48d0854a21c0a43952a5 : SUCCESS in 3m 42s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/dead35eac3724a878b69b39d4f3d7ae6 : SUCCESS in 10m 05s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/39a3a474ca7c4d41ab4b91e297a8746c : SUCCESS in 6m 40s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/24dd3afc8c964fbfaa95a3511574c085 : SUCCESS in 10m 53s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/2aecd625a1e4477eba4b0a1991e7f808 : SUCCESS in 5m 41s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/2f7334d2466944d586f747994dcffba6 : SUCCESS in 9m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/13ad4be7faf84151bdcbc89166a8c553 : SUCCESS in 9m 15s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/270f92f07cbc4d38a00d20d88037a436 : SUCCESS in 8m 37s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/04abbd42840a4261892d82cfa789921c : SUCCESS in 16m 46s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/1edb9b2017ce4016b09b5d7a12c92709 : SUCCESS in 5m 14s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/1c48f443c82342329cd54666f5417821 : SUCCESS in 23m 23s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/fd1b867b1953404d9802ebb4ff2ec683 : SUCCESS in 13m 33s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6bcdb768246140d1b3ead6f2e827fb9a : SUCCESS in 4m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/771f3d85f8fe4978ac3a0a238c56a6aa : SUCCESS in 22m 06s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/78da35917d514ff5a6612b5613b43bcd : SUCCESS in 5m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1fd86f0278a2441bbf7b4bf7323945b5 : SUCCESS in 6m 11s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/24fabdd570b74285a13918a905ba1673 : SUCCESS in 11m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8d552d5ac8d04e9d89beb671df964d34 : FAILURE in 49m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6e7da7187f384ab081e35a261216d1f8 : SUCCESS in 1h 26m 30s","accounts_in_message":[],"_revision_number":1},{"id":"55e59e1aa991676576f806316c3daa02657625fd","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-03 05:01:10.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7cb4ac08b4660a5b47b4ef71b6e4f8218879c7b6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 05:30:11.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/27fa42c0d8944c6cb72f0ee6eab52b92 : FAILURE in 16m 29s","accounts_in_message":[],"_revision_number":1},{"id":"4534084858b83bb36ed02736e1adaf051c06e1eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 06:30:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a762186f2e7a4bc998c52c50055fe503 : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ed5f11386011411c83170c47d52d9fef : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/969ec25a734d4d1ebd97bb52844c9d5f : SUCCESS in 4m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cadb4c6e6faf4d6bb8b1738ebd47b8b1 : SUCCESS in 5m 58s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/72201ef37add4e3496870be39fecbbda : SUCCESS in 4m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bb92082ef49540a5b03c5400751f2f51 : SUCCESS in 4m 33s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4c163b9668694c6cb95f7136bf37e564 : SUCCESS in 8m 58s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/43eeb7db73ba4f269ca4ca7f10cea505 : SUCCESS in 9m 28s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/914363e5ea6e46a2a94d22c357b3986a : SUCCESS in 9m 29s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f00737fc1265444e851abfa99e047fd7 : SUCCESS in 6m 36s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/9029efe994084d908b41fc7a1b235cea : SUCCESS in 11m 05s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4a71da2d0eac479d89fdee15c27f8809 : SUCCESS in 9m 07s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/d21d5aca1b8d4a689d6ced8f82277b0f : SUCCESS in 8m 48s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a6366dba66f1482eaa531d5c27b2eba6 : SUCCESS in 13m 10s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/01f79caefe7e40d7843b1bb672f100ba : SUCCESS in 6m 20s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/7fb85934c29642689d067ce39b7adfdc : SUCCESS in 30m 58s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/0bb31b1315714fd4b4b626349ab89820 : SUCCESS in 10m 32s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b70d3f0b9d8e4cb68ba6d08f9be570ba : SUCCESS in 3m 43s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/076fcdebe4414c5fb91db5b1bb477c80 : SUCCESS in 20m 44s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/177906894b72459e9d1b7aa2324a9819 : SUCCESS in 4m 52s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ae0ff3e7f1ca43f4b1b50041c9bc72e8 : SUCCESS in 8m 22s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/a4c6889f70234443b5fea87b5c785fbc : SUCCESS in 12m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0e6b72769c0b4a228ffa6555c1f666f2 : SUCCESS in 47m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ce06a43645a94d1e8eb3d6d2185c56c7 : SUCCESS in 1h 10m 48s","accounts_in_message":[],"_revision_number":1},{"id":"8c5f39a4ef3f7bee198408dd524e08d8b79fada1","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-03 15:19:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ab6776a02ffcebc9131f2135061fee036076799f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 15:21:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c5aa600e30514a0ab0b37543d40799b0a246ae47","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 18:17:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cce8d4fbcc49437da52c54a7d361adfa : SUCCESS in 5m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/131254039e2445168f84073450e29dbb : SUCCESS in 6m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/defefbcb486940aabaab6da6f540feb6 : SUCCESS in 5m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/73a9bdff95834ce982c1dd432de280cc : SUCCESS in 6m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c22f2668ad39452f92c348a2b54944cc : SUCCESS in 4m 33s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ef8e93bc8ce740a8848c23e5466a3f39 : SUCCESS in 9m 45s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b9a345ab78cc4fd7aadce3f7821302dd : SUCCESS in 9m 17s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/aa2035de449d4c51b85ca98a92891ba0 : SUCCESS in 10m 38s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ee6c42c224f3404b90d77aed09ae856d : SUCCESS in 11m 13s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/e3636d7216294a19a4ca306a8b856487 : SUCCESS in 11m 32s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5b4f0ad8526445e5b944fea94140123b : SUCCESS in 12m 06s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/258574195111455fb6709a501946ec62 : SUCCESS in 9m 57s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e2426447c2b546a7a498ced3b388e85a : SUCCESS in 16m 05s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/18c225db618d4240a22f3769e3ffe6b7 : SUCCESS in 6m 52s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/983056d81c4e4184a1b108525eb03af1 : SUCCESS in 25m 52s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/733db18346594d6cbcb6c32489a04a37 : SUCCESS in 18m 18s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b40c677209b54dcc97547a5a45f0874a : SUCCESS in 4m 46s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e41c454184734995b37d9dd28d7a2575 : SUCCESS in 20m 21s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/57c4a84ac2cd406bb28331f3d5a0a61b : SUCCESS in 6m 46s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6ff566703ef446aea64629ee318acec6 : SUCCESS in 8m 18s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/6f0bdfda72944f35b2ebb8d11e372a57 : SUCCESS in 12m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e00be53116994deda6ef78c3850777ff : SUCCESS in 54m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2c5f86e2c90b4cc3830d01f60371040a : SUCCESS in 1h 25m 07s","accounts_in_message":[],"_revision_number":1},{"id":"df60b71832e09d29c24ff1c0a5eadd493abc6564","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 18:17:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bbeb3dbe43d34c3c4c13e5a0219df3aa734cb48e","revisions":{"bbeb3dbe43d34c3c4c13e5a0219df3aa734cb48e":{"kind":"REWORK","_number":1,"created":"2022-01-26 12:32:16.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/57/826457/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/57/826457/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/826457/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/826457/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/826457/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/57/826457/1"}}},"commit":{"parents":[{"commit":"0e67f1d4bee22d1d2d56ec24fe3282c9a5a7ffd2","subject":"Merge \"update constraint for oslo.policy to new release 3.7.1\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0e67f1d4bee22d1d2d56ec24fe3282c9a5a7ffd2"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-01-26 12:32:15.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-01-26 12:32:15.000000000","tz":0},"subject":"update constraint for python-manilaclient to new release 2.6.2","message":"update constraint for python-manilaclient to new release 2.6.2\n\nmeta: version: 2.6.2\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\nmeta: release:Commit: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\nmeta: release:Change-Id: I45600e997dcc33dd05ae3d03ebb9289ffb7af344\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+1: Tom Barron \u003ctpb@dyncloud.net\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I8e338cefb460ce1618e938e768e829e2375c6841\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bbeb3dbe43d34c3c4c13e5a0219df3aa734cb48e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bbeb3dbe43d34c3c4c13e5a0219df3aa734cb48e"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
