)]}'
{"id":"openstack%2Frequirements~826604","triplet_id":"openstack%2Frequirements~stable%2Fvictoria~I560fa08e8fc3d2b2df1510fb9611eb8732361389","project":"openstack/requirements","branch":"stable/victoria","topic":"new-release","hashtags":[],"change_id":"I560fa08e8fc3d2b2df1510fb9611eb8732361389","subject":"update constraint for python-manilaclient to new release 2.3.1","status":"MERGED","created":"2022-01-27 10:18:53.000000000","updated":"2022-02-01 21:24:20.000000000","submitted":"2022-02-01 21:24:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"826604-new-release","meta_rev_id":"122e7f5505fbd4f8b6d803c45ff85d13d2d18058","_number":826604,"virtual_id_number":826604,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-01 21:24:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-02-01 20:02:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-01-27 10:22:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-02-01 20:02:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-27 10:22:16.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-01-27 10:42:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-27 12:00:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-01 18:42:40.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2022-02-01 20:02:44.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"79df2ef19293721b075dce42e23e1a66a0146b2d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-01-27 10:18:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fc0f0fdffc9a1352669f3886eae349dc321b04b8","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-01-27 10:22:16.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c2344bf9d7869e8ea3b0cd572c8796e540ae1116","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 10:42:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/79973bc94bd64897859bf9cc841fc12e : SUCCESS in 17m 02s","accounts_in_message":[],"_revision_number":1},{"id":"696c56fe96e4dc1b6ee5195cef05fec7679d0389","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 12:00:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/840f7015cfbc4eb8b74da0b0f43ff7cc : SUCCESS in 3m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f1baa9913fc244aca474b3b3a75d7ae0 : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/12a15d748f0d4cc084eba0585df62833 : SUCCESS in 4m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6af51bea50844c8e93e5c634f2e2c023 : SUCCESS in 3m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8a2f8835a6d54e208a44ac3a34454382 : SUCCESS in 3m 47s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/381567d1659c4dc0a87431bf198ba3e6 : SUCCESS in 10m 35s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/71f847547d804340b6a5715c53a3f7fd : SUCCESS in 9m 31s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/3a4c3de8fba64a3da1aecfb607cddcae : SUCCESS in 10m 05s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/44872fd44d4f4a549387d7c8d838debe : SUCCESS in 5m 34s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/84abcaca46974db5b5454974ed7578fa : SUCCESS in 9m 52s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5adb17b661f049d6ba1e8663302b7992 : SUCCESS in 8m 47s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/52c9d454fddf48f99a880a257dd2a5af : SUCCESS in 7m 32s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/1c2d45f3898b4e198b183db84743da12 : SUCCESS in 13m 52s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/3afb48e16c6d426d91156e99ecc28aec : SUCCESS in 29m 01s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/a45c00cea80841138d60efbc351fa1a3 : SUCCESS in 18m 07s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/11661fbcc48b4c3da58bace99e405d24 : SUCCESS in 5m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/292a785eb9e14a85af6799b477c837ab : SUCCESS in 19m 05s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/e5c608125a824c3b96c0e18637a3d25e : SUCCESS in 5m 20s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1856dc4f903847efac761eaf4e021bf2 : SUCCESS in 8m 08s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/fe46a9bba9164737ad3d597a706c6c38 : SUCCESS in 11m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/16d9663fe4314fed90dd86a7411e3e6a : FAILURE in 55m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/00348153b3214f698e4c4c8436e0884e : SUCCESS in 1h 24m 19s","accounts_in_message":[],"_revision_number":1},{"id":"70085087bef0a29fea670ca2329088fd5bdf66c5","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-01 18:42:40.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c4c5ad47bb7a80529c37a03fe0d805429e5f94a1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-01 19:04:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/273624e68d1b452b8ca6916ad0882c3e : SUCCESS in 15m 56s","accounts_in_message":[],"_revision_number":1},{"id":"f40a93d49608ab256a5ce1711ab96d51809767d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-01 19:59:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87d21aa7ee084625a69fe2271a013ff0 : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/122faf11f2704437aa3bdaa1c77b3268 : SUCCESS in 4m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/74b9c338382746d792bda7be19d544ae : SUCCESS in 4m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/296488a4f0d348fd891515341608b600 : SUCCESS in 3m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4625313894604ccbb58a8bad32fac7a7 : SUCCESS in 3m 50s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6e38fcb3e6c54841918dc329645bc1dc : SUCCESS in 9m 48s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9ff615574ce242058e1546b620185774 : SUCCESS in 9m 13s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/0c899930139944afac4c846a6ff945b7 : SUCCESS in 9m 10s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/0ed16416c8c74b59b8340327b5c3100c : SUCCESS in 5m 09s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/0d80b84cbc7a44c49d8265144afb6cc8 : SUCCESS in 10m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2793fbabb0344254904df3629a54adf2 : SUCCESS in 7m 53s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/e043a3ea02c347189e8902a9323118fc : SUCCESS in 6m 41s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a786b29405fb4247aac4d9e984c6728b : SUCCESS in 12m 50s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d8e7d6544ec74da98a2c56efd0cf12d3 : SUCCESS in 23m 01s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/a202afe343304a63a46a87ffe6efa766 : SUCCESS in 12m 28s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e31e03897646467ba35d354426e7ada7 : SUCCESS in 4m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8c8c35e011204fc68cf8296862db13dd : SUCCESS in 18m 37s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/6adc0e537c27466ea4e873181576319a : SUCCESS in 5m 26s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/cd4b0e2768d1412b867718606948e0f8 : SUCCESS in 6m 37s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/57e058451dfc4f129ddbaf45bea4900b : SUCCESS in 10m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/aaefd6abbf7e4d21893ab4f3ff5db5c2 : SUCCESS in 54m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/867ea84cb25b4dc7bd88a29283eede15 : SUCCESS in 1h 12m 59s","accounts_in_message":[],"_revision_number":1},{"id":"099ed2f85636ebd683d724e5465a55fda0a11984","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-01 20:02:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4727a8ef98bfde2784b76f3560168af8606ee61c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-01 20:04:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ee4e23f5cfd3396d33421ca700886b0a783be13a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-01 21:24:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d650c48c72034fc99eed9107ac864a88 : SUCCESS in 4m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e873641e4a414bb2bf529f71f484d7e9 : SUCCESS in 6m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/18c9281a696343ae89b7a8f5305183e0 : SUCCESS in 5m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c8b6f80393404921a4c85c2c3269cc82 : SUCCESS in 5m 12s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/30860e53a57d4318a390375758b9a946 : SUCCESS in 4m 54s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/80d6cbc9388346ae8c032d877608b709 : SUCCESS in 10m 18s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/918765f9ec0c484e9372f428b3493b68 : SUCCESS in 9m 31s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/2e0f4c1b48984b60b69d3296e52e9eb4 : SUCCESS in 9m 18s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/fbc67519d9d94e2fb6db2592d3a2aa34 : SUCCESS in 7m 33s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/b8f88a1b28c44539bb4276e058837844 : SUCCESS in 11m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b5a1443f0e3c4f59b480a9c0fff1408c : SUCCESS in 9m 50s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/da7c20e2beea45a1a4c2b7c0e86ebad4 : SUCCESS in 8m 13s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/d89acfb1b7db47938e21f3410e75cbed : SUCCESS in 16m 52s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/6a0ec0c7b73149c4a9b89aaa5790addf : SUCCESS in 30m 30s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/11e5562959194dbc8a83377b7d2cc17f : SUCCESS in 14m 39s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5b7e9e6393cc49b4aed900fb88642371 : SUCCESS in 5m 35s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a4de8a647ba746b1ba0bec59e58daead : SUCCESS in 20m 11s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/cf52325bb83a46f786f8960b87cb9095 : SUCCESS in 5m 51s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ca22491a918b42ad80d7b477066a63cb : SUCCESS in 7m 29s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/0a0d49c0d0e2449c8f99f76f1668b296 : SUCCESS in 11m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ec2568563da842cc8693bb1f32d112fc : SUCCESS in 56m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6531bac2790f43088ac7f7474e5f3fb1 : SUCCESS in 1h 19m 10s","accounts_in_message":[],"_revision_number":1},{"id":"122e7f5505fbd4f8b6d803c45ff85d13d2d18058","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-01 21:24:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8c6579c78d27315a4498c7964b56225855e64b07","revisions":{"8c6579c78d27315a4498c7964b56225855e64b07":{"kind":"REWORK","_number":1,"created":"2022-01-27 10:18:53.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/04/826604/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/04/826604/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/826604/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/826604/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/826604/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/04/826604/1"}}},"commit":{"parents":[{"commit":"f8214a80076fd47ba9d03477af12b9314d7734f1","subject":"Move upper-constraints for validations-libs to 1.5.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f8214a80076fd47ba9d03477af12b9314d7734f1"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-01-27 10:18:50.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-01-27 10:18:50.000000000","tz":0},"subject":"update constraint for python-manilaclient to new release 2.3.1","message":"update constraint for python-manilaclient to new release 2.3.1\n\nmeta: version: 2.3.1\nmeta: diff-start: -\nmeta: series: victoria\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I690adeb5d0074c02ecf3b1159e72ae38e05dd673\nmeta: release:Code-Review+1: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I560fa08e8fc3d2b2df1510fb9611eb8732361389\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8c6579c78d27315a4498c7964b56225855e64b07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8c6579c78d27315a4498c7964b56225855e64b07"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
