)]}'
{"id":"openstack%2Frequirements~826997","triplet_id":"openstack%2Frequirements~stable%2Fvictoria~I98a9ce76a67367d702e31f39e97a68b5f06ba389","project":"openstack/requirements","branch":"stable/victoria","topic":"new-release","hashtags":[],"change_id":"I98a9ce76a67367d702e31f39e97a68b5f06ba389","subject":"update constraint for os-ken to new release 1.2.1","status":"MERGED","created":"2022-01-31 08:37:33.000000000","updated":"2022-01-31 17:02:46.000000000","submitted":"2022-01-31 17:02:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"826997-new-release","meta_rev_id":"4a2b9aae49b1c86681a68d7339895e8fc4356601","_number":826997,"virtual_id_number":826997,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-31 17:02:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-01-31 15:46:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-01-31 08:39:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-01-31 15:46:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-31 08:39:25.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-01-31 08:55:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-31 09:40:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-31 15:46:43.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"997f0e3a35c1c51cb8e2a7f131da62f431d2b494","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-01-31 08:37:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"82e73444afc5946e95f7fb73464f34beaa50a069","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-01-31 08:39:25.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"87802d191fb4bab23a1502df930c52ef6ec234a8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-31 08:55:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/61c10cc4c61e42db9cb2866ba281e4c0 : SUCCESS in 16m 07s","accounts_in_message":[],"_revision_number":1},{"id":"c64da8757c97959cbcb738e574605d4f4adaf146","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-31 09:40:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/533581a3f0f64ea6800620c7e787d1b3 : SUCCESS in 3m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/95b9e6be9ae04f5ca5111b0794b64bbf : SUCCESS in 6m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/92390348d0114300927526ef6f06d414 : SUCCESS in 4m 14s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e4de3539854243ee9d9a2ca9423fc1d8 : SUCCESS in 6m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/80230130b3854a4bb74865f352d4d23c : SUCCESS in 4m 08s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ced2625e0eff4bdab92986abb771441f : SUCCESS in 19m 01s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e0c991b11b8f4523821e74332e32db3e : SUCCESS in 7m 00s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/f60f5ff1bd454e319c62014ec92c89f8 : SUCCESS in 9m 58s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/466654d3d9864c629e9ee0dfd8e3f47b : SUCCESS in 6m 07s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/f12511483a214821a62345de41430ff0 : SUCCESS in 9m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7533d200ca954d8a9d4a9f7dda051f11 : SUCCESS in 9m 01s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/fc79c648591c4186862965348f101ea6 : SUCCESS in 7m 12s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c8796113d7ae4674b4b49fe3ecd48ae6 : SUCCESS in 14m 27s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ba508f82dc6b49029654f73c2961f6d3 : SUCCESS in 28m 18s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/4ff674da87d447299efafc619ee97d4f : SUCCESS in 12m 16s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/170112a9fba44770963f45d56041def0 : SUCCESS in 3m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0f8be334577b44ea8809540aefdf8aa1 : SUCCESS in 18m 55s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/adc4104029dc452c812f197f43a2e85c : SUCCESS in 6m 34s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/024c398695ae4effa471a89c45a904c2 : SUCCESS in 6m 10s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/9650f2097a1949d3a02eebac8e934ffd : SUCCESS in 12m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/741fcd2a69f04f4ea884bd404b5ab439 : SUCCESS in 50m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b3535aa2bbba4ad8aabcf161e9aa825f : SUCCESS in 59m 07s","accounts_in_message":[],"_revision_number":1},{"id":"5fcbc85beecc7a9724333d9240b631f2f8e5e2e5","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-01-31 15:46:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"61a711925bd8e6e310e2299fd80212fcb90645f3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-31 15:56:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"794d5ec05a1f839057331f3549a747d4d48fbb58","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-31 17:02:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c643ec903a0746369492fe4c45c35f1d : SUCCESS in 8m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/afd4fc0ee1f64d3e80be424674a82287 : SUCCESS in 10m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/acf12affb82a4820bbf3eafc5c72998c : SUCCESS in 8m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/db7bf80f3cd24c1e9d60d2c7a4d5a590 : SUCCESS in 7m 45s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/61c89bf5c1074c7aab3072e35e73aa73 : SUCCESS in 7m 46s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c0534fd113fe4549a3b5ecfa2b2c9536 : SUCCESS in 9m 44s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/17cdd2e12fcc444ab06ad4498e849a49 : SUCCESS in 10m 52s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/5f2f335d0a1a450b91e9b77b85e340a5 : SUCCESS in 11m 55s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e28a28fc161b468ca14cbc8a5e6818aa : SUCCESS in 8m 32s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/0415a8e9f4514e3da16c44ac34e5f81a : SUCCESS in 10m 33s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5553436ef1d64237bd1fe08c3323e318 : SUCCESS in 12m 50s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f295b01c65454d62966f047dcce08bd7 : SUCCESS in 14m 48s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/96824cd96483425a8c78b90b5a858b58 : SUCCESS in 17m 18s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/4d61f06d9bcd4b6fad3ce4f839700a21 : SUCCESS in 26m 40s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/e9371977c3104de9b6f5663c2dc7901d : SUCCESS in 16m 46s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8d71100c5c724779be0a921f98f335d6 : SUCCESS in 8m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/194315a6a74d4369addd4dfbe1b2b765 : SUCCESS in 25m 36s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/47aac93e8c8941bebb79637f71226315 : SUCCESS in 7m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9b66a5c97a9448e7a989155c3d01c028 : SUCCESS in 11m 54s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e7e6e7d87c6f456f9ce474b76e093d52 : SUCCESS in 15m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6ed896be5cc5400592d20a0f750b588b : SUCCESS in 53m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0ba89bec50314acfa3199f6f8b660e98 : SUCCESS in 1h 03m 06s","accounts_in_message":[],"_revision_number":1},{"id":"4a2b9aae49b1c86681a68d7339895e8fc4356601","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-31 17:02:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3b56b35b6e67d98226f8032a7257c469c2e88495","revisions":{"3b56b35b6e67d98226f8032a7257c469c2e88495":{"kind":"REWORK","_number":1,"created":"2022-01-31 08:37:33.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/97/826997/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/97/826997/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/826997/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/826997/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/826997/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/97/826997/1"}}},"commit":{"parents":[{"commit":"f8214a80076fd47ba9d03477af12b9314d7734f1","subject":"Move upper-constraints for validations-libs to 1.5.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f8214a80076fd47ba9d03477af12b9314d7734f1"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-01-31 08:37:32.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-01-31 08:37:32.000000000","tz":0},"subject":"update constraint for os-ken to new release 1.2.1","message":"update constraint for os-ken to new release 1.2.1\n\nmeta: version: 1.2.1\nmeta: diff-start: -\nmeta: series: victoria\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I1c41c5d4033cec62169aa1efe39af5049ca857bb\nmeta: release:Code-Review+1: Lajos Katona \u003ckatonalala@gmail.com\u003e\nmeta: release:Code-Review+1: Bernard Cafarelli \u003cbcafarel@redhat.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I98a9ce76a67367d702e31f39e97a68b5f06ba389\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3b56b35b6e67d98226f8032a7257c469c2e88495"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3b56b35b6e67d98226f8032a7257c469c2e88495"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
