)]}'
{"id":"openstack%2Frequirements~827322","triplet_id":"openstack%2Frequirements~stable%2Fxena~I5e0d17476e351fc4b8eb8714a9f25c31d6a74bc7","project":"openstack/requirements","branch":"stable/xena","topic":"new-release","hashtags":[],"change_id":"I5e0d17476e351fc4b8eb8714a9f25c31d6a74bc7","subject":"update constraint for neutron-lib to new release 2.15.2","status":"MERGED","created":"2022-02-01 16:56:07.000000000","updated":"2022-02-01 20:41:14.000000000","submitted":"2022-02-01 20:41:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"827322-new-release","meta_rev_id":"a85cdc4136c52bcd6a425d64f0f7e446a3aaa4b1","_number":827322,"virtual_id_number":827322,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-01 20:41:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-02-01 17:25:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-02-01 16:57:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-02-01 17:25:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-01 16:57:53.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-02-01 17:16:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-01 17:25:15.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2022-02-01 18:12:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"38ed772a6d892da117db8d75853f2f4e1ffdadd7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-02-01 16:56:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"48f89563c9d804b69e9110acc4d6d32ed0a5f147","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-02-01 16:57:53.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e25a8832a7689ee41b55b91e7e5f379f2cf62871","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-01 17:16:43.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/69ad056d54f948f3b12e38d18a541a6a : FAILURE in 17m 54s","accounts_in_message":[],"_revision_number":1},{"id":"8b7931e632aeaa6a637c3a911cbd525706524252","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-01 17:25:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c6c562104a5a0bf82704fb0f87e3e6c860af4f7d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-01 18:12:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ccd7983b90d94bcab2d82f5849eb7091 : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1acc4d8836374160b0bd04f6c7839640 : SUCCESS in 4m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cec9c7f2f9ae4174b2f26c06a1198fe8 : SUCCESS in 4m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/21faac5114404cf89e11dd4ca0a9cd37 : SUCCESS in 4m 23s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1f0d6038fcbc40079a78e7eeb98b9bfb : SUCCESS in 4m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/609463d91b624071820defa56c33bb94 : SUCCESS in 4m 57s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9a0985ffb0594e2788d6d9f9b86bf911 : SUCCESS in 8m 16s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8e8786a2d9c84f789f7934d33d221cd7 : SUCCESS in 8m 59s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/128ea86070734cb59708fc35ecd3f291 : SUCCESS in 5m 19s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/eedc31b243884ea7a67f58124eb7c0e1 : SUCCESS in 9m 18s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/48ef580c6f8942d8a338ef45c88cd024 : SUCCESS in 7m 07s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/3ded9c073ae04aa8a9af184a8aee2a90 : SUCCESS in 10m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0845479334dc4ab3830b7d3d4c5d66b3 : SUCCESS in 9m 14s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/7b0a71ddbfcf4b25b036070ac68f2737 : SUCCESS in 7m 42s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/e35bbccf6f634e1eb9c977b0e2d80cbb : SUCCESS in 14m 49s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/ca3066e5255e4f84953bf264daa0dc82 : SUCCESS in 15m 18s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/8864c9ce92a34b5fa485e58430aec671 : SUCCESS in 5m 15s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/29e85b5c26f3427aaf50c531c0942d23 : SUCCESS in 21m 28s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/ed7614af39fb439fb82c55fe26b0a22d : SUCCESS in 12m 08s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/bf0aeefcfa85402a9d509bb5a3825af2 : SUCCESS in 6m 25s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f8d419112d894fe7bb495e07af7a786a : SUCCESS in 4m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f0a5658cf28a41fda2e86968774a5453 : SUCCESS in 17m 45s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/a62c3a30ba8e4b12b8f9470c6a51dcbc : SUCCESS in 6m 36s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/953f3e3ef9394a319ff22b185dacdec2 : SUCCESS in 6m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a9b477f035664c76af1978ebeeeaa4b4 : SUCCESS in 7m 07s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/56e09a1b2eac4e088811ba922593a673 : SUCCESS in 13m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2fc3b7b677644fe0a09a65c6012edf21 : SUCCESS in 48m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a91ec23bc59d4dd7b876b4b76d0bd6cb : SUCCESS in 1h 13m 37s","accounts_in_message":[],"_revision_number":1},{"id":"a66dfcfc1427442edc82457f95b0ff18ef45fc50","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-01 18:13:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1c66a95fa03ca765db1f342f353b64b31a3a6628","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-01 20:41:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a017c04b23d740908b1ce6c5e0ea03fa : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/19f62b48985040bb8c3818fc77d1c105 : SUCCESS in 5m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a08ff1af89aa4424bc94602e62531cf1 : SUCCESS in 5m 52s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e9121aaabeb040099628a50ff6975214 : SUCCESS in 5m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b2bc022c300145839aa51d81bfd87c9e : SUCCESS in 4m 51s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/24971dddc8234b0bbb17fc6184b51244 : SUCCESS in 9m 18s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/79109fffbd694586a08e46c14fc24308 : SUCCESS in 8m 58s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/4cd2161637cb4c75b79a32b6e9a165c6 : SUCCESS in 10m 35s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/21cc27ce42fe49649b1883b264730ab0 : SUCCESS in 6m 11s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/b36d0db4310e4ad49a6bc24c7bb44110 : SUCCESS in 9m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/750357bb26bf49bebdcb847e6317c525 : SUCCESS in 8m 26s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/cb8cbacaf20b4b6ab809db007d61d579 : SUCCESS in 9m 17s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/40997d26c96348ad9b75c76a9b406f7c : SUCCESS in 15m 28s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/091bb39407464543b59ff506be27ee46 : SUCCESS in 10m 54s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/f0566c1e0c4f47199e4d4aa321799002 : SUCCESS in 5m 16s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/ea974100807d4922b10751429428ee62 : SUCCESS in 21m 05s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/dd301a4a8b0242b88610e0504f24a4cb : SUCCESS in 13m 26s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/195d266ace644fd28bdc5ff300a26950 : SUCCESS in 5m 39s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9c755ab7092c464ca7178c1c03667cb0 : SUCCESS in 4m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/824afea4aaf7489191d989b6a260cd47 : SUCCESS in 22m 33s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/28bff8740f6741209b36522f8582144c : SUCCESS in 6m 23s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/b8d9f3abd57d4e1399b4b8565c718f28 : SUCCESS in 5m 54s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/36956e8840eb4fb2b726b8b216f7a569 : SUCCESS in 7m 27s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/c68beed5e5a149878dbe52e6e1909cb7 : SUCCESS in 14m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d031936a85414c54919481ddb9059054 : SUCCESS in 42m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0225090258c94633a1ee14c244cc9bde : SUCCESS in 1h 13m 25s","accounts_in_message":[],"_revision_number":1},{"id":"a85cdc4136c52bcd6a425d64f0f7e446a3aaa4b1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-01 20:41:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"dc552ba7ddeaad57b18d2ceb1cffbc08fbabc223","revisions":{"dc552ba7ddeaad57b18d2ceb1cffbc08fbabc223":{"kind":"REWORK","_number":1,"created":"2022-02-01 16:56:07.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/22/827322/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/827322/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/827322/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/827322/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/827322/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/827322/1"}}},"commit":{"parents":[{"commit":"2f8a6c78b2fd51d59996b44697379ae6e27ac9bd","subject":"Merge \"Bump dogpile.cache to 1.1.5\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2f8a6c78b2fd51d59996b44697379ae6e27ac9bd"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-01 16:55:59.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-01 16:55:59.000000000","tz":0},"subject":"update constraint for neutron-lib to new release 2.15.2","message":"update constraint for neutron-lib to new release 2.15.2\n\nmeta: version: 2.15.2\nmeta: diff-start: -\nmeta: series: xena\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: yatinkarel \u003cykarel@redhat.com\u003e\nmeta: release:Commit: yatinkarel \u003cykarel@redhat.com\u003e\nmeta: release:Change-Id: I7ad7baccc1d092d6938e9231f68862e06c3e00ed\nmeta: release:Code-Review+1: Lajos Katona \u003ckatonalala@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I5e0d17476e351fc4b8eb8714a9f25c31d6a74bc7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dc552ba7ddeaad57b18d2ceb1cffbc08fbabc223"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dc552ba7ddeaad57b18d2ceb1cffbc08fbabc223"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
