)]}'
{"id":"openstack%2Frequirements~827323","triplet_id":"openstack%2Frequirements~stable%2Fwallaby~Ia57550a30064674503820444e4f476f058efbd9e","project":"openstack/requirements","branch":"stable/wallaby","topic":"new-release","hashtags":[],"change_id":"Ia57550a30064674503820444e4f476f058efbd9e","subject":"update constraint for neutron-lib to new release 2.10.2","status":"MERGED","created":"2022-02-01 16:56:42.000000000","updated":"2022-02-01 20:41:20.000000000","submitted":"2022-02-01 20:41:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"827323-new-release","meta_rev_id":"bcf307831adc370f425894843037ea2d736999c0","_number":827323,"virtual_id_number":827323,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-01 20:41:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-02-01 17:25:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-02-01 16:58:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-02-01 17:25:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-01 16:58:02.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-02-01 17:11:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-01 17:25:14.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2022-02-01 18:26:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6793008b11d7c5c0ede724b24efeef74e0f2f437","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-02-01 16:56:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1e18da505d08def797c8d3f2cb3f1874e00583b1","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-02-01 16:58:02.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"201d7207476f951dc88ea739f1729f497497114d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-01 17:11:54.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/8fbf6aaea59d4d059ca11b5209e1b796 : FAILURE in 11m 03s","accounts_in_message":[],"_revision_number":1},{"id":"bc779397fb0ae80d69bf545d3a0bb76eb4cb377c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-01 17:25:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2762f02c49ff80130bf594bc0b0cf1704ecd0f60","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-01 18:26:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d94632fe8e6e4d43b23eab9ea9976a86 : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/93e991ad1de341889777ea55e0d70fac : SUCCESS in 5m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b6246b360c32437fa8a0ac30bac5a3ba : SUCCESS in 4m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/98a77a9f47f448a3b40cdb5ff3851b73 : SUCCESS in 6m 19s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/84e69607ef3b4b5eba07db060b4d5147 : SUCCESS in 5m 35s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/18fffb9054e844f1b6550b9f7d74b07a : SUCCESS in 4m 17s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/26dbdf42a58f496598bc93b049bba6f9 : SUCCESS in 9m 11s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ff7ffc0de8b24daf8637c984cafd87bb : SUCCESS in 8m 37s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/93cf8d6ae3c44004ac163d88caaac98b : SUCCESS in 8m 13s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/b923f402a6ae418288962619852f4c39 : SUCCESS in 7m 25s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3ebc504fefe946d2b75785031b22df33 : SUCCESS in 10m 53s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f1a367ac0125415dbd75451523db70b3 : SUCCESS in 7m 59s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/ffc6c8b3093a41a0a0eea07b57a47ccc : SUCCESS in 10m 43s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/dcc90e41576f4e239ec5aca62ee13e50 : SUCCESS in 14m 14s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/4568964c237e499ba5120d8e8259bb2d : SUCCESS in 5m 29s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/99512e6052f04cfbab35cb26282f1df2 : SUCCESS in 30m 49s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/93515761be3a48a4925b4ea9f210e6d6 : SUCCESS in 14m 31s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9a42f0a1509b4142bb6b8c528bf489ff : SUCCESS in 4m 36s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/402a8ac453684a2c94d6f6483e70f0cf : SUCCESS in 21m 09s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/c0697a119ee344d1b0884fe3426c240c : SUCCESS in 6m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3c8a04eea3984cc3bbd845628414d4eb : SUCCESS in 7m 24s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/cae07470f2704d6180f50e4995d43889 : SUCCESS in 12m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/86f2990b1d0b4a708f766273552146bc : SUCCESS in 55m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5294a4e2f679496a838bbc6b82a4db68 : SUCCESS in 1h 25m 55s","accounts_in_message":[],"_revision_number":1},{"id":"edb056aff5f40606269fe7886da9005b00fceff2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-01 18:27:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"050ce6817cbbfe193ee1adf4e4c6b7645ebce7a9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-01 20:41:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cbf0d505896f476fa98a544ca2378767 : SUCCESS in 4m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4805d43ebf7d4361a9024d8f9605da23 : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1a51b14b3fe844ed8ef96673a5756c19 : SUCCESS in 6m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/89e2d97fc8c2470fa17f45b701fc3acf : SUCCESS in 4m 29s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fcd22c8ddd7347f4ab882a968abca96d : SUCCESS in 4m 15s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ff5fd3441b11444b9850d5b087448887 : SUCCESS in 9m 41s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b4ae2a60445740aba0c783ce8089abc3 : SUCCESS in 9m 56s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/5c235871edfb4f1ebfc27936e01ecb0e : SUCCESS in 9m 02s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a85bc1d1204441a6b4ff1b8f64be152b : SUCCESS in 7m 47s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/e1e1a0a7e9c04dc2a4a326c919f46589 : SUCCESS in 11m 33s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5b9897b53687449e8b85a291c32279ca : SUCCESS in 8m 42s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/c12c3c597ca34cfaadea6ebe2cb08397 : SUCCESS in 10m 01s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/ccd82d012e0c43d6930ba0308596da3a : SUCCESS in 17m 58s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/b0d6063fb7af46519085ba638b01b39b : SUCCESS in 5m 41s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ee416dfea21344a7bff43d345323d388 : SUCCESS in 29m 28s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/3f8f2ce63215446f8ef95958c56a1c71 : SUCCESS in 13m 14s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5895a15d10e84983a43501728842899c : SUCCESS in 4m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4511fb2bbdb24e9484b3f3464cfee301 : SUCCESS in 17m 08s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/aff66eb0e71f4304bd39d7c67c1b947c : SUCCESS in 6m 07s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d60ee700c6c34e228e3c76fba7f87683 : SUCCESS in 7m 57s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/5c1fa3aee9b94a41b809b69a68293bd2 : SUCCESS in 10m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0fdc4e91a3db4b6eb4b038998d07ecda : SUCCESS in 52m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/49b2abe7828749c09dcd89d67fa4053c : SUCCESS in 1h 20m 41s","accounts_in_message":[],"_revision_number":1},{"id":"bcf307831adc370f425894843037ea2d736999c0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-01 20:41:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"03bb2441311149e8c395d60810255890f8c316f8","revisions":{"03bb2441311149e8c395d60810255890f8c316f8":{"kind":"REWORK","_number":1,"created":"2022-02-01 16:56:42.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/23/827323/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/23/827323/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/23/827323/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/23/827323/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/23/827323/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/23/827323/1"}}},"commit":{"parents":[{"commit":"e3f7f8fecd9dadb148fb0212446429906ac05899","subject":"update constraint for oslo.service to new release 2.5.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e3f7f8fecd9dadb148fb0212446429906ac05899"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-01 16:56:34.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-01 16:56:34.000000000","tz":0},"subject":"update constraint for neutron-lib to new release 2.10.2","message":"update constraint for neutron-lib to new release 2.10.2\n\nmeta: version: 2.10.2\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: yatinkarel \u003cykarel@redhat.com\u003e\nmeta: release:Commit: yatinkarel \u003cykarel@redhat.com\u003e\nmeta: release:Change-Id: I7ad7baccc1d092d6938e9231f68862e06c3e00ed\nmeta: release:Code-Review+1: Lajos Katona \u003ckatonalala@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: Ia57550a30064674503820444e4f476f058efbd9e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/03bb2441311149e8c395d60810255890f8c316f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/03bb2441311149e8c395d60810255890f8c316f8"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
