)]}'
{"id":"openstack%2Frequirements~827338","triplet_id":"openstack%2Frequirements~master~I1588e69fafce5fe9e8e6590dc1a53a58a2b1048c","project":"openstack/requirements","branch":"master","attention_set":{},"removed_from_attention_set":{"14288":{"account":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"last_update":"2022-02-02 04:58:57.000000000","reason":"removed on reply"},"10239":{"account":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"last_update":"2022-02-02 07:38:25.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1588e69fafce5fe9e8e6590dc1a53a58a2b1048c","subject":"Cap zeroconf to \u003c 0.38.0 on Python 3.6","status":"MERGED","created":"2022-02-01 18:40:14.000000000","updated":"2022-02-02 07:41:00.000000000","submitted":"2022-02-02 07:38:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"827338","meta_rev_id":"1afb98ad3f37a6d63aa7c86b71b0dbfba4f95823","_number":827338,"virtual_id_number":827338,"owner":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-02 07:38:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-02-01 22:49:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-02-01 22:49:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-01 18:44:27.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2022-02-01 18:46:34.000000000","updated_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"reviewer":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"state":"REVIEWER"},{"updated":"2022-02-01 18:54:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-01 20:56:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-01 21:05:59.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"a3253bf3af3392908d26d5da51b6873d078e0362","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2022-02-01 18:40:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d982716551acb44296572d0cd32bc99f67ef719e","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2022-02-01 18:44:27.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"5e6dbcd905fc17829b9b9f21009b0526289a3b02","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2022-02-01 18:46:34.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"ccb98366ce12ff2556edf53d5cf9dbab845f2918","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-01 18:54:14.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/996bba85bc3c4c29bc1e00c2154d0c99 : FAILURE in 9m 43s","accounts_in_message":[],"_revision_number":1},{"id":"261809716d85da3e1b5af2c729ceb8c10cc41322","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2022-02-01 18:57:07.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4db55008d1335c4111efe050dfead7659165a012","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-01 19:11:06.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/4411fa0236a84f0f8a537b113e943feb : FAILURE in 9m 29s","accounts_in_message":[],"_revision_number":2},{"id":"c17febfbbb4a112a9ef0bf00f91ee31635877c02","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-01 20:56:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b29d7d39ca64e2bab9c19f29dba0de0 : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42903235e7e24b9fb00a1810c7d72b03 : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d407c53bc58945f0bcb4670a41829886 : SUCCESS in 5m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1ab80f7450fa42869fe4ed959cb01549 : SUCCESS in 5m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d157f2d69baa4883a3413f4dcc90111e : SUCCESS in 4m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5f8abcc781db4abeacb07abfadb8d39f : SUCCESS in 4m 58s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e09799243c3f4f418d5dc0ddebb1fa07 : SUCCESS in 8m 51s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6d7b66f85b964dc886f556849f53abfb : SUCCESS in 9m 02s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/0561ead42d4a485ab3ccc46a6605dbdd : SUCCESS in 7m 32s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/c138b820d9f4412dbd02d72c08c08415 : SUCCESS in 10m 10s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/a934b9716d3144d69f4618efaa374682 : SUCCESS in 5m 41s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/84aa77df89bf453591362684080c39c4 : SUCCESS in 6m 16s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/55d49da91b514896b0a42bec500cb088 : SUCCESS in 11m 49s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/00f2e0dd6f514a768ced29fb0f96b6ab : SUCCESS in 10m 02s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/1c64980d087d473e98b0190f226e7ec2 : SUCCESS in 10m 08s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/5d3d2c28e35f443ebf84266144b7c0d1 : SUCCESS in 15m 42s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/9f3ac071b59046248250a3b816d93313 : SUCCESS in 6m 16s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/0a3e81220ee14f5390f989905738e2c3 : SUCCESS in 15m 49s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/201882b4409743618ebf1476e5937ae7 : SUCCESS in 6m 55s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/2f152edfc9614f359d13ed6c50a0ea76 : SUCCESS in 25m 39s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/e70546ba0b0049b193be816c283b8fcf : SUCCESS in 12m 14s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/97faa48ab774418a9b1bb1ed9a6bf80c : SUCCESS in 6m 43s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/af6daa5acb634dd0b03afd1a7550969c : SUCCESS in 4m 02s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e73ba83610444e37aad26ce043d86559 : SUCCESS in 25m 11s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/d5a91e710b014288993def0c761e62f8 : SUCCESS in 7m 05s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/d1afdaac6a054c209755126a1e2da9a1 : SUCCESS in 4m 55s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/314021f952fe47eba3e8ef96b0229a4d : SUCCESS in 6m 52s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/4dcb878ad7ab4d8fb229f617f66d8aec : SUCCESS in 10m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/56d4f32c27974f81b749ea7c4b7f8cc4 : SUCCESS in 42m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ed050935b4bc40feb1de51225359f886 : SUCCESS in 1h 53m 18s","accounts_in_message":[],"_revision_number":2},{"id":"43465a1705822c191b7de769d36c810455bd3629","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-01 21:05:59.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"02d5d1eb0db5756b1e411a4f3dcfcb5981a75a67","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-01 21:40:40.000000000","message":"Patch Set 2: Verified-2\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"62ec4b1bd232fc3fd919ddc9e67a5a986ab89185","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-01 22:49:03.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"703ea1a883746bc5bfa43b2a5222e103c4c26353","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-01 22:49:28.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"058b7c7be004b7dfbaa6608e045a8b45d42ba748","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-01 23:03:09.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/d46242250ad24d54a9a07a59f8109314 : FAILURE in 10m 01s","accounts_in_message":[],"_revision_number":3},{"id":"9dea29fc6b3edcfd5fbf496ee894d690e662da4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-01 23:56:40.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36d9b19391964148afbab0e8091b9ff4 : SUCCESS in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad062f9238724d67a483a60f84eae07b : SUCCESS in 3m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/35b09f58901b4a4ea9704dc62cf96567 : SUCCESS in 4m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/303610e2c8144267b475d58776552d73 : SUCCESS in 4m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/45b255ee09394b2b87195ecb61f27ee8 : SUCCESS in 4m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/747212f68edf4f58b3e4b3696b127eac : SUCCESS in 3m 22s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/84e86ae25bdc425ea74a8c9e33cf2dfc : SUCCESS in 8m 55s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e3fecf8fc2af4710bc7f432d2b7f2e4f : SUCCESS in 9m 09s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/36c1c527c27746d2952cb57a37937852 : SUCCESS in 5m 32s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/9dc765ae9f3344cd9b1e04c923c1d4bd : SUCCESS in 9m 13s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/ddebab1d0af84c7e8eba9f5b4a28f53c : SUCCESS in 4m 58s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/f1a8ffa7faef4c8e9b7af2e8ddd9661b : SUCCESS in 5m 59s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/2f556dca21804b7db3a99a1a9a25f30b : SUCCESS in 10m 40s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8308508f1646417190708149028a6951 : SUCCESS in 8m 20s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/d0e0e1e2018d42628d00ea4832a13baf : SUCCESS in 8m 54s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/21690bd4870648c5a29800780a423a64 : SUCCESS in 15m 35s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/0339fc2b39514b2aae91a6dadd49b791 : SUCCESS in 4m 20s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/f2d2bf9f4969471f9ac5b58c3ad08279 : SUCCESS in 16m 10s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/95be4652db594f2187d9d2aef1309338 : SUCCESS in 5m 50s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/0eb7aef0a2684736b24762825d1def1c : SUCCESS in 21m 33s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/682349e136eb40c8a91346f5a787fb26 : SUCCESS in 12m 29s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/1daaf9ce618a480587732bec9c179c4a : SUCCESS in 5m 50s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6bca5bdcadad4dd0b3e7bb9a02f6bc14 : SUCCESS in 3m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a7ca438f7edb4151a04a5858d63367a7 : SUCCESS in 19m 33s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/50a4a1c0705048578b8829de64bcbc31 : SUCCESS in 6m 34s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/fbffd3b956c54b7ea3b8e0b453a3529b : SUCCESS in 5m 26s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/263f2cd863364225a334542e9e799dba : SUCCESS in 6m 15s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/673b68e8a11d4a81a5fbb9c2f45dc017 : SUCCESS in 12m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2dd69568320141ec934ed8721191c1a0 : SUCCESS in 39m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e8de61acdae94d3f93a7b70ea4e74463 : SUCCESS in 1h 06m 01s","accounts_in_message":[],"_revision_number":3},{"id":"e263744732b3e117f2b0ba02f579a16582fdc8b6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-01 23:57:26.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"e1cd442771774f3d7d402cd08c94e2892467ecb5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-02 01:39:03.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25b058903ce14c6d823fcb1588d907ca : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/712963aeed0c4633b732480ecd9749d9 : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/69e7e318d2a443029b11861c8627acf8 : SUCCESS in 5m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/64621edd5ae343f4a09426d58b8d251b : SUCCESS in 4m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a79c5db4010944dca487d629f2377939 : SUCCESS in 4m 47s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f98a70cae5b24902b8e7ee5a29c3dcc6 : SUCCESS in 4m 18s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5671d2107b684537aa790f5e883786e3 : SUCCESS in 8m 44s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/4c3b4c3defd2488281aa9a31225765b4 : SUCCESS in 7m 37s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/7662fb37a0b04caba60cc6c97fef7439 : SUCCESS in 9m 39s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/b8ecddb70c844a09ac6a385eb6b15168 : SUCCESS in 5m 24s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/48c004e2edbe48ebbd185f80c50bfb00 : SUCCESS in 5m 49s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/0102999ddb834988a30a86133a36fa1d : SUCCESS in 9m 14s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/31015a0e1fba4404bc8fc4ed85149244 : SUCCESS in 9m 13s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/0d02a59db7cd45008e5af814a17ae713 : SUCCESS in 7m 52s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/ca18c99e11704da1ad899c35fd484429 : SUCCESS in 11m 35s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/a538cce05b3e4d3b858a94769800e4f3 : SUCCESS in 3m 43s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/6133ddb2db9243ea82a0a614ec846617 : SUCCESS in 26m 22s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/fa747f0cd65d4a8f944c78f3a933546a : SUCCESS in 6m 07s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/a502f72b43c14caab26db7a3b72eeb75 : SUCCESS in 20m 29s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/0df90abd232c40d69e48063bc8913fef : SUCCESS in 9m 53s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/d76da048987a4ec7b167936d6f764424 : SUCCESS in 5m 27s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d143c75eb5c644a78342703e136aee7d : SUCCESS in 4m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/58b784fb422740d8aee0cabc8015b682 : SUCCESS in 18m 53s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/5f6c069f26be4679892e357f95f0fe42 : POST_FAILURE in 12m 19s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/1ef8d1f13b494cdd9ea0563a6c436644 : SUCCESS in 5m 34s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/206e663fc09d4acfab7af8a0ef32b55d : SUCCESS in 6m 32s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/5ec32b45b3dd4e36baf0d7a8a0cf7613 : SUCCESS in 12m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1b373357df1340d7baaa81eace47d5a4 : SUCCESS in 41m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/276fbb04fe7c484c86d656d554c73a7f : SUCCESS in 1h 39m 14s","accounts_in_message":[],"_revision_number":3},{"id":"f5767a9d2a6ba4cdcfeccfb21cbb41194a1b100d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-02 01:44:36.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"60d95c6b259ab21b4666f4c3dc1e6ed4517ae6dd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-02 02:03:06.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/f6298234b7794af192513360fff25168 : FAILURE in 15m 47s","accounts_in_message":[],"_revision_number":3},{"id":"8c78f810fdb9be4ec80cd8317c8c1d6a7b527fd4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-02 03:09:15.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49241993890f485b89f5ea614c35fcdd : SUCCESS in 3m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2878f147b5904758a8d54d56618320e7 : SUCCESS in 3m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b032c7f2e0d8457f904ad842498dc0ce : SUCCESS in 5m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cb8c965c2c9b409a9fb6b9357f2efc0e : SUCCESS in 4m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a8edeeea495848918d36af9b5794a630 : SUCCESS in 3m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8421570d7a7f4cd4806d275f057be97f : SUCCESS in 3m 36s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e2179f35ac1541efb50f17b47a001b08 : SUCCESS in 11m 17s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8bad240550b745d898d1df2b3ca763ca : SUCCESS in 9m 09s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/7cc35d40de73462e8df06939bec509fe : SUCCESS in 7m 00s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/325b408d6bfc4e9eb6637e05980f6a89 : SUCCESS in 9m 26s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/16942674f5a94163992685b91e554162 : SUCCESS in 4m 35s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/282ad35a4ea84d81a9779f742b3105dc : SUCCESS in 5m 07s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/9764faaa3afa40588248b8ad7517c369 : SUCCESS in 10m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/20cadb3bf14e4968adf6e059e88afc66 : SUCCESS in 7m 50s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/09eef740c99049919ff5b1fee365a22a : SUCCESS in 10m 07s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/196f68d653f647288ce35825fed83311 : SUCCESS in 46m 39s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/7c647fd77d2c46d3843585e162c0424f : SUCCESS in 4m 12s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/953ec9160c174a1a89def9e71d4b2089 : FAILURE in 26m 00s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/301a0609b0234fb2a7e1b42456f68d0b : SUCCESS in 4m 44s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/223d6ff80d184fbdac23cb6433acd2db : SUCCESS in 24m 15s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/580ff213a840410eb4c43b97df476093 : SUCCESS in 19m 47s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/53967f7dcd454c329ed6c9ba7e2ae0a2 : SUCCESS in 7m 16s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d5f7999051334da5ae1555ff6c852d9f : SUCCESS in 3m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ce38cee9c513408082e03fa40513ecc7 : SUCCESS in 19m 18s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/c0318c399a534523841a5d0e3333f141 : SUCCESS in 5m 32s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/a613002d96e14978946e386b17754111 : SUCCESS in 5m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c1e8baaa36454b2dae9dafb7298b8222 : SUCCESS in 6m 29s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/e3bb6d85e0454779ae6856f35b99983b : SUCCESS in 11m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8887dca738b24c479ef125cb9923703f : SUCCESS in 49m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/71b4ef2a14c64735b8e7c267dd1d380f : SUCCESS in 1h 19m 23s","accounts_in_message":[],"_revision_number":3},{"id":"763102713f0d002dda7d34f4959dd5ca3eb2989d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-02 03:10:54.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"db878201c67c3d960e3868c39d2e9b02a175197a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-02 03:26:18.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/5e038f6cf1994c9ca8de028a1599f9b3 : FAILURE in 11m 20s","accounts_in_message":[],"_revision_number":3},{"id":"21e445a30c0deb19b7adad90f075446c43a0a485","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-02 04:49:30.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85392e333b8e473182b6224ad51bf798 : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42cf3106f0924f30893a027ac64a74a5 : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0e1f38a2a4214b81b0f5499a3cbc4388 : SUCCESS in 3m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f16a235cb3754d81be067660b84f569d : SUCCESS in 4m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b0e48eaa780d491689147532dc9ec298 : SUCCESS in 3m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7a6b4a7d5624461b96a8998239481b9e : SUCCESS in 3m 27s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6b9df6fab7114e2ca3b970b6d39933b9 : SUCCESS in 8m 02s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/56bef3b50f1947a2bc12a171e7337fa7 : SUCCESS in 8m 34s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/97e4c2fdff5f479289f5b7c7c0c9af76 : SUCCESS in 5m 17s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/49cd1e868d8442f58691dd6e62a05d8e : SUCCESS in 9m 32s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/7dfa02d494094908b7c4a3b9bfbfb736 : SUCCESS in 4m 49s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/c7e589e0f2f74fd6b2d191b7c443571b : SUCCESS in 5m 15s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/c6d6b2037d244d34975b5496f888f884 : SUCCESS in 10m 32s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ce0f703bf5654e1eae119f75f3255146 : SUCCESS in 7m 28s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/ed7071efa9534f378f8bda229be43bca : SUCCESS in 6m 50s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/5c0b4e93ea594b7b9d581ec5181f1010 : SUCCESS in 12m 02s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/757a32412d7b44bf81433bc32fee23b2 : SUCCESS in 3m 56s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/5e0116b477b648818d9b71d5f3258e97 : SUCCESS in 15m 38s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/9805a7861d0b4ba1a1878d4756df8291 : SUCCESS in 4m 34s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/a06e107e75d94b03bb61b7ae8e15d7d2 : SUCCESS in 22m 16s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/72c1b8a5084a48e883930a0621258895 : SUCCESS in 10m 46s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/36362950ebab4882abffc2deb19fbb8a : SUCCESS in 5m 02s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8ccb8474e1a7488eb9d6e789c7d5a91e : SUCCESS in 3m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e18f446c0ab64f748f934f62775f4503 : SUCCESS in 17m 09s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/3d2984d73d7e439cb6b35475d3ebd67e : SUCCESS in 5m 58s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/762586a1271243c6a6b95a15b1d4fd0c : SUCCESS in 5m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a3017478fc764ff78fef676f6b7bdc19 : SUCCESS in 6m 35s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/8591117eeeba4e559e4b27028a9859f3 : TIMED_OUT in 41m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/36d02ae1c5184b999d109f8fa9f008d3 : SUCCESS in 47m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/23701ed3eacd4d5ca9a7b6d297f98cdf : SUCCESS in 1h 31m 37s","accounts_in_message":[],"_revision_number":3},{"id":"3888a59e041d1ff93d042b436f5b6754510b86a6","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-02 04:58:57.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"9e57957c04644e9c3231688131f216a1f921c4ab","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-02 05:15:01.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/9cd0702ae9754909b3431b4be69c43ba : FAILURE in 11m 10s","accounts_in_message":[],"_revision_number":3},{"id":"b5555c463255a0230000e681dcb1f169b45427a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-02 06:22:26.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/103d95009719481596186dfbcc001ce7 : SUCCESS in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e5a4968cf464cd09ac9e61fcc5dff41 : SUCCESS in 3m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a045e26983da4fd1b851280b70eacc34 : SUCCESS in 4m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6fb5de7574c7473a890c6412791b1377 : SUCCESS in 4m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3469d732e9924b5ea9f39cfa8176775a : SUCCESS in 4m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/94719827d114421b8595cef835197893 : SUCCESS in 3m 29s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6f84f23709124512a2a953d1ba090cf5 : SUCCESS in 8m 21s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/4dba9999705249739894988da66f5d95 : SUCCESS in 10m 21s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/030669638a5b4f1eaf10ea4db16b449b : SUCCESS in 5m 33s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/10bdbdd359354af881d60b6d2e86ec5d : SUCCESS in 10m 20s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/97daaef701f744bda79cc56db078630f : SUCCESS in 4m 45s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/d2c29ba168684ce683424edd1448ea07 : SUCCESS in 5m 27s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/1c99995b291e4fe9827000bf4aa10e8e : SUCCESS in 11m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e78285bc75aa48c182f1cd66c993d39d : SUCCESS in 8m 38s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/efc3b790d01c4ef0b90c1db1b0a48de3 : SUCCESS in 9m 10s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/e7a2cde8add6411795ff372f6fdec046 : SUCCESS in 16m 01s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/632fb910abef4fdb82dfd9b3ad6d8551 : SUCCESS in 4m 21s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/a582144ba0b243ffbea8931f85a1388f : SUCCESS in 15m 02s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/debf663424e14115bbe2b0cd363ddedd : SUCCESS in 5m 51s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/d9420920cb364fdfb5f0037a65121810 : SUCCESS in 30m 12s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/9485ace4ba024782b58ffc133013d0b8 : SUCCESS in 15m 24s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/054ceed250174622ae914442146a7fb1 : SUCCESS in 10m 06s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e8b945a90ad94c40a8de9abdedb145fe : SUCCESS in 3m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/eb47b94acffe4b88955417c55d02308d : SUCCESS in 22m 19s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/abb218e0b7f341c7b76a0e4fdd09cca9 : SUCCESS in 5m 31s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/2646287ab0704c34867d65ad5c67dd9d : SUCCESS in 5m 06s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9e0bb7a2736644a99c657f0580fc5ac9 : SUCCESS in 6m 13s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/7975f22411ad42ef8b398f1b7d1d8d57 : SUCCESS in 13m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7dac04898b93421a950d21d1ba0f5eaf : SUCCESS in 47m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/96ed0c02548f460e8dc41553100924e9 : SUCCESS in 1h 20m 59s","accounts_in_message":[],"_revision_number":3},{"id":"ba0197ed2405da1b8a36bab79782c03a9e0ab082","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-02 06:23:10.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"828337b9d3f2ddb5cdcf02bb380a232b526495d0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-02 07:38:24.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/529d61048b3046f2b21f2ec7f2810444 : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d030a68e83e24f9c9079c6d64771a0f4 : SUCCESS in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/63f1d01ee1e54196ac479e01568f5a0e : SUCCESS in 4m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c3842cc196624ba89dc7eb804760299a : SUCCESS in 5m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/615e1b9cf316460581e3112c353d8f68 : SUCCESS in 3m 38s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fb2ccde5c2ff4b6a94d12c75ff6bc901 : SUCCESS in 3m 18s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/87a0cf1d5f754c97a7e43dbf23ef241e : SUCCESS in 8m 40s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/67e429faf22843d4b003ff0b33976c3c : SUCCESS in 9m 09s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/515ae870c9a14a8ea292e20b977cd7ab : SUCCESS in 9m 26s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/2add5a2fba9b4e6fb1104a751394894e : SUCCESS in 4m 37s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/01f1eec33126441fb019f38873ee10d4 : SUCCESS in 5m 01s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/3a8290e27c4a4c728053458b655f5738 : SUCCESS in 8m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6cce3aa7957b46d093886bfff2c349b1 : SUCCESS in 7m 52s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/5902845505e1439691f18a80389f9156 : SUCCESS in 8m 55s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/b68d7227aa984af684f24da679fa8613 : SUCCESS in 13m 42s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/d2a0f04fe1244131bf0169ec78b299e8 : SUCCESS in 4m 09s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/40eacf7c0ce344bf8c3f196281c0aabb : SUCCESS in 15m 12s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/342388cad60e4d3aa0df87aab0b3123c : SUCCESS in 4m 37s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/a09fb512c7a84dd7ab68ea80b2dad578 : SUCCESS in 22m 50s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/5f663f0d3a2d43b9ac2e4440a070bc14 : SUCCESS in 11m 52s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/0f86a10fc3264cb19b1ab319b489728b : SUCCESS in 6m 03s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/95352fdb428e4017bd731662a807bfea : SUCCESS in 4m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8600b476703842379636638bcfe70827 : SUCCESS in 20m 17s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/0d45d3684894427599ee830e2e6cacf3 : SUCCESS in 6m 48s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/d80bb4fedb174977bdad8acc0bcbdea0 : SUCCESS in 5m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8c319e046a2143629b0c575a89570fe8 : SUCCESS in 7m 52s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/bc215d1db78748bda10fb8e864ceb760 : SUCCESS in 10m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/237b3b0310f1484d8f9b2a0a66653fb5 : SUCCESS in 43m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/96388d05dba446f48004031be7f99f5d : SUCCESS in 1h 11m 03s","accounts_in_message":[],"_revision_number":3},{"id":"109196894c6bc0fdb320f28a91f164c6a366678b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-02 07:38:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"1afb98ad3f37a6d63aa7c86b71b0dbfba4f95823","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-02 07:41:00.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6e7c554095f4a15b55b6e0ef321a3fa : SUCCESS in 58s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"abfab1bb8feca121ec2ad52c8b203dde6fc99a3f","revisions":{"fcc106e3a0061db1504205aa4dea78d6e7630626":{"kind":"REWORK","_number":1,"created":"2022-02-01 18:40:14.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/38/827338/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/38/827338/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/38/827338/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/38/827338/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/38/827338/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/38/827338/1"}}},"commit":{"parents":[{"commit":"836b7cecf6bc2d71ef169b254026f80c2372e59a","subject":"Merge \"update constraint for neutron-lib to new release 2.19.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/836b7cecf6bc2d71ef169b254026f80c2372e59a"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2022-02-01 18:39:05.000000000","tz":60},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2022-02-01 18:39:05.000000000","tz":60},"subject":"Cap zeroconf to \u003c 0.38.0 on Python 3.6","message":"Cap zeroconf to \u003c 0.38.0 on Python 3.6\n\nSee https://github.com/jstasiak/python-zeroconf/pull/1009\n\nChange-Id: I1588e69fafce5fe9e8e6590dc1a53a58a2b1048c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fcc106e3a0061db1504205aa4dea78d6e7630626"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fcc106e3a0061db1504205aa4dea78d6e7630626"}]},"branch":"refs/heads/master"},"ad879d7b5d3151b0c42cd2ffdca041aca700499c":{"kind":"REWORK","_number":2,"created":"2022-02-01 18:57:07.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/38/827338/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/38/827338/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/38/827338/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/38/827338/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/38/827338/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/38/827338/2"}}},"commit":{"parents":[{"commit":"836b7cecf6bc2d71ef169b254026f80c2372e59a","subject":"Merge \"update constraint for neutron-lib to new release 2.19.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/836b7cecf6bc2d71ef169b254026f80c2372e59a"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2022-02-01 18:39:05.000000000","tz":60},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2022-02-01 18:56:55.000000000","tz":60},"subject":"Cap zeroconf to \u003c 0.38.0 on Python 3.6","message":"Cap zeroconf to \u003c 0.38.0 on Python 3.6\n\nSee https://github.com/jstasiak/python-zeroconf/pull/1009\n\nChange-Id: I1588e69fafce5fe9e8e6590dc1a53a58a2b1048c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ad879d7b5d3151b0c42cd2ffdca041aca700499c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ad879d7b5d3151b0c42cd2ffdca041aca700499c"}]},"branch":"refs/heads/master"},"abfab1bb8feca121ec2ad52c8b203dde6fc99a3f":{"kind":"REWORK","_number":3,"created":"2022-02-01 22:49:03.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/38/827338/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/38/827338/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/38/827338/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/38/827338/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/38/827338/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/38/827338/3"}}},"commit":{"parents":[{"commit":"9e3a1a58b3481b6ecd94c7304705a52bf909791e","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9e3a1a58b3481b6ecd94c7304705a52bf909791e"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2022-02-01 18:39:05.000000000","tz":60},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2022-02-01 22:48:46.000000000","tz":-360},"subject":"Cap zeroconf to \u003c 0.38.0 on Python 3.6","message":"Cap zeroconf to \u003c 0.38.0 on Python 3.6\n\nSee https://github.com/jstasiak/python-zeroconf/pull/1009\n\nChange-Id: I1588e69fafce5fe9e8e6590dc1a53a58a2b1048c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/abfab1bb8feca121ec2ad52c8b203dde6fc99a3f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/abfab1bb8feca121ec2ad52c8b203dde6fc99a3f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
