)]}'
{"id":"openstack%2Frequirements~827650","triplet_id":"openstack%2Frequirements~master~Iaeb3800e32c36b7dde66ef0cdee6dba14fe9151a","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Iaeb3800e32c36b7dde66ef0cdee6dba14fe9151a","subject":"update constraint for oslo.cache to new release 2.10.0","status":"MERGED","created":"2022-02-03 11:49:39.000000000","updated":"2022-02-03 18:36:36.000000000","submitted":"2022-02-03 18:27:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"827650-new-release","meta_rev_id":"d7324f0b845e724b9619e4f7011b35cd2fc36992","_number":827650,"virtual_id_number":827650,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-03 18:27:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-02-03 15:19:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-02-03 15:54:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-02-03 15:19:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-03 12:09:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-03 13:25:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-03 15:19:06.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2022-02-03 15:54:49.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"2b550f215c9128a294028c2fc639d640431abbd5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-02-03 11:49:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9863cf7511ca01faf549d3fe572274f67ce96412","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 12:09:51.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/7b513ae5b18e4da1b1873a7f94357d91 : FAILURE in 17m 39s","accounts_in_message":[],"_revision_number":1},{"id":"30d8524717c8aed6a835163f0fdabca1daa3d741","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 13:25:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27f5099daaaf45638120ab2f037a2669 : SUCCESS in 5m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/688f3f1bb3064f40aa42a2dcbfdb98f7 : SUCCESS in 5m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1efd3e5454d443e2bfe2d6183843de60 : SUCCESS in 5m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f9fa0b8143174ab39b96e00455674e82 : SUCCESS in 6m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5acd716ef5af4ce6bf2cc9f463e5d450 : SUCCESS in 4m 42s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bbdc3ec8f42c4d18aef4b4b200ce236d : SUCCESS in 4m 53s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e044d31d01064cd2b3119e019cee49e9 : SUCCESS in 11m 46s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/af628549812744cd8fa2ce789fdb0742 : SUCCESS in 10m 44s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/e20cd4f059194f7990a4fd709240c454 : SUCCESS in 7m 01s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/872951f2ea364c31a3642b9571147834 : SUCCESS in 12m 16s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/71e3f737cce746afab4d1a0cb84a5105 : SUCCESS in 6m 34s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/3b3a41502db949f5a2fe6884756aa904 : SUCCESS in 7m 33s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/6ca82b3a2bd84ac2a144ebf3e3e6c635 : SUCCESS in 12m 01s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2242c65107fe4de08cc5c415cfd50b89 : SUCCESS in 9m 32s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/584eb5d08fee4b4a9a7818d2bf9fb077 : SUCCESS in 8m 47s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/90f59776e7464aa2971b15aa785bcbb0 : SUCCESS in 14m 44s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/80f7502cb52649c3a2f7ccf3ebd7f98e : SUCCESS in 5m 04s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/0c77d11d12ef4790bcd1b53647007b40 : SUCCESS in 17m 51s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/89db9ea2f14c44569a6626838b025f56 : SUCCESS in 6m 24s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/45db2f42778346ba96d64c476486eb27 : SUCCESS in 24m 37s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/f9d1815b836e404f840bd8aa6d897cbf : SUCCESS in 12m 48s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/71750f0756564ea2bdf7551418c43ad0 : SUCCESS in 7m 15s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7fd93931d0f841ee938aac492caf244d : SUCCESS in 5m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b94ccf180b6c4e55aa8e14a70db6790a : SUCCESS in 19m 08s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/5af5da3743da4af3958920c0493f1673 : SUCCESS in 8m 10s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/7a0e6ec098ed43419a357e32a097ce7f : SUCCESS in 7m 07s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/32e80191d61e431bb6eaf2d1bf3e1f9c : SUCCESS in 7m 32s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/f572ce41364b404f91f10b9259282bb1 : SUCCESS in 13m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c1b725cf63834b1e89138bd75f0db1ec : SUCCESS in 46m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/70d783631cef4024886a7b64df338de8 : SUCCESS in 1h 28m 37s","accounts_in_message":[],"_revision_number":1},{"id":"9a7047daffa0a0be2fcf92fe3fffdfa00a033870","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-03 15:19:06.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b3834a843a7738a27a5c4ae5651b694f7ca82b71","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 15:22:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d88ae4f825ae25478a11cdd90928d37d9758e5bc","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-02-03 15:54:49.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"18fd2c098a916ebcdb3a751fd17079971fad7d9b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 18:27:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d42ee29b9d484f3c853b67280e42f1a8 : SUCCESS in 6m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb39b37f213244e58b991495c753f0a8 : SUCCESS in 6m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/76842191b51a43f0a4b486eb41c82490 : SUCCESS in 5m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/78e81076693742c09bdc073c4d9acf1f : SUCCESS in 5m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/05022817d4924743a992e96929d24011 : SUCCESS in 5m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d4f4740836a5483fbb74ad44c397c49b : SUCCESS in 5m 46s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1083952cd8494185b1d2d8d953094074 : SUCCESS in 10m 32s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/297f14efa74a423b8bf6f9d1dcaf9f8b : SUCCESS in 10m 12s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/9eb1924f496f4fa0b134a268ec0f3066 : SUCCESS in 10m 28s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/eb88f34538874bb5b1b3225d69227bc7 : SUCCESS in 5m 51s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/670ae95deb6b496fb792becfcd350c91 : SUCCESS in 6m 42s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/241213a8560644249653fa3ef29cf8fc : SUCCESS in 10m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7426d502c824433aa0fb7fe3d8a52200 : SUCCESS in 11m 34s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/44ef326744df426ca1cbf8d19b783c0f : SUCCESS in 10m 13s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/61026f2807ae4b91a77f5b4c273ec4da : SUCCESS in 17m 28s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/c21b0cbee4d14c3bb599cc7928f653ab : SUCCESS in 7m 03s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/8a3ec95f40b744679f6c2fcfa9678fda : SUCCESS in 16m 48s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/a90d4c8d478644ec9477f69d05a99e32 : SUCCESS in 8m 08s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/f5f9d5bb56064a8398206a8ec32159dc : SUCCESS in 23m 23s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/7085dc4383ea4164b108625674630596 : SUCCESS in 14m 41s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/0ee89259bc214eafafdef2b30b2703f3 : SUCCESS in 6m 32s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c59cf4a8cdc4429b893c0d3a73a58479 : SUCCESS in 5m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cdcf46116c244dc3959ecf59286bdc8f : SUCCESS in 25m 15s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/28db583f2e7b4af29201ac899edb77f0 : SUCCESS in 8m 07s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/760abc59abb64194ba0008a4ff50fe17 : SUCCESS in 7m 34s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/14a390b38586480fa9b398efafc03ae3 : SUCCESS in 9m 35s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/3c8187e570ef474bba08c5d33c7e9a0a : SUCCESS in 13m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c8760b532da244efb72b35989c375147 : SUCCESS in 53m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/83a3bf6faba3410fb6b2ba2150d97680 : SUCCESS in 1h 31m 30s","accounts_in_message":[],"_revision_number":1},{"id":"e35e3b90342bec77ab174ce714559e80b583ccfe","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 18:27:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d7324f0b845e724b9619e4f7011b35cd2fc36992","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 18:36:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe894b0066a1427c9441ba3ed3696ece : SUCCESS in 2m 04s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cc9439a8355322899ddf1d7d957da5c3175f9d5f","revisions":{"cc9439a8355322899ddf1d7d957da5c3175f9d5f":{"kind":"REWORK","_number":1,"created":"2022-02-03 11:49:39.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/50/827650/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/50/827650/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/50/827650/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/50/827650/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/50/827650/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/50/827650/1"}}},"commit":{"parents":[{"commit":"abfab1bb8feca121ec2ad52c8b203dde6fc99a3f","subject":"Cap zeroconf to \u003c 0.38.0 on Python 3.6","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/abfab1bb8feca121ec2ad52c8b203dde6fc99a3f"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-03 11:49:37.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-03 11:49:37.000000000","tz":0},"subject":"update constraint for oslo.cache to new release 2.10.0","message":"update constraint for oslo.cache to new release 2.10.0\n\nmeta: version: 2.10.0\nmeta: diff-start: -\nmeta: series: yoga\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I27c955b47fafcb9db3be70d2a8f508fe9def348b\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Iaeb3800e32c36b7dde66ef0cdee6dba14fe9151a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cc9439a8355322899ddf1d7d957da5c3175f9d5f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cc9439a8355322899ddf1d7d957da5c3175f9d5f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
