)]}'
{"id":"openstack%2Frequirements~827652","triplet_id":"openstack%2Frequirements~master~I1a52e7de8274a655a6dc89178843396573f43323","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I1a52e7de8274a655a6dc89178843396573f43323","subject":"update constraint for oslo.messaging to new release 12.12.0","status":"MERGED","created":"2022-02-03 11:54:45.000000000","updated":"2022-02-03 19:50:22.000000000","submitted":"2022-02-03 19:47:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"827652-new-release","meta_rev_id":"667fac47855613ea7cbb8106ba9fdf4066b88b78","_number":827652,"virtual_id_number":827652,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-03 19:47:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-02-03 18:03:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-02-03 18:03:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-03 12:18:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-03 13:44:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-03 15:55:24.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"CC"},{"updated":"2022-02-03 18:03:08.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"f86427588154f49d1489db5c342b27b950c77782","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-02-03 11:54:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4e717a6d79c75efc202979db6e1f34ca9683da65","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 12:18:04.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/6cc5c5739c3c45fc91132ccd1b1a0f4c : FAILURE in 18m 47s","accounts_in_message":[],"_revision_number":1},{"id":"6535f7f389f6ed3b71c6b1c862c541c5379de4bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 13:44:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/952143b2111844e1adea47fb16dcb78f : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cffcbbeb03a547648881114877dda6ac : SUCCESS in 3m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/332ac0cdbec342d7a0a69577ce3e94e8 : SUCCESS in 5m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3c6ff08362e54f97a5f2a7841ca1081c : SUCCESS in 4m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/247fa2396b914839a9cc2e5975f75c1a : SUCCESS in 4m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6f3a99619647451b93dde1aa5854ceca : SUCCESS in 4m 35s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9a138bbbf7c84212a9bcafa6e1ec5297 : SUCCESS in 10m 09s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f4abd52a110242cbb40cd0f7d3c951b7 : SUCCESS in 9m 21s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/a38143bf242f4d93bd0ea89abd6fb856 : SUCCESS in 5m 59s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/11329b3644c442268d6c86ef17f77f80 : SUCCESS in 13m 56s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/5dec384a1ff94021bebc8905e6836f61 : SUCCESS in 5m 56s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/4e8cf91855f64bfb9fe096d8661531e4 : SUCCESS in 6m 14s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/6dfa03819da14641aeda8c90c02de8c7 : SUCCESS in 10m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/834fc0ea0a644ac7922a88dc953a56f9 : SUCCESS in 8m 47s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/adc288b2d7a44622b966b28dd3ee3702 : SUCCESS in 9m 55s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/37d7829f2310435889cbdda4a0598e79 : SUCCESS in 14m 10s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/9e5eac1dcee141928362b12c16080918 : SUCCESS in 4m 28s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/533ba16422c2469d8d73cb6cd28354a9 : SUCCESS in 18m 19s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/70a14b7e9c224e50a49c740c7c560eb3 : SUCCESS in 6m 49s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/1fa39802eb274c3bb3b7b2a1b8dc0402 : SUCCESS in 21m 08s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/c0f56a0b7fde49a494175af2de72204d : SUCCESS in 12m 49s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/bcf1be99fae94aaaa4fa52207335ceda : SUCCESS in 6m 05s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f4d3fbe41c68474f8efc72fd855a9833 : SUCCESS in 5m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/176f1818eb4e46ff8c00456a76b1f98d : SUCCESS in 18m 36s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/c5f5651cfff649b7b496358e67d4cb22 : SUCCESS in 6m 13s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/d72ecfcbfb514db996c658ba546b0a45 : SUCCESS in 6m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/628aefb8d57046cd8eae44c825aff0b0 : SUCCESS in 7m 22s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/eabb8732a86c402fa7c3bf188bcd35fa : TIMED_OUT in 41m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/11bdf89bafc84b1e81c8ba07143d8191 : SUCCESS in 37m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/402832d61458445c8ead7d46b5e6193a : SUCCESS in 1h 41m 43s","accounts_in_message":[],"_revision_number":1},{"id":"367088aee53cdb038bc0e8ddda8bd7281cd3a302","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-02-03 15:55:24.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"149f2475d67584a1ef200a10e6d35111dfbfa2c5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 17:39:24.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/4d2003c074e84f67ae7acaa2d5a0a499 : FAILURE in 14m 35s","accounts_in_message":[],"_revision_number":1},{"id":"8ec4ab245ced0706d154b15e63ecf81414a4c888","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 18:01:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea48ca0891c84f7c9672d86f880814f8 : SUCCESS in 6m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30a99ae6c8e34e04bb77d4479a6623e2 : SUCCESS in 5m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fa69e342c4b34c26a2178d3bac6bd576 : SUCCESS in 5m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6785cea1bf724785a3d9c3136ee2daa6 : SUCCESS in 6m 38s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7cbaedbb813942b4a9dd2231580953a0 : SUCCESS in 5m 40s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6e6f16f109a54e0b88fc51938c83e6e8 : SUCCESS in 6m 43s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9b168118efb042028e91c543e899fd96 : SUCCESS in 12m 26s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/971acb0fc99547e29751f1a168934ce9 : SUCCESS in 10m 36s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/faceebd0ffa34b5f9014ab6f50983c18 : SUCCESS in 7m 02s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/3a69775e67564e899927b9493f59a5eb : SUCCESS in 11m 39s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/32c31f1ada8247f0b63a6b143e8c9b6a : SUCCESS in 5m 50s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/e15d6b6aaf134fad9d556143b1cac7da : SUCCESS in 7m 24s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/5f6adeefe045479a9f6f3832bfa31e0b : SUCCESS in 11m 49s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1467761212f442949bf3aa2a6fe168e6 : SUCCESS in 8m 28s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/601e66815cb4437e8a6dd54278987b59 : SUCCESS in 9m 26s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/c3d33620a47540a68a1a3f096e6a461e : SUCCESS in 18m 13s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/320a6fb8c41a4eb29be3dba479eaba46 : SUCCESS in 5m 18s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/b7e56e6e476f4fe893901dc2b29a6fed : SUCCESS in 18m 40s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/69a03633c41b40c0bea7c23f2ac65eb0 : SUCCESS in 6m 25s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/9523dfad86144b9fb8e55626d11201df : SUCCESS in 24m 55s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/37b8b6eba4df49feae052a700c012737 : SUCCESS in 13m 50s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/55ea0e5c527f49a3a9ac3040712e056c : SUCCESS in 5m 52s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8505b385126541f68eabfe3964168841 : SUCCESS in 4m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6c722dd984c04a0ea6d1075ba1fec614 : SUCCESS in 21m 20s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/0c3f8ab2a2bf4bbc8366df18632e88ac : SUCCESS in 7m 53s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/ace76b0cf9294b3bb9ff6cfac6edddd5 : SUCCESS in 7m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/010f1fedd5834d669bd4ea46722c05cb : SUCCESS in 8m 11s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/8a39df55a1c6472b8a72f45bb79efe2b : SUCCESS in 14m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ba01db1e7f524468898be96aa5bc3a76 : SUCCESS in 49m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fb1b2a18afdc4c36b347103b1856bc1b : SUCCESS in 1h 26m 33s","accounts_in_message":[],"_revision_number":1},{"id":"e3fa8b02084eb62e183fcf3d6ebdd703c62c152c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-03 18:03:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7893e7d891466b5b7b8fba9c85a8cc5c94b9de05","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 18:04:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a8211bc312fbb5e5de08f489f2278b82da062e97","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 19:47:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8cb912540114853a2840b61711a72a2 : SUCCESS in 7m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b155e8e8d3894e06903dcc8e26dc60ce : SUCCESS in 6m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7f18283116804389b3e700c5326b02a0 : SUCCESS in 6m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4a7b33963d644131ba13e5ad1fd11d02 : SUCCESS in 6m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e692a88b2a724b889fba8d60515d2a93 : SUCCESS in 7m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e4f1bf89d1144f93ab12e783a1913a93 : SUCCESS in 7m 09s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0837ac436e8d4ceebba5ea738e4e99e7 : SUCCESS in 13m 23s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/dd809afe21de4bb282b811e6f1abd8ca : SUCCESS in 11m 49s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/5319d5f5c7c24976b045d67e0ffaaab6 : SUCCESS in 11m 20s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/4771823a0435499da74270bb34d2e37f : SUCCESS in 8m 13s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/aa7e3d9af86f450da4c4d1351bc461bd : SUCCESS in 7m 39s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/27bf8aa0e46f409fa487da1eaf05377e : SUCCESS in 13m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/78257ca777c3466bade8a5e5158f7491 : SUCCESS in 12m 13s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/e77679b8924f4af283fe60bf61d3067c : SUCCESS in 11m 16s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/0bdacb0bedee4dfeb78977d59940b0f5 : SUCCESS in 16m 05s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/56f1336169124836aa97743891c7ad3f : SUCCESS in 8m 07s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/35746ab972f14da59aab56db95b72c57 : SUCCESS in 18m 58s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/95accb7e3cbd4b5f84620e7c889086f2 : SUCCESS in 9m 04s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/2fe2f5411b084878af44564b9368d8d1 : SUCCESS in 23m 43s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/fbc6f46d71044a47a4b5d4e4e65403a9 : SUCCESS in 15m 02s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/1af3c1da1f87451ca786a60555177acd : SUCCESS in 7m 57s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2351417bde514ef4a1efed31c601ecbe : SUCCESS in 7m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b82d9b715a4e4ac6ac63a5ac4be2de97 : SUCCESS in 22m 31s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/264913b4878a4cdf84c254d3206a4608 : SUCCESS in 8m 57s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/93ca6ef249a943ccb6293ba4310a7ebd : SUCCESS in 8m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/233bba807ba345b2a3abfac86211b53e : SUCCESS in 10m 20s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/e08b6dad5b1743ee991e25d234aa5071 : SUCCESS in 14m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8dbc58f3e9104ab8a21e029b12fc4b3b : SUCCESS in 47m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/93637caac30f426391db0abbf84a6bbb : SUCCESS in 1h 35m 19s","accounts_in_message":[],"_revision_number":1},{"id":"1f9ac5ef9b17771a779955220bd672ca7270f57a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 19:47:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"667fac47855613ea7cbb8106ba9fdf4066b88b78","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 19:50:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e57b9e1dc6aa4ee0b9f1334f6772a1b9 : SUCCESS in 2m 00s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8e3f76297d4d3a12f5c3b24f3694372b463b91d7","revisions":{"8e3f76297d4d3a12f5c3b24f3694372b463b91d7":{"kind":"REWORK","_number":1,"created":"2022-02-03 11:54:45.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/52/827652/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/52/827652/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/827652/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/827652/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/827652/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/52/827652/1"}}},"commit":{"parents":[{"commit":"abfab1bb8feca121ec2ad52c8b203dde6fc99a3f","subject":"Cap zeroconf to \u003c 0.38.0 on Python 3.6","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/abfab1bb8feca121ec2ad52c8b203dde6fc99a3f"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-03 11:54:43.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-03 11:54:43.000000000","tz":0},"subject":"update constraint for oslo.messaging to new release 12.12.0","message":"update constraint for oslo.messaging to new release 12.12.0\n\nmeta: version: 12.12.0\nmeta: diff-start: -\nmeta: series: yoga\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I27c955b47fafcb9db3be70d2a8f508fe9def348b\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I1a52e7de8274a655a6dc89178843396573f43323\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8e3f76297d4d3a12f5c3b24f3694372b463b91d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8e3f76297d4d3a12f5c3b24f3694372b463b91d7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
