)]}'
{"id":"openstack%2Frequirements~827696","triplet_id":"openstack%2Frequirements~stable%2Fxena~I88c554503086272a13f4db4a28649bccf8e92a7a","project":"openstack/requirements","branch":"stable/xena","topic":"new-release","hashtags":[],"change_id":"I88c554503086272a13f4db4a28649bccf8e92a7a","subject":"update constraint for horizon to new release 20.1.1","status":"MERGED","created":"2022-02-03 15:33:14.000000000","updated":"2022-02-03 21:26:51.000000000","submitted":"2022-02-03 21:26:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"827696-new-release","meta_rev_id":"e29a1b98921bbd2a0f0675796362354871d837f9","_number":827696,"virtual_id_number":827696,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-03 21:26:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-02-03 20:08:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-02-03 20:08:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-03 16:53:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-03 17:23:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-03 18:06:29.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"state":"CC"},{"updated":"2022-02-03 20:08:06.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"5ccde257e4ca5b0e80ab60a66c4d53f71bb9ba17","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-02-03 15:33:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e4199cf9b8d7c138b5a313ccc2e989a7dfc07df5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 16:53:59.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/d5d7efeb180c46948eb8e002a015306f : FAILURE in 18m 13s","accounts_in_message":[],"_revision_number":1},{"id":"5cc721180ee37e9b0675fc95c2216e6716746fc4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 17:23:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/772c3acd47474c7d9adb62f0f510d566 : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a20f481abfc94683aeb616fa8362252b : SUCCESS in 6m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1c9cf2ba38ed411ab8e275a8377e1283 : SUCCESS in 7m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b2df8770a8cf4f07a2968893e68abaad : SUCCESS in 6m 45s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6e3e34221012475b9f7b6b89a9d0e048 : SUCCESS in 5m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c06613a5e8cf4241862b5a520fd251d4 : SUCCESS in 6m 11s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6e5c98e42a7c429ba7191b69d558008a : SUCCESS in 9m 38s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0e6abcd0cac043ae943f8150a10d0109 : SUCCESS in 10m 51s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/8e4801df60cf4437bfc7feaa857793d9 : SUCCESS in 7m 56s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/35d48917345d48889d3c9590dedaa4f8 : SUCCESS in 11m 46s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/817cefb692d946c8899d9c75bd9ec6a3 : SUCCESS in 7m 07s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/11b767a865fc41e69c2f9f00c0389756 : SUCCESS in 11m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e4ccbfb1f71d483fbecf4e91c8d40190 : SUCCESS in 12m 28s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/5f02c1d91f0047d1831873463ff753c8 : SUCCESS in 10m 21s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/240bddde8ffd40cea7041000d234aba8 : SUCCESS in 21m 57s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/85d106fd9e3248b28885a650962193d2 : FAILURE in 19m 53s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/6f78f46fe16d4160abf11b6cdbab4ef8 : SUCCESS in 7m 33s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/23bc02923185497ea7a23d7b8f4b8ef6 : SUCCESS in 28m 19s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/5fdc054596fa4d9ea3e6d49447cd2e55 : SUCCESS in 12m 35s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/1d03d4ad9ac14db3867bd22bf576a0f6 : SUCCESS in 6m 33s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e2abd781066040a8aa125536e4f4e4df : SUCCESS in 4m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/81a35030de324b2b9e1b47b57a463370 : SUCCESS in 22m 18s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/18bddbaac82f429d9e73fa14f9a0e598 : SUCCESS in 7m 48s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/1c8da9c014914b05a0c9ac8b91e19d53 : SUCCESS in 6m 52s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6b7ae9f0eca54b688ffdcd918d5998ea : SUCCESS in 8m 06s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/c33d946486df4f8992ae4680bd20fc12 : SUCCESS in 11m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8cd00301596844fcb39aaf751c5ba050 : SUCCESS in 49m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d8a11154cd2445b0bcfc9c598cd97f3b : SUCCESS in 1h 19m 43s","accounts_in_message":[],"_revision_number":1},{"id":"fee4aa17c60c533da1fbc4a903e2827a6f7113a7","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2022-02-03 18:06:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f5533fe0286677dda2614195e4a5f57ab777b2da","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 19:21:40.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/bd569f50f5ae4d559db69495acfa9c72 : FAILURE in 21m 03s","accounts_in_message":[],"_revision_number":1},{"id":"99c3cd50d93a41e02e9f13dc06570c28ee254cad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 19:54:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24b2d4f14ee1478593634178035e6e9d : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/00527811fa0d49cf90b3adf34dce5d34 : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3728858caa954bdaa802e3d6bff0a4c4 : SUCCESS in 6m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e5bce929d4224347913c5a5db4bfcbba : SUCCESS in 5m 26s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/14e9b706a715498e8f915670890f349c : SUCCESS in 4m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ebedcdf34d7a486ba7106dfe911a221d : SUCCESS in 4m 48s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/add2373ed4cc4b369c46b73345d824b1 : SUCCESS in 9m 33s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/7e1f824a84bc40e68a39494af70d4927 : SUCCESS in 10m 15s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/4a0c82ccba894601b4c09691bd8d0229 : SUCCESS in 7m 03s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/082b844d28cd49b58d7f5abcaed720e8 : SUCCESS in 10m 53s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/3ab5932dab2d48ce8ec2bb5776335b20 : SUCCESS in 6m 12s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/6690a05b4b1e4e44a9e633b38c2200b6 : SUCCESS in 11m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/17743ad5083848f0aa725228aae8af6e : SUCCESS in 10m 58s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/04dd7405ec23405286679c35454fabec : SUCCESS in 13m 53s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/0cef081cd4e24a4699da11c52aed2af9 : SUCCESS in 15m 28s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/1d9f8c27c4d343ba9936a20e2a6e7e6c : SUCCESS in 10m 25s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/68afb7eed7984911b1c917fb4907dfd9 : SUCCESS in 7m 39s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/e767e30f843a467884a21c7c05c7b414 : SUCCESS in 32m 35s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/635625ec07804306a507bba0ba1dd6c8 : SUCCESS in 13m 08s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/bace923bbae24d2e91f26ec1b470973a : SUCCESS in 6m 39s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c529dc3d630b4c6bb32d5eb10d1028c9 : SUCCESS in 4m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/78aa9b82023449d494d360778d66c16f : SUCCESS in 18m 59s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/1c76e84ae09a483fa5f640e34d389c1b : SUCCESS in 7m 44s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/a4979a2aed814a9699e16548607dd894 : SUCCESS in 5m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/04ad6a278f7845fe8a9aaaf89075839f : SUCCESS in 8m 55s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/37ab6530434046a381593ea2ccf5f17a : SUCCESS in 14m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f32a04426d4b4762988a080b19658c68 : SUCCESS in 47m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cbade7e4e73648c79ee3674d991cf959 : SUCCESS in 1h 21m 52s","accounts_in_message":[],"_revision_number":1},{"id":"9f478f84b696a15b0115de4f7f694990095a7579","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-03 20:08:06.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c669cff25076c5cc390dcb72ff6d3b256a6bd841","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 20:09:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"974345e07c60adaaf20c9e8078c358dcd93be25e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 21:26:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1aca64645f364698ae5ea4bef3f4fbc3 : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2ed9491757164e4baf78a2b135770476 : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/abdefbc90487492a9c7b6978c148f95a : SUCCESS in 5m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d9db4937b12b4fc38642b481db403944 : SUCCESS in 4m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3ec84ec41b3d46389ffcd187566b45ec : SUCCESS in 4m 54s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/338879854ece485d85869e077e024cf2 : SUCCESS in 10m 54s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2782148810ad47338235d79ff971842b : SUCCESS in 11m 10s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/ca6572e403a44bae9fd0b148e52e04e0 : SUCCESS in 10m 42s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/d30445142c1f459ebd0b6fd65433701b : SUCCESS in 6m 59s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/03b4e0ef3ca94c31ac8847e2f66fcd24 : SUCCESS in 10m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/586c2b2b99254f15a81d7fc7ccb5511b : SUCCESS in 10m 33s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/df2249b7a7844f5d9143660b4360713b : SUCCESS in 8m 04s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/f366692725ef48459c2f051eaa8e8677 : SUCCESS in 14m 49s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/b04c3f936c37480296273d220cb72a61 : SUCCESS in 11m 33s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/de1c054839f3402690aa90ab832a6e41 : SUCCESS in 6m 21s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/6203516dc9854dfdac57e10795c648fb : SUCCESS in 25m 43s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/93b223a7f9de4b7d8b1ad3c530899004 : SUCCESS in 12m 08s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/0b689ab37a304f3d89303281be27ebd4 : SUCCESS in 6m 38s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1636f17f4a91418f843c894e538f20c6 : SUCCESS in 3m 58s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/adc6dbe2cf4c4be5979c70b8e8b7f055 : SUCCESS in 19m 41s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/424c74f3253748f9a100aa455a524610 : SUCCESS in 6m 28s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/46670bc2b1e448aab118216c5196a029 : SUCCESS in 4m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9221ab53d666473ea44f0cfab1af3a76 : SUCCESS in 8m 05s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/676529e97dc94e409fcda5e6557d566b : SUCCESS in 18m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2460baf54a6a4ed0b23aa1c7a1357351 : SUCCESS in 52m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e478ad336dbb486b863dc6a65c7ff6fb : SUCCESS in 1h 08m 29s","accounts_in_message":[],"_revision_number":1},{"id":"e29a1b98921bbd2a0f0675796362354871d837f9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 21:26:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f3cf1c18d715f1f848d0b95956922e255d8b0686","revisions":{"f3cf1c18d715f1f848d0b95956922e255d8b0686":{"kind":"REWORK","_number":1,"created":"2022-02-03 15:33:14.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/96/827696/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/96/827696/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/827696/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/827696/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/827696/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/96/827696/1"}}},"commit":{"parents":[{"commit":"b6344ced106db11dfe8e79aba2f49324eb209707","subject":"Add neutron to stable upper-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b6344ced106db11dfe8e79aba2f49324eb209707"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-03 15:33:12.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-03 15:33:12.000000000","tz":0},"subject":"update constraint for horizon to new release 20.1.1","message":"update constraint for horizon to new release 20.1.1\n\nmeta: version: 20.1.1\nmeta: diff-start: -\nmeta: series: xena\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: manchandavishal \u003cmanchandavishal143@gmail.com\u003e\nmeta: release:Commit: manchandavishal \u003cmanchandavishal143@gmail.com\u003e\nmeta: release:Change-Id: I3308eadb54628e4f9efcd2733c056bac5dd97e95\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I88c554503086272a13f4db4a28649bccf8e92a7a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f3cf1c18d715f1f848d0b95956922e255d8b0686"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f3cf1c18d715f1f848d0b95956922e255d8b0686"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
