)]}'
{"id":"openstack%2Frequirements~828104","triplet_id":"openstack%2Frequirements~stable%2Fxena~I6505027d10b9fdc2648e1d2a011c9f098e8f498d","project":"openstack/requirements","branch":"stable/xena","topic":"new-release","hashtags":[],"change_id":"I6505027d10b9fdc2648e1d2a011c9f098e8f498d","subject":"update constraint for oslo.metrics to new release 0.3.1","status":"MERGED","created":"2022-02-07 10:31:52.000000000","updated":"2022-02-13 21:55:34.000000000","submitted":"2022-02-13 21:55:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"828104-new-release","meta_rev_id":"cd98c0d0e9f0c01eb03e03e5b4638916b08b2b84","_number":828104,"virtual_id_number":828104,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-13 21:55:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-02-07 15:22:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-02-07 10:33:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-02-07 15:22:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-07 10:33:29.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-02-07 11:17:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-07 11:57:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-07 15:22:11.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"be8a961ba4b4c2ef19f7a0c4e09651d5e91c946c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-02-07 10:31:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b629f3376b99d986b24eb81135138e6c9650303d","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-02-07 10:33:29.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8325aaaac3caceea34239fbff992a838aef1b3f7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-07 11:17:02.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/582461147d6b46b096fb4ccb7a62e0e4 : FAILURE in 14m 32s","accounts_in_message":[],"_revision_number":1},{"id":"d3b39e21349e3cb1d509be4f12dd78ce573f211d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-07 11:57:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50f04548e1ab4996bd3b47845d3e2550 : SUCCESS in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8d11574eae0d40369e9010c3b6614692 : SUCCESS in 4m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c10418155edc4f638390c86168e739ad : SUCCESS in 4m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5de936d00aa541b9b1f495d78ed077fe : SUCCESS in 4m 33s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7c3d6fc6b8d8428a80362ded29bb23e4 : SUCCESS in 4m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8d7b7d18d7f1402d94d3c6cf6f4d8f87 : SUCCESS in 3m 49s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2db0d5199f424fdc98653972fd15724e : SUCCESS in 8m 44s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/780f380c465341a8a260e16b89469fc5 : SUCCESS in 9m 00s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/e7accce9bf2d40a79273644a9c3fc183 : SUCCESS in 5m 44s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/7e108a41814744c297d410469ac8959e : SUCCESS in 10m 11s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/059c3c18ce56429f8ea38172f5aef0c1 : SUCCESS in 6m 15s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/9b9e91facc0745638789622503496fe3 : SUCCESS in 11m 36s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9ae68e8463074a33b8a9dcc61903ff43 : SUCCESS in 8m 20s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/499ed9ff6b6d47a6a2a17a35c17bcfa2 : SUCCESS in 9m 16s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/16a9f3930740482da03f92063593af6d : SUCCESS in 19m 26s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/f4e6d1b56a46431f893892f1c8e93fdf : SUCCESS in 8m 35s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/0cc91f3c545d4388aaf05b4146110994 : SUCCESS in 5m 10s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/3faa9d0f104f4fb89793c1d87c69f42c : SUCCESS in 28m 03s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/289eaa10fffd45c1818f0931d63215fe : SUCCESS in 10m 15s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/8a386ee1614944f7b4d9b1f617fd4b7d : SUCCESS in 5m 14s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/fbcd468fd941492289afaf6b0aa6ba94 : SUCCESS in 4m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e7af8239a2e44a6f83501a6832d33acd : SUCCESS in 22m 33s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/3fabbfa46cd04dd587e6e19c0c38d1b3 : SUCCESS in 5m 55s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/7a2d56ef965248e5b15d283ac399d397 : SUCCESS in 4m 58s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b4c7de295b53444b844db1a6e427f9cc : SUCCESS in 6m 40s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/bd630a08fb744945ac4d06c16d19b9da : SUCCESS in 10m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/94411ae11e7141d5a0780ab3245c3e98 : SUCCESS in 52m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8a13218cdbdf4a0c90446e6052e60b82 : SUCCESS in 1h 17m 53s","accounts_in_message":[],"_revision_number":1},{"id":"ffc787369150ead469fc1c701ac541e765491b9e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-07 15:22:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1db733b5e9fc2f92ab05ec68eb5ab6124b1a7060","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-07 15:25:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bb702155f99e0d96a14fa65a9481575d0d3ede39","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-07 19:22:03.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4897434f511347dea0a7e12ca79cfc6e : SUCCESS in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/052cb9760ef542aa84e8867fe0a0fcf9 : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1a30c5a45a064b27970232f9a8b36053 : SUCCESS in 4m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cac5ada701e848548c5bf38f1eb624b3 : SUCCESS in 4m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ddaa9e7d5bee4b9aa1b666e2237fe0e1 : SUCCESS in 4m 13s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/da006dd19b5f47868629882adb26befd : SUCCESS in 11m 44s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9300b479085a49d2a5dd050941d69cd7 : SUCCESS in 8m 53s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/7a8d799d692d4b55a9595022e56a80d5 : SUCCESS in 11m 01s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/12c09bbbfa9c4c348563ab098d51228e : SUCCESS in 5m 39s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/9d19f02b783f431fa1ca749e90561b2e : SUCCESS in 11m 40s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/95158cd7dab84f6ba4dd6e6b2a2a2e27 : SUCCESS in 8m 52s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/19259a0603e748e8a75bacadd0fa86b2 : SUCCESS in 8m 35s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/e10d852e9fc2413fa4f4c0dba673aae3 : SUCCESS in 18m 55s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/9ed19804f13c4a209e73bc1f7c9cf0c8 : SUCCESS in 10m 00s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/a836288d7db545bc82d7bbb25f214a22 : SUCCESS in 6m 27s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/37396b7d9de0427b925dc8cb54e84b01 : TIMED_OUT in 41m 13s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/c53f6cd0908e465fbb6f2876da854a13 : SUCCESS in 12m 19s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/89cf9b729d924e9c89ca04e2208dd57a : SUCCESS in 6m 15s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/54fd2cd32645460eaf2eda89b2c26780 : SUCCESS in 4m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c87f8dced2934eaf83bb4b341e59b40f : SUCCESS in 21m 52s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/921d7ba5e66a4ea8ba97169b9f51b2ff : SUCCESS in 5m 52s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/4a6a98083eb845c39a94dc74cf0f2d73 : SUCCESS in 5m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1341c155ad7e4183bf0a5330cf43ee6f : SUCCESS in 7m 21s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/be314b1ce1924ce686e527952b25942c : SUCCESS in 12m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/57d0e0736f304d9f93b12492b852fd8f : SUCCESS in 53m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4362e0224dc44f2f8571013051b6c667 : SUCCESS in 1h 31m 38s","accounts_in_message":[],"_revision_number":1},{"id":"0b095f9ab260357deb4ff86a50daaeb59d4034cb","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-13 19:55:30.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9636fc95a26027a054d3479b39f03ad3c442b348","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-13 20:17:24.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/c0b4855f4b36438987d3ba2d38efd34a : FAILURE in 19m 39s","accounts_in_message":[],"_revision_number":1},{"id":"c76611d840f9c35855505fd886958569fc064993","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-13 20:59:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc3c2eff27b74f4f94aef97cbc01315d : SUCCESS in 10m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/984007285246420b8fdf47cc0734aa6c : SUCCESS in 8m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c0fb2eb1347446ddbd2988a3998557b0 : SUCCESS in 3m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0830ee37a6cb46ae813f574d5b0416d4 : SUCCESS in 4m 11s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2198840a525d4e00bfbd2b744c4c4775 : SUCCESS in 4m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b06e4badcec14260a9df774816e76ead : SUCCESS in 3m 30s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/92ccf60a9e8e447398166a7a37f3b223 : SUCCESS in 10m 48s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/25795dc615864208bb4e4dde6ffe7c39 : SUCCESS in 8m 32s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/76063e82fc8b411399d093cda365a998 : SUCCESS in 5m 44s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/e47255045ac44c66b6ad38f33b89da75 : SUCCESS in 9m 04s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/d308cff8e38c4aa08b533189a05df817 : SUCCESS in 5m 05s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/de9790a7c3144436a8b1b4dc9ebe4ae8 : SUCCESS in 10m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c1734ab9ee684a398e59a5153dc6baa9 : SUCCESS in 9m 20s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/b51058cb01584bb883c6d8f7ef3089c0 : SUCCESS in 8m 12s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/be7590b1afe745d39e5275350424c42b : SUCCESS in 19m 40s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/2e1a2635ce784223a9b80fbf6f8aea66 : SUCCESS in 8m 08s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/aa39e1ff2e76427ebb242ffe87e5cd85 : SUCCESS in 5m 07s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/2152c2e511b543ad85af1570696fe51e : SUCCESS in 20m 17s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/315f7ff75a744a72b0a0cbecdecaa0af : SUCCESS in 11m 13s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/c0009d588bcb48b0a48eaf94436117a6 : SUCCESS in 5m 52s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e7eac42bbd024687b2c33556b4e89def : SUCCESS in 3m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/44a74337a43c406685e5883b78fdbff2 : SUCCESS in 16m 54s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/422ead99d0fe4d1fb108699a388ab85e : SUCCESS in 6m 39s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/1c429aed144348fdaf0c7c770a54c50a : SUCCESS in 5m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/24f7c862f4a946cda89aff167c4939d8 : SUCCESS in 6m 36s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/a0e9179197d84ca8b9d1410f705facfe : SUCCESS in 10m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b9042a4ce6dd4fe39b150c9b02fdc449 : SUCCESS in 43m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5f313039b94542a8925d4270be1bf7d0 : SUCCESS in 53m 06s","accounts_in_message":[],"_revision_number":1},{"id":"5cabd8cc790fed5eb42f21296012fa9e4d5ef67b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-13 21:00:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5d459503113790cb0f0a3c27ec0fe961dcab1974","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-13 21:55:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f78926267b5345e9a3661a300184af4c : SUCCESS in 3m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/068c4ee9a9ab417b857c48f9203183ca : SUCCESS in 3m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1344617db6aa4770aac2e42ee2fd54c1 : SUCCESS in 4m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/625212d6022b48998ad6fc8fd8928ff2 : SUCCESS in 3m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/491f40e9509f4a85958cb131a6414ca7 : SUCCESS in 4m 29s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b64de047c5d44b96a07a678fc6c4c833 : SUCCESS in 8m 25s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ed5975dc070142db8f3312fd3acfffc2 : SUCCESS in 7m 54s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/e2cb6ac605594dff903b6748fb0a32ab : SUCCESS in 8m 48s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/b3863c7368504a229ac2fbe640bc5c51 : SUCCESS in 5m 16s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/30644f586764482b823f8ee5eaae3b95 : SUCCESS in 7m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c397de93e4734bf7901f4755d2d76f04 : SUCCESS in 7m 20s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/6d0a79b8143f4cf18628bb80162900c4 : SUCCESS in 7m 21s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/92e216e7c63e4d97b49ddd758d3f5327 : SUCCESS in 18m 21s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/eab6160ca4df42d19ce0220fdf5f77f7 : SUCCESS in 10m 30s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/566c42f620fc4fb88392d0cbde0419b6 : SUCCESS in 4m 51s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/6c4f4fab12e74bf19de0ce1736093c22 : SUCCESS in 19m 04s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/4e5f8ca01588499697fb92421ad52a16 : SUCCESS in 10m 55s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/4da396fd27744d4388b637adc247bc0a : SUCCESS in 4m 42s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5cf4f490e6ad4172936d57cf227dbafc : SUCCESS in 3m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c582c1915d994d60bc550f60ff955c94 : SUCCESS in 16m 36s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/8557ba0f9b794641907b83cffa484684 : SUCCESS in 5m 51s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/3b749ff9c65f4dbfbb6ca222fb0e67ad : SUCCESS in 5m 18s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c2429c1519154efab673c0a743fd164c : SUCCESS in 7m 21s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/9f2c4f4c3e484ca8b2c37f6c11a3cd2c : SUCCESS in 10m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/86a86cf0f593473face9b252c6f204f8 : SUCCESS in 52m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1deb8b5b6c484ba2a2d24146e4dad622 : SUCCESS in 54m 15s","accounts_in_message":[],"_revision_number":1},{"id":"cd98c0d0e9f0c01eb03e03e5b4638916b08b2b84","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-13 21:55:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5d886748b001cd6ad883b4f6cdc009efe121c910","revisions":{"5d886748b001cd6ad883b4f6cdc009efe121c910":{"kind":"REWORK","_number":1,"created":"2022-02-07 10:31:52.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/04/828104/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/04/828104/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/828104/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/828104/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/828104/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/04/828104/1"}}},"commit":{"parents":[{"commit":"19b48c99f4354b364675c0c2f02efdf95fc990ff","subject":"Merge \"update constraint for horizon to new release 20.1.1\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/19b48c99f4354b364675c0c2f02efdf95fc990ff"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-07 10:31:29.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-07 10:31:29.000000000","tz":0},"subject":"update constraint for oslo.metrics to new release 0.3.1","message":"update constraint for oslo.metrics to new release 0.3.1\n\nmeta: version: 0.3.1\nmeta: diff-start: -\nmeta: series: xena\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I431930ef1ca69a324898acc36d6b4c30fa892e44\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I6505027d10b9fdc2648e1d2a011c9f098e8f498d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5d886748b001cd6ad883b4f6cdc009efe121c910"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5d886748b001cd6ad883b4f6cdc009efe121c910"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
