)]}'
{"id":"openstack%2Frequirements~828503","triplet_id":"openstack%2Frequirements~master~I31c56063ab6e00391e9a89adfd7fbc9c9f6e5f7e","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I31c56063ab6e00391e9a89adfd7fbc9c9f6e5f7e","subject":"update constraint for taskflow to new release 4.6.4","status":"MERGED","created":"2022-02-09 09:20:07.000000000","updated":"2022-02-14 02:39:11.000000000","submitted":"2022-02-14 02:36:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"828503-new-release","meta_rev_id":"da6855e1e530b19a17f9e4a09b66d0c18611c8bc","_number":828503,"virtual_id_number":828503,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-14 02:36:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-02-09 18:28:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-02-09 12:16:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-02-09 18:28:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-09 10:41:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-09 11:00:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-09 12:16:34.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-02-09 18:28:14.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"2ae90712df2731ff87944c7fecf7cd6715b6710a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-02-09 09:20:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5f680a25124be178b6f910c34a013eca6b4cfb88","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 10:41:24.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/229fbdfb5a584b999ab9f390c0102056 : FAILURE in 14m 58s","accounts_in_message":[],"_revision_number":1},{"id":"35aa2592522d048f393bb573bc32449013541e65","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 11:00:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5bca68e32a04ad3ba12509c8cc77e01 : SUCCESS in 6m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2febfa937bb453abbdd0b5440f294ac : SUCCESS in 3m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7308198c32814752ae49a1600999a355 : SUCCESS in 7m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/28803d5b248f4ed9b661bcd3ca357751 : SUCCESS in 5m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ffdec5ceab3341748a005c512c8c7fa3 : SUCCESS in 5m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ce53b1e56df34c088089c7e6079d5a83 : SUCCESS in 5m 43s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a358af316c2b48078d1c79b4d2d2b213 : SUCCESS in 11m 01s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2f7cfe5037ea4181b9845e43247bdc53 : SUCCESS in 10m 29s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/5cc7137bb79f4b7f819db930828973fa : SUCCESS in 7m 10s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/a2c29594528c441fadc659ecc9deef72 : SUCCESS in 10m 59s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/d2de40fcbd7c42d787ba22e747d58132 : SUCCESS in 5m 35s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/504279f72ac94c508bcd01021cc58fb3 : SUCCESS in 7m 51s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/c03140d877df499a9f77cc1b793624a0 : SUCCESS in 10m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/bb65210dc0124cba9ebabb19a01ff0f3 : SUCCESS in 10m 09s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/75f1524ec6af4ded89717d0dff9686e6 : SUCCESS in 10m 24s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/b8cc3b8b86744717baeeea03c0f998da : SUCCESS in 13m 37s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/7a2f8dd583314632a4db566c62262625 : SUCCESS in 5m 21s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/4cb17c08d4a94fe2b175b7e0ec9692ad : SUCCESS in 25m 50s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/e10e9a7512aa4d9e8e7166510a5146aa : SUCCESS in 7m 10s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/606214ef09444fc38d2e895af0c434dc : SUCCESS in 26m 10s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/eab8b81b27c9455b9d16b342d2eb481c : SUCCESS in 12m 48s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/2a1bca8c17bc490e948767bac21de369 : SUCCESS in 5m 51s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/cef195beba8f4569a8abae365cc3cee0 : SUCCESS in 5m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a2d43302de0f4d799cc6a8266e8de315 : SUCCESS in 24m 50s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/77952044d9404a159b80f3b735ea05b3 : SUCCESS in 7m 19s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/84e1eb5881fd4e1c9606da83609977b7 : SUCCESS in 5m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d645a7dbd2a24e22af53feb7e28629ec : SUCCESS in 9m 17s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/35278de30343484e84afb58d68e91fba : SUCCESS in 12m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a873be51c1b943f7b398c752a5b1d3c1 : SUCCESS in 42m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3ace06cb6a2248c78f3e8cc93ef6830d : SUCCESS in 1h 28m 38s","accounts_in_message":[],"_revision_number":1},{"id":"ca9c78f2f6e9a28451b21de018b4193e1e1adccb","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-02-09 12:16:34.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f7210ff94afb1f92bba014644a2b105f3ec63ca5","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-09 18:28:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9575ccc65e5223963eca97fde0f01a256d6d9ca9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 19:18:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0a2beb6a8d2f2a3c8719d005c1b0bec24e505392","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 21:25:18.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ec34d9baeba430292653ff2918764b8 : SUCCESS in 9m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2de461bf41c84d449b0b597ce4e82136 : SUCCESS in 9m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d8487176a32d4e5fa940e9169cea91a1 : SUCCESS in 9m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0efbd34fe746493e94b1ae8a3d434351 : SUCCESS in 9m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fd54e86c9d2d4979bdc2e7c48f65c726 : SUCCESS in 8m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/226605e3da3a47708fdd1354ec19a885 : SUCCESS in 11m 16s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d9e6334ee32c4035bf794ff757aff7fa : SUCCESS in 15m 20s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ffc2674f599145beab5a3dd560ab0a4f : SUCCESS in 13m 58s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/2e13a4b60f2d43c1ae5c396d173decc0 : SUCCESS in 15m 32s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/f3a1e9b902ff417f915c9e7eff05049d : SUCCESS in 9m 57s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/216f3fdda06244009e27b38765656635 : SUCCESS in 9m 47s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/440cc6c0a3ab4ce5af6f76aedf795496 : SUCCESS in 13m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5c901012cd9746aab60298ba5b6e3b91 : SUCCESS in 14m 27s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/6d71cf39f6f74e0797b59b2a4e7e429a : SUCCESS in 13m 24s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/22968a17a17e4ed6b4a2e0cb29eddc26 : SUCCESS in 16m 31s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/977d72c2de4240f3a5dda6da586915aa : SUCCESS in 8m 38s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/66c36e41a7624858bb4e9b4d4af008ac : SUCCESS in 20m 55s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/9b3c7aa9cb8c4246908d85a0d8774317 : SUCCESS in 10m 11s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/685f09fe9e964684a22ea0aed29f97e9 : SUCCESS in 28m 40s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/bc152bcbe47e45c5a11e107bce1a4a12 : SUCCESS in 16m 37s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/417a4b11411e40e9bac246a7fce9ce33 : SUCCESS in 10m 45s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/cc867e3800914203b706f5e70b23dbc2 : SUCCESS in 8m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ffed4f4664da4ef294e8c6e6f8e6c861 : FAILURE in 25m 29s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/ae8eff28c8e94479a487ca50baf82522 : SUCCESS in 13m 30s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/4367305c374f4a4c891fca3faf443715 : SUCCESS in 9m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0d40ea4b79a34fdc9ae9c5841ba4e5eb : SUCCESS in 12m 10s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/be12c74103a94cfe8e02b88f22114d96 : SUCCESS in 17m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3945dc37022545b69166e078a23faa0a : SUCCESS in 51m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3a45f5355148477b9fcb815e4d134699 : SUCCESS in 1h 27m 02s","accounts_in_message":[],"_revision_number":1},{"id":"3b6a7c881d04e0bd3e69ae0e99578fd3acb791eb","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-11 02:45:44.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e3d3e0c81d32daf5096b4e683401097c4fb20407","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-11 03:57:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e4456c092ce447d87831abff1ecf2c8 : SUCCESS in 6m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/177d587c25c643b88cf758bede84bc8f : SUCCESS in 6m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ee79ead4ce984ac48255555a213c574f : SUCCESS in 7m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/caad32f57d784ecfaad3d2e679df60e5 : SUCCESS in 7m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f0f66493160443d5ac0c3c64fe01a71b : SUCCESS in 6m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/498ec8e18b974a0aa10839778e7f20de : SUCCESS in 7m 18s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b6c6d4b8b5e5480db6ad0d2855e66bee : SUCCESS in 12m 17s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/525f13e8f07645018c5a4a596d6ffdeb : SUCCESS in 12m 30s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/fc19d3cda9264b7893733e8131696aeb : SUCCESS in 9m 05s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/20babb32587b4053acee1f5f1d9b298b : SUCCESS in 13m 34s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/c57b303f2adc4e97833706e4f5c00d6c : SUCCESS in 8m 19s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/4f41a9fa9b76479487f81cd9f8b97149 : SUCCESS in 8m 22s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/fe9999d8d734488299aef93ae8d9335f : SUCCESS in 12m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d2cc3ee4eaad4b7497ec5e47139e5e94 : SUCCESS in 10m 01s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/c0362cbf355a4c258adb5c2bb763dbb1 : SUCCESS in 10m 50s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/aa07351cfb3649a89c4634d2395fe06e : SUCCESS in 15m 06s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/f2df790f957440afaeb37be556f7ef7c : SUCCESS in 7m 42s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/cdad8b3e60f64e2caa6cbf5a22df1758 : SUCCESS in 26m 48s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/cef772cb2e3a4b1283fb5494903bd7b9 : SUCCESS in 7m 53s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/06c5c4955a7e45aca0423835db73b1f2 : SUCCESS in 27m 46s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/e9cb3a28dead46a3a4034b0b84103a3b : SUCCESS in 12m 06s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/ec6b43a62a404f689ada23203903e9d6 : SUCCESS in 7m 48s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/670eeeecf8a74afcbe4d6d3792de70c2 : SUCCESS in 5m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/289b7124764147c4ba06d7bcafc24cc6 : SUCCESS in 21m 56s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/6d11e7eead8a4ff4b3949dfd9abc368b : SUCCESS in 8m 18s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/deedaad9115a4153b55dd231602c04cf : SUCCESS in 7m 32s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7b2ff3d1ac5a463e9d9d9c016350d00c : SUCCESS in 9m 11s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/86a7d458bc1c4e5797ffbe608ccdc785 : TIMED_OUT in 41m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6b3545381efc412b958a3d41d9237a1d : SUCCESS in 46m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7bdd71dc4cc44bd3a86a47bb7bc999bb : SUCCESS in 1h 04m 42s","accounts_in_message":[],"_revision_number":1},{"id":"063ed5eed5ee3fe92cdf23dd7bc4e997ce286815","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-11 04:21:12.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/e675c66244a6458d93d03015554abd61 : FAILURE in 19m 02s","accounts_in_message":[],"_revision_number":1},{"id":"4e3bfee5e583d00934bf8dfa769c82875d244e62","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-13 19:55:43.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"fc934c138ea3940fc90c11de794f23f16af7d02d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-13 20:13:51.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/c472cf2a57d748f8ae7c05576c9a5ecd : FAILURE in 16m 07s","accounts_in_message":[],"_revision_number":1},{"id":"e1fa9e94b2b10adae854664b4552061c8ec7700b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-13 20:57:00.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd4f3a65a8b1487bb57f4403fc4d2700 : SUCCESS in 3m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25bee3eecf684fc6ba83c27db2465832 : SUCCESS in 3m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bb7e5418a14740bebacccb87f1cea179 : SUCCESS in 3m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/86ae27a7df2d49599e3f703d29a6fb38 : SUCCESS in 3m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a452098c108f4e7b896c91101d931be5 : SUCCESS in 3m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1b386841676b4e338ff42fdd0bcc580b : SUCCESS in 3m 37s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/16c2240ef35344db862c1c10c53568bc : SUCCESS in 9m 21s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/94a5e6ac7e604b978c6ca536688e5c05 : SUCCESS in 8m 52s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/77ec061c0988402bb951f15c092984f4 : SUCCESS in 4m 34s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/635377ba4f7b40af9d740d59c65b5da2 : SUCCESS in 12m 18s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/7f6a96faa2b941eba57923d8a4b0f8f6 : SUCCESS in 4m 20s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/51aae45d0f304b63ad3b1b318b988156 : SUCCESS in 5m 47s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/798ed3b3b745468884a810238ab4cf91 : SUCCESS in 8m 01s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f078998f75194f12a8a2d4e5b6446ab0 : SUCCESS in 7m 41s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/976d31e2ba6b4c25959d3917c53597cf : SUCCESS in 7m 53s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/03538d8368b04692a41209af751c7b69 : SUCCESS in 11m 39s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/1225a436df5d4243ad8346312675c540 : SUCCESS in 4m 06s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/1fb251add6714f96b7805ed932877849 : SUCCESS in 14m 48s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/c29ece0968dc44c59e61625323698b85 : SUCCESS in 5m 42s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/20030403a4854ee0a042ef85ac8942c4 : SUCCESS in 24m 07s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/1ca8c025e540444fa191d6fc1ee19f8f : SUCCESS in 11m 30s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/2ecb79f24a424bc7a47153ed00effe50 : SUCCESS in 5m 59s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6e5cce0b49224b05b32772ddc45bbeca : SUCCESS in 3m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b335387b5f824bb8a9bd5791fcf97dfa : SUCCESS in 19m 41s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/d5433995c5994b14ba9be67d65f7f67d : SUCCESS in 5m 34s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/18b0b68dc93d4b4faa3c83b52d992e2d : SUCCESS in 5m 22s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4c26289aab0c4addb7309e3c89429d7f : SUCCESS in 5m 57s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/8b3120b91244447ca5319e13db17f7e1 : SUCCESS in 11m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c23501629b0f4e84a4270b9d7356ece5 : SUCCESS in 52m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a750077d62274e04ba821b8992e046c7 : FAILURE in 55m 32s","accounts_in_message":[],"_revision_number":1},{"id":"79410785e7d8c9ae0345648ba980cdd1a308bdc9","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-13 23:04:31.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ed8461329f576bcd5bdc87841109789227d3b1cc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-13 23:17:46.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/c9fb4e600c4840f4ad0e2a7bd4aefbbe : FAILURE in 11m 21s","accounts_in_message":[],"_revision_number":1},{"id":"aeb159e94f901173e8bdb1f673af3a51ab31dad7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 01:05:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0be0d5825c64abab6d7999c4e6e3532 : SUCCESS in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6b5757580f441f990945570e114e891 : SUCCESS in 3m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2b20310fef3a41dabc1d6da194f438c8 : SUCCESS in 3m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b8787da745bf43dbaf590712709fe77f : SUCCESS in 4m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e5892b2ea9564ce884a3cb01a0ce1437 : SUCCESS in 3m 52s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3732453d2cf64689a75ccc8e9531cd8c : SUCCESS in 3m 27s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/dbe409cef6594283b267606cca5fd132 : SUCCESS in 10m 42s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c85b40f06f864455aa5c279bc119c6d8 : SUCCESS in 8m 30s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/7f58ad7d3c3148b6afae1efd5d98127f : SUCCESS in 5m 07s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/37e821f0b47f47fdb976d5999ab683d8 : SUCCESS in 9m 22s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/a3df88db4f4b4932b39ce56403120956 : SUCCESS in 4m 39s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/9041566bda3846ed966c434af27c1e6c : SUCCESS in 5m 11s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/b3784dff3fd84ce2a9ca7e1875c0b4e8 : SUCCESS in 11m 26s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4daf79a2707644e2899873b24e6115eb : SUCCESS in 8m 00s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/bba4c725887b4d8bad64189394f8bb5f : SUCCESS in 8m 19s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/5b31489261074ccd808f0dda2d2095cc : SUCCESS in 12m 21s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/db23e729c23c486f9fcd311a90ab1950 : SUCCESS in 3m 48s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/f01d315e19994787839380cf47f6e935 : SUCCESS in 16m 36s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/0a8c6dde8b8046e080de9f04675b3fa4 : SUCCESS in 4m 39s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/70aa841dc2134a868e82d6748e5fe03f : SUCCESS in 27m 26s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/1a6e7a3c64cf481e81790a7458d0d911 : SUCCESS in 12m 00s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/caf7d4b6bf534aa2ab58e669600b50e9 : SUCCESS in 5m 17s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/03e77396d08645b7a290772ccd8a8b7d : SUCCESS in 3m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1e4ebf3e85e442d1923eff2bfb953045 : SUCCESS in 21m 11s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/f94216574b40493b936529f858a5779a : SUCCESS in 5m 27s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/3c20ee328370462b9228f2a39d0330d4 : SUCCESS in 5m 25s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/08ce41ab265041b6b417c58406a63b01 : SUCCESS in 7m 26s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/0861fb8b13a24f09967ba1858b1131c9 : SUCCESS in 12m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/26a403d8b9574f7f82f8f2723ac9ba47 : SUCCESS in 54m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/94a423a0c1584a08ad469866c9a21ae5 : SUCCESS in 1h 54m 25s","accounts_in_message":[],"_revision_number":1},{"id":"1550716a8c79985daa2dce4af4930b3cbdde7612","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 01:05:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b6ad4dad518d22151c7f9b5db30fde3fd1447449","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 02:36:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5843fa28f4354654ae7d5ad851478cb7 : SUCCESS in 3m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a31ae316ec6c4e8092cdf795c5e2d568 : SUCCESS in 3m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/950ee5f97fb143e8b9b66f4f6af628ee : SUCCESS in 6m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/152f3463ee2840b6a644c2f743afe7b0 : SUCCESS in 3m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fa5e912f2af4471c8407e603fdf7c3da : SUCCESS in 3m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8c623429a2184c1d845ccbba14c0d819 : SUCCESS in 3m 52s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/80d4860b34604830814b2c2133d0514b : SUCCESS in 8m 39s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ec2b1892faf84502b4ef4b20b2926548 : SUCCESS in 6m 51s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/7ce4e738eb3d4d6a90114b2322541cee : SUCCESS in 8m 21s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/eca5481e3dd8423394e1de76f2a41353 : SUCCESS in 4m 17s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/fa22a71b617c42ac99c93ba4f8e29ca8 : SUCCESS in 4m 35s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/e9496cf64aa3452392ecb1c52f8c0d87 : SUCCESS in 7m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/bf2928b7cc2a4a70a6ecddc3e44253d2 : SUCCESS in 8m 19s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/5df05184f808404aa8d9fced5071430c : SUCCESS in 7m 06s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/70ecff5b84e8430580426086e2fa638a : SUCCESS in 9m 18s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/021f03e365f247819fb5e94c1baa269f : SUCCESS in 4m 59s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/3c43815371a342a5b3a93d926cb7bb19 : SUCCESS in 14m 44s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/a0718ba3b9404a80aed68f2ba3f2cafd : SUCCESS in 4m 17s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/c56a1d90409e4039b88a27bbeab399c8 : SUCCESS in 18m 44s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/0ecd831c693d42328f68e52f8735956f : SUCCESS in 11m 46s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/0a66505334b84349bce045f2e5d4705d : SUCCESS in 5m 09s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/214dcd93a80d47d9bd362369dc69c991 : SUCCESS in 3m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a37eb57beaea4b1dbec4fe63e77f93bc : SUCCESS in 17m 48s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/58cf6a9092b34695a1670af792564f39 : SUCCESS in 6m 43s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/62ab9aa3baaf482892928bd78ab868b8 : SUCCESS in 4m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3f1d0b5dc31747d9b114969b6fbf8cd6 : SUCCESS in 7m 37s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/e840bf9a495b4d30aa212652302bdb86 : SUCCESS in 13m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6beac64b40e04e52854ca89dee8120b5 : SUCCESS in 1h 15m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b1496e5e23b244d3b3a8241bf8bfc80b : SUCCESS in 1h 01m 59s","accounts_in_message":[],"_revision_number":1},{"id":"1d5419c3239976f2a3d04d62276f69743a12cc72","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 02:36:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"da6855e1e530b19a17f9e4a09b66d0c18611c8bc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 02:39:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd57cf8420be4dddb39b5d8c01354e72 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ad3ad158a689c6220dd7233c95085f9e7f9322c3","revisions":{"ad3ad158a689c6220dd7233c95085f9e7f9322c3":{"kind":"REWORK","_number":1,"created":"2022-02-09 09:20:07.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/03/828503/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/03/828503/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/03/828503/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/03/828503/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/03/828503/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/03/828503/1"}}},"commit":{"parents":[{"commit":"698f8e89b69fc29b7f494ead0f04c9d629e10cb5","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/698f8e89b69fc29b7f494ead0f04c9d629e10cb5"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-09 09:20:02.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-09 09:20:02.000000000","tz":0},"subject":"update constraint for taskflow to new release 4.6.4","message":"update constraint for taskflow to new release 4.6.4\n\nmeta: version: 4.6.4\nmeta: diff-start: -\nmeta: series: independent\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I89863c57f449cfad2903f6832ee89e95f039b911\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I31c56063ab6e00391e9a89adfd7fbc9c9f6e5f7e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ad3ad158a689c6220dd7233c95085f9e7f9322c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ad3ad158a689c6220dd7233c95085f9e7f9322c3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
