)]}'
{"id":"openstack%2Frequirements~828690","triplet_id":"openstack%2Frequirements~stable%2Fwallaby~I04bbb0a73aa47675c69450c9c216bdb5f0bec8ac","project":"openstack/requirements","branch":"stable/wallaby","topic":"new-release","hashtags":[],"change_id":"I04bbb0a73aa47675c69450c9c216bdb5f0bec8ac","subject":"update constraint for ceilometer to new release 16.0.1","status":"MERGED","created":"2022-02-10 10:47:54.000000000","updated":"2022-02-11 04:44:06.000000000","submitted":"2022-02-11 04:44:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"828690-new-release","meta_rev_id":"45b203116188d5bb0eb62c0a64a6eddcd7f28057","_number":828690,"virtual_id_number":828690,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-11 04:44:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-02-10 23:00:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-02-10 12:38:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-02-10 23:00:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-10 11:07:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-10 12:20:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-10 12:38:56.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-02-10 23:00:39.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"a277b2d7d72eed9a4c30f91fab3bfa2f145048b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-02-10 10:47:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3fff228edeea87cd5c007cf838b3ec3cc4f71b86","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 11:07:20.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/43d3a14b3710407db2ec147efaba61dc : FAILURE in 15m 54s","accounts_in_message":[],"_revision_number":1},{"id":"1379baad81c174351f24a3881d59e83973780260","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 12:20:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2369236bde094ec9bb9a198f9f6652d1 : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0c3be58824e84336a8ec1ddff5151525 : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5d82d5a9890b48b78e9bf45b9004155a : SUCCESS in 4m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/560ece55d1b44bad9d7b3c7c0eced120 : SUCCESS in 4m 46s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b4b879fbdd1b4bb095b7ef2a0e5c6392 : SUCCESS in 4m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1d4244ae56954d6dbba21c875598d496 : SUCCESS in 4m 07s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/93b7b6a1acea433390860a7af23c7213 : SUCCESS in 10m 24s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/dd6cef6cc42848d7affc665c2d56e1a4 : SUCCESS in 8m 30s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/840ecb447353488eb2c9fdbe612ee007 : SUCCESS in 9m 49s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/3ed0a339ae93481bb4ff6d4bbfc6d76c : SUCCESS in 5m 50s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/de2af091c1eb47759aaa56c2d0df5c22 : SUCCESS in 11m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ad81eea25b824227901ea2a7af1f6d1d : SUCCESS in 8m 46s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/c22585d122b9472db82c919aaa4ab0a8 : SUCCESS in 8m 12s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/869e6ad7808449868451eaa65b0786d2 : SUCCESS in 15m 16s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/443b034aeb444de5b701d02decd552a6 : SUCCESS in 5m 13s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/bca28eb6ad7d44d0acaced72cbf248e5 : SUCCESS in 27m 52s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/d9e70f25689844b18fb94c929e19686f : SUCCESS in 13m 19s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/292d5b14f0b243078c43e6c99aac35b4 : SUCCESS in 3m 58s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b5e09d915aa147439fea8db5e6df2c01 : SUCCESS in 20m 30s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/2e938dc1d1e149089173890cf4194384 : SUCCESS in 5m 46s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e582b6613aea45c79400c3b5ac2a7909 : SUCCESS in 6m 13s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/66d096c630864302a5e6992e358a9f73 : SUCCESS in 10m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/edcfecc2676849419595240047103873 : SUCCESS in 50m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/eb432a82cf124b87a77568e3d1a59045 : SUCCESS in 1h 26m 48s","accounts_in_message":[],"_revision_number":1},{"id":"dfc93cb47b4f8f33336dac909c222f9d51c46586","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-02-10 12:38:56.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5789e039ac2215c3438790e4cd0c6d3a26d0c377","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-10 23:00:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4ce68443bfe5c2ebd979cafda75a57aec485502a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 23:02:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2f040f42ab114b8e3b624298e08a7b9e23c949cb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-11 04:44:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a88ca80d55fc4362b6a9a9aaca9c8c24 : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c62c0bb754114ffeb4311368c42f7715 : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/65af5947f6c84ffa87e7ac2d7a7c18a8 : SUCCESS in 4m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a3761b20505c45a2a9152c9c9266e893 : SUCCESS in 8m 32s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f4bf25505b0f4919b569249c4d3881f7 : SUCCESS in 3m 58s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/16cad38a1c3c46b29dd52bd3c4073347 : SUCCESS in 9m 23s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/25d50190769943f0a3860df36f136230 : SUCCESS in 9m 04s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/82fecc8567564d69ae89ef8a19894670 : SUCCESS in 9m 18s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/b008e37db5014d4a972a5e86246961ee : SUCCESS in 5m 41s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/91b2784e0d7d447f9496f715c2ffc8b6 : SUCCESS in 9m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/974a29e3529b459196d780e0334352ca : SUCCESS in 8m 17s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/ff806bc0e04b446192ba843fd2286e0d : SUCCESS in 7m 40s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/f5dda1705b2c45c4be872dd2271ea689 : SUCCESS in 13m 12s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/91004032b443467aa41fcf74df224bcc : SUCCESS in 5m 12s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/6bd4540f338948b7b01cb91cccc4c068 : SUCCESS in 30m 14s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/06c9f3c25f1e4754b3afae3e4b9841f0 : SUCCESS in 13m 46s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/fed5ef74d24d47cb9b7e26ecaf596b34 : SUCCESS in 4m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/24b7ac32ddac444a9883ea23cee2eb17 : SUCCESS in 16m 59s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/b1674393e73d46baba5ee1f2357ab7a0 : SUCCESS in 5m 18s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9181c527b3aa4bfbaa0cb09729120795 : SUCCESS in 6m 38s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/d282e8b99e9645f6b0590ac3b1c047cd : SUCCESS in 9m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c1d0d88928114434a382f66c51f5edd9 : SUCCESS in 49m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e6b2fd26860a4cfd93a0578a91e23fb2 : SUCCESS in 57m 20s","accounts_in_message":[],"_revision_number":1},{"id":"45b203116188d5bb0eb62c0a64a6eddcd7f28057","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-11 04:44:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"12d141827c21cbc0860d348fb91fc465c685f036","revisions":{"12d141827c21cbc0860d348fb91fc465c685f036":{"kind":"REWORK","_number":1,"created":"2022-02-10 10:47:54.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/90/828690/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/90/828690/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/828690/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/828690/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/828690/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/90/828690/1"}}},"commit":{"parents":[{"commit":"6dce31e7272206f0f1c85bb3dc2767f5a50fb41c","subject":"Merge \"update constraint for ovsdbapp to new release 1.9.1\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6dce31e7272206f0f1c85bb3dc2767f5a50fb41c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-10 10:47:49.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-10 10:47:49.000000000","tz":0},"subject":"update constraint for ceilometer to new release 16.0.1","message":"update constraint for ceilometer to new release 16.0.1\n\nmeta: version: 16.0.1\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Tobias Urdin \u003ctobias.urdin@binero.se\u003e\nmeta: release:Commit: Tobias Urdin \u003ctobias.urdin@binero.com\u003e\nmeta: release:Change-Id: I6ac44990dc80582afa897fc3cb9f5784897456b5\nmeta: release:Code-Review+1: Matthias Runge \u003cmrunge@redhat.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I04bbb0a73aa47675c69450c9c216bdb5f0bec8ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/12d141827c21cbc0860d348fb91fc465c685f036"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/12d141827c21cbc0860d348fb91fc465c685f036"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
