)]}'
{"id":"openstack%2Frequirements~828693","triplet_id":"openstack%2Frequirements~stable%2Fxena~Icca2962ace4b54828855d148b49a4fa9f8adb03f","project":"openstack/requirements","branch":"stable/xena","topic":"new-release","hashtags":[],"change_id":"Icca2962ace4b54828855d148b49a4fa9f8adb03f","subject":"update constraint for ceilometer to new release 17.0.1","status":"MERGED","created":"2022-02-10 11:03:34.000000000","updated":"2022-02-11 04:44:08.000000000","submitted":"2022-02-11 04:44:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"828693-new-release","meta_rev_id":"22e6a7106fffbc59e259b24dc08131af3288aa99","_number":828693,"virtual_id_number":828693,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-11 04:44:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-02-10 23:00:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-02-10 23:00:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-10 11:23:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-10 12:24:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-10 23:00:40.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"890241a740c12e323eab73a0cc784e6822e8f4c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-02-10 11:03:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ac64d7fd9a9c47b7302532829596db2b582aeefc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 11:23:35.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/55e9b54658104ca09c415b2c707538d6 : FAILURE in 16m 18s","accounts_in_message":[],"_revision_number":1},{"id":"38dcad65871e9d332cbe82e6c32a27a4a07c6d0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 12:24:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/698ff5948fef47f38d0cec60a4c63b90 : SUCCESS in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2ef7238da6514b5fa978ba663a165b8c : SUCCESS in 6m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c1f5c50996804a3dbf2dbf6bca42969b : SUCCESS in 4m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/511999cad8434d2d99988fb8e950c359 : SUCCESS in 4m 59s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/91bb28b6805f418fa26004b5253173e0 : SUCCESS in 5m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a675955a9beb4f5a83fc70f20e65a8f9 : SUCCESS in 4m 07s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0064fbdaf68e4fcd9ddc40f2f0c451d8 : SUCCESS in 13m 16s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b4e05a4e045443d28ae159781b1519a4 : SUCCESS in 9m 40s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/cb695996042c4f08ac2ec24387218734 : SUCCESS in 6m 08s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/c046cbbd20e24d38a64c1de461b9c3e2 : SUCCESS in 9m 23s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/abc7413165a44a539cb90cfab588f93c : SUCCESS in 7m 42s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/9569df2ee43f443fa396dca91e355c76 : SUCCESS in 10m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/663d8a36e89f42bda231eb4d93b2f1fc : SUCCESS in 9m 47s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/cf6ef156b5414dc19d14f9fb4ccea1d6 : SUCCESS in 9m 28s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/ecb492e0a4c742488482d349e9cc74ae : SUCCESS in 15m 13s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/d0f1e1bdc0b7454e9879522f3c46435a : SUCCESS in 11m 14s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/2411ea21fb2c4e90b6dbdee73b6cc2d1 : SUCCESS in 5m 00s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/58f77b9b4cb94799884180d902fe0ce4 : SUCCESS in 23m 58s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/04dada88b767489c98bc6a1e243f026e : SUCCESS in 11m 24s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/895a2458910246779e86bea7067cfd05 : SUCCESS in 5m 44s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/64bce50a6c604481a0d0da175b73ef0f : SUCCESS in 4m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ebeea4236b854d18ac63626991e24bf0 : SUCCESS in 17m 55s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/1c17380c6a964ca5b84f42751a75facc : SUCCESS in 6m 51s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/c14b20f8d325466fa1da1769fc244764 : SUCCESS in 5m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/cab396a661d84ecaa81308fcaf89f41e : SUCCESS in 6m 48s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/4d527744235e45c6ab58907ccb92acc0 : SUCCESS in 12m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/34107a5d26dd4351ae43b090d0c10fbe : SUCCESS in 46m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b0fc60fee7624e6186a735a9ac6a8bdf : SUCCESS in 1h 14m 39s","accounts_in_message":[],"_revision_number":1},{"id":"a50f09ebafbd83cba8d669af446711af232f0d04","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-02-10 23:00:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c7530015bfcb06d73a2899ffeaf3da02063c0d1b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 23:02:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"943cb6359ae3f0603b6e20dd2527674d10d32c8e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-11 04:44:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b5ef4e3c7eb431a88d7d0112e00d826 : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/85bc536b7b924acc91fa2e44a5c4e66e : SUCCESS in 4m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8f91ac10feb249fa81518a89f2cdd3da : SUCCESS in 4m 38s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7a0cfed8d7ba40c1a273a11dc344f4d6 : SUCCESS in 4m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7477bfebc2e84da692290e2e9e4392e3 : SUCCESS in 4m 08s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/14dd3c1b0abd46e6bbfd428a941421ef : SUCCESS in 9m 07s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9dced3b8ac914986aa9be64ff8c8f4c3 : SUCCESS in 9m 45s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/a4742e6184c04fefa391dc720332ffe2 : SUCCESS in 9m 57s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/e0887f21a972450d900991580bf26ebf : SUCCESS in 5m 34s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/f34353b0001a4047828b6f17e3b973f7 : SUCCESS in 9m 19s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/aeb7a7215b924e4e8b2b9449182cce81 : SUCCESS in 9m 11s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/7e03b1a5fcf843aaae058ccc5d1ac641 : SUCCESS in 7m 21s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/5190bdc47fd74dec912dda175fe16d0d : SUCCESS in 14m 57s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/b97f2b23df4a476e83c8e1dcb20c79c5 : SUCCESS in 11m 52s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/03de15fc66cb498596c7f150e23a24b5 : SUCCESS in 5m 37s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/f36beae5a3464403ad7b3ec0b6361a58 : SUCCESS in 24m 20s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/71408f5d923e4817b291a90d9841dc35 : SUCCESS in 12m 01s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/55a3faf689c0447cb6dc410f116febc1 : SUCCESS in 5m 50s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8459a1fd43b04af3bac561073b1760b5 : SUCCESS in 4m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/21dd472b8313405cb629396d7720e59f : SUCCESS in 17m 44s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/5a243cbecd4d4798807a64f21d141bda : SUCCESS in 6m 34s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/608343731fe24551a1b4fc463d7716ac : SUCCESS in 5m 26s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/462d2ebf8df041abac9942ddcecada75 : SUCCESS in 6m 30s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/1e9206c6b00c482c8fd6014a27c79237 : SUCCESS in 13m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a91ded3137c6482f984fe52ebb09ee99 : SUCCESS in 52m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f31fa526ecda415b9ac5676cc7acb10e : SUCCESS in 55m 42s","accounts_in_message":[],"_revision_number":1},{"id":"22e6a7106fffbc59e259b24dc08131af3288aa99","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-11 04:44:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a135f20faffb90fed69cd304a67f2cf87f0c1865","revisions":{"a135f20faffb90fed69cd304a67f2cf87f0c1865":{"kind":"REWORK","_number":1,"created":"2022-02-10 11:03:34.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/93/828693/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/93/828693/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/828693/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/828693/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/828693/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/93/828693/1"}}},"commit":{"parents":[{"commit":"630e737e2f3072b54aa01f465834e142407aea50","subject":"update constraint for oslo.messaging to new release 12.9.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/630e737e2f3072b54aa01f465834e142407aea50"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-10 11:03:33.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-02-10 11:03:33.000000000","tz":0},"subject":"update constraint for ceilometer to new release 17.0.1","message":"update constraint for ceilometer to new release 17.0.1\n\nmeta: version: 17.0.1\nmeta: diff-start: -\nmeta: series: xena\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Tobias Urdin \u003ctobias.urdin@binero.se\u003e\nmeta: release:Commit: Tobias Urdin \u003ctobias.urdin@binero.se\u003e\nmeta: release:Change-Id: Id8762695c32cfc9f41efbc7686c4304b5c761e4e\nmeta: release:Code-Review+1: Matthias Runge \u003cmrunge@redhat.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Icca2962ace4b54828855d148b49a4fa9f8adb03f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a135f20faffb90fed69cd304a67f2cf87f0c1865"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a135f20faffb90fed69cd304a67f2cf87f0c1865"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
